)]}'
{"id":"openstack%2Fkayobe~938658","triplet_id":"openstack%2Fkayobe~master~I15d17f45d15c04aced0a4c0fe97e1d798ff46799","project":"openstack/kayobe","branch":"master","topic":"infra-vm-perf","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-01-14 09:44:45.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"35010":{"account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"last_update":"2025-01-14 14:03:15.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-01-14 14:03:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15d17f45d15c04aced0a4c0fe97e1d798ff46799","subject":"infra-vm: efficiently handle `hostvars` during deploy and destroy","status":"MERGED","created":"2025-01-08 13:42:19.000000000","updated":"2025-01-17 13:36:37.000000000","submitted":"2025-01-14 14:03:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938658-infra-vm-perf","meta_rev_id":"d732cdee798e2ace904e346782f9941c0f8298e8","_number":938658,"virtual_id_number":938658,"owner":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-14 14:03:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-01-14 09:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-01-14 11:25:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-14 12:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-14 12:25:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2025-01-17 13:36:37.000000000","post_submit":true,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-08 14:55:25.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-01-08 15:31:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-09 09:17:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-01-10 08:03:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2025-01-14 11:25:30.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2025-01-14 12:25:01.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"c4f8b6cfdc50940cb5e04e2d3d3a3c03f9150e6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-01-08 13:42:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f7fca34b565e9562940d453ffaecf09e3216648","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-01-08 14:02:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"850776de0c9ed7b2c018e0b80661674ab6c4213b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-08 14:55:25.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"397f7080ee266228bd353425f1ebccb01f1ccae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-08 15:31:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b99a07d228f948cd87a0704fb32f0ec7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8aea7063e514d2aad2314c9b77d6923 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f209cdcf2654e3d992274c473540038 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9142cd913984c3fb926b1034c62085e : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ba48661f0dc49ac936b46a8fab42f41 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f18a423adacb49daa580e02d832d67a2 : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4431af9752b44489a0d6a22635b2618c : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a321b153c63424da6b2634c2ba7976e : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3b9d817e17f649d79edd2cb2716a0db2 : SUCCESS in 9m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/52a8efb7a0dd40749db38cc803160942 : FAILURE in 46m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e01e3b9b75bd463e9085a933c2316703 : FAILURE in 42m 08s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d7e9e2be4ba4d348928b156cdcf3cc5 : FAILURE in 40m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a4e8ee69f35047449a18851c020a61d6 : FAILURE in 47m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c2abfa7342204323a74dde86bd95e491 : SUCCESS in 19m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/62eead242e3f445d8a4bba17f066750d : SUCCESS in 14m 13s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2d44af3e7f34e03b28b2e1973b1c65b : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f6e842b7acb1414db7ecb6f9b3f5f4a9 : SUCCESS in 1h 15m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c818ee6440b84fa08902f1bea5293b57 : SUCCESS in 1h 28m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f099b1e4a50e4b82ad2c5b13c2252c5e : SUCCESS in 16m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5819d0b6095f49d09886090e0197a21c : SUCCESS in 15m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f4d5dfb7d5c341629e121df237202d5e : SUCCESS in 50m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a0dceda09a4744b3b8e8fe26f0fa3130 : SUCCESS in 33m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4051a8cf04941108274a357cfd36534 : SUCCESS in 24m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/136427bea51f4751a57c5e848dedf587 : SUCCESS in 14m 49s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/06f336e72fb6459cbafedfec54593de9 : SUCCESS in 15m 26s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6dfa498601b4ef49f675c9fe3313234 : SUCCESS in 11m 16s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/42dbdd128253472cb10e8ae41649f706 : SUCCESS in 9m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d50f2f5b007d49eb943f0bfd0ae829a2 : SUCCESS in 15m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9bcac34fa554b39b3fd5f3a588d0ec5 : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":2},{"id":"9aaa06785ec11addd744ad3d511fbb68ab677d1d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-01-08 23:55:36.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"48f4ab3801e962c4d6f7841dbbda284a7ea0e087","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-01-09 00:10:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"de533859d592a4f55f19d8415896fd178820bd49","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-01-09 00:29:55.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"939440f41f06ffe13846cd138036ad85c1b19a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 01:41:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ffaf8c4b5d941c3af2e82f6a1021980\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f710b7a83c80484b9d8c4185c99a92a4 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6c50829441461cabeadc73d861f3bb : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e186db6545124fd8b7231ce56a2d1e61 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8356ba2435a44b48bff54ba1775c0b2d : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4b3a9d5017d4778bfc5fc9eabe6996c : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/57c3079b31f245e4837ad0b6ec5cdb39 : SUCCESS in 3m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35c68ecf3685436e9c327026099cac7f : SUCCESS in 8m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fbb14f8dc3b54008bb423c2dbda1cc37 : SUCCESS in 11m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ab964f997d5840968d46eee16f801d5a : FAILURE in 38m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/54c4dd4606b448acbc526becf9fc22c2 : FAILURE in 33m 53s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/118c960269e34fd299e064c9750396cc : FAILURE in 45m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fd9011b6fea946b7a89330a788a94ed5 : FAILURE in 40m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7b980bc2c9904ebda281b1d125e2b85f : SUCCESS in 19m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7d69e42123ef4d1fbfa4d65ecc8abf28 : SUCCESS in 19m 53s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7d9899de44e460d876bd7311562319d : SUCCESS in 12m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a5284446dd454ebdb7bdfc987d9a6c44 : SUCCESS in 1h 28m 27s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed02db093a9b42fd9f8ece1b17ec3d09 : SUCCESS in 56m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3def20a51cd648b8a09489ed8f0862d7 : SUCCESS in 18m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55497aa8a38f4d0e8af8538844f00755 : SUCCESS in 17m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6b90d940f80447f8ad6e6f33ecc770b3 : SUCCESS in 34m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ac43ac67bfe644528eb5583d19a15e87 : SUCCESS in 29m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f14a23b0f76405a94e839b6ab53a32e : SUCCESS in 22m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ae6464b95ffa4f64a825694c22a8e92c : SUCCESS in 17m 43s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/429496c95a134091a8c941ff5298f7f1 : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4905c92bb4154e998319725c66e02caf : SUCCESS in 10m 41s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/72913e4e295d440b83ebb98092cef85f : SUCCESS in 7m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/10d41a34093d4743beae88aa424a8ed1 : SUCCESS in 15m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15c02f4e6f88479f89d2f99cfe8233c6 : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":3},{"id":"fbea859075b36defa9b2985861ca09c6c20725e6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-10 08:03:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f502fa9f17d6f9132c3fd84be92237119b6fde2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-01-10 13:48:11.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4b642a4326dd9abcf8113cdaa7cdbfe2614ea8a6","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-01-10 13:50:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"74eac828399710397e2a75e4c86863520a16deb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 15:52:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed73c17575bc4589ab0c51a43336d74a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d677f8c531e4e3f930f2c06022abbc8 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d9605d1f46a46949a88f9624353db18 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e6eca18c75946ab88b7f3b7df3bbe64 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5b0c6c9be4440799ccbe8e0d8cd8b7 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04e68bf565b348ad898b40aac0f8b1fe : SUCCESS in 4m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7c1f5ab61e6b4a9fa523c0014f74da36 : SUCCESS in 3m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c71f458897d24a548049c89b30d239ce : FAILURE in 8m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/45b7421855c746b99a8dfcc89764e714 : SUCCESS in 11m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b5435bcce25c426495e26ca24ea4b2c5 : FAILURE in 40m 41s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3b0daf539cc344359bfef89242a683ee : FAILURE in 40m 19s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05d653a8252c49abbdeb54b1fd33b46e : FAILURE in 36m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0c64bb62045347df9dab6ebe84f1bff9 : FAILURE in 52m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d2469dc576214b0d8b4b401b33e9f0b3 : SUCCESS in 22m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bf6f6c4c209f4fd4aae90ee5effa2392 : SUCCESS in 17m 42s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d38d5419b33e43f98806020c18cf5b08 : SUCCESS in 13m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/575c7e22470c4780a0ffdc8ab764a01c : SUCCESS in 1h 53m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/972288e1669f4adcbd83814a5de447ab : SUCCESS in 1h 42m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8ca06155f7c24ee1a063aaadb0f5ce92 : SUCCESS in 19m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e546249e4504460b1fb4d738ecbeef8 : SUCCESS in 17m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/cd070aa8cd7b436db7309f860f04fc85 : SUCCESS in 45m 52s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bd06b79595fd4f488016b587db7f512c : SUCCESS in 33m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fe26f6806074e54b8a5f15998a479c9 : SUCCESS in 24m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7b9148236e064ba7bdfde0f82e4451bb : SUCCESS in 14m 55s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/7af98705a4974e138737c6807b87bb24 : SUCCESS in 15m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/468324f74b614bb08d85ba3af476bf50 : SUCCESS in 11m 49s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/dc6700acf0b74d1e96a827e5fdec4a69 : SUCCESS in 10m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ddb2df101cba4d42bf2ab532a48fcaef : SUCCESS in 15m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28cb58d521ca4603be835bfd390cd201 : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":4},{"id":"46600df37ee02bc53f278fe2aa3d17f9334c1a7e","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-01-13 20:40:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"28ba579ac63c594ce62c767861b0320bf1a9c9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 22:48:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b1939763d3d4dc9ab003db5ae26c4bf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66e9575e3365415c97892aaa8378864c : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11106146ee004ab6b93c30c8da2bdc41 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1409ace7fea94838b86209c6601e2a24 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f065f73c9bb4d468f28dc15a091e25a : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8446202988ed40b79fb52e25e7063ea7 : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a3a14703e1c642d59d5010dec1d57ea4 : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/389dc7fc8f274c41846648ae7c8573fa : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8d53b593c5b04145bf87e8e2dd8a2f27 : SUCCESS in 10m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/89284a211a784db5b702560f29c34430 : SUCCESS in 1h 06m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6518a12c1f5b46bc94bd324ce10aa82a : SUCCESS in 53m 58s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2be2220797c4b4aa701f9bd8316c33f : SUCCESS in 1h 02m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d97b333372a247dbae6a070729592864 : SUCCESS in 1h 09m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/48a23cad2a1a4ceaa9d78872105dc3dd : SUCCESS in 22m 51s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8ff68bb1948749feb7c0bea872a7f000 : SUCCESS in 18m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3be977abf29742fe9530ea7693c3702d : SUCCESS in 16m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/93c6fb2426dd47bcb6580d39d5a20b10 : SUCCESS in 2h 01m 59s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cd24e745ceb45ab92487f45161b1cc2 : SUCCESS in 1h 35m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1fd04f25bf08465582196cdaa3b20713 : SUCCESS in 19m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47efee894cf44db1a00aa2a097dfad41 : SUCCESS in 14m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/cbe26020e7d14360ac4be35926a3d37a : SUCCESS in 41m 56s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3330466b43964979ba86dfebb2a4a4b9 : SUCCESS in 33m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da6a06e4e61841d98555363e9f810cf8 : SUCCESS in 27m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d2b85e9fe7764e8ea0fe7a2da70e4772 : SUCCESS in 14m 55s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1bddf377151e4ad79d20654096eccc7f : SUCCESS in 17m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52ba1dd2097c4246a3dcf405251eaf50 : SUCCESS in 10m 57s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/60129b178cd44a8f85fd8dc1c24ccbeb : SUCCESS in 10m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/69779a40b7554ca2a4c46807bec053f6 : SUCCESS in 14m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0be598e3d56d424c89e2a5401f35801c : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":4},{"id":"53ecf9a4c59122a947735434946894ead5281c74","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-14 09:44:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5d84eb21d51655a47eed31967b9ded427578ba0c","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-01-14 11:25:30.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"56c645b30988694e81edbf71a24b45e192d7debd","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-14 12:25:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e705a3aade87c49ee7def40200708e6d6e88d3e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 12:25:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"385922210e05f6f2048c1ce40231eca259596a75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 14:03:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c87b778892064eb0a692460864a5c5ee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34b55d6b1983448294d438041ef34c81 : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a4cd8967c9143e2be28676bc41c4587 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f8bbb493754e19aa147a554e617e04 : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36bb9d3527b14d14b03bfc5a4debc815 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10e34e0de55642c987ef9c9594bc8e9a : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da6e2c147ad74daf815d8f623e7c9ba1 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b125a36ff24460f992812ad8ba8d229 : SUCCESS in 9m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/08b892bac0d847d1beec8a56ae25ebe7 : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89981f5b7bc34e028f7be81992524fd6 : SUCCESS in 33m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cb2e1e335f7b47a4b701d4d7bc3fb2d2 : SUCCESS in 1h 05m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1669561e9504435c9ffd61920871804d : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc4a5b58518d44f2814a83bfaaa4a412 : SUCCESS in 13m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/208c3ce1826d4097b9281d02d1cca531 : SUCCESS in 1h 10m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b195b491ff3d46ad9a572909f949ab15 : SUCCESS in 1h 36m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1ca889b629fd4ac3a79f26bd16b078a1 : SUCCESS in 18m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7171e176b8a54809a1841fa426f9ee06 : SUCCESS in 15m 27s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3ca6347aef3e4e8d849871893c88883a : SUCCESS in 36m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4db860bf6ecd4350881f92fc0ebf2704 : SUCCESS in 31m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/817a4b8a346c465eb89d33be70d90fa1 : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a81fa25452404e44a4cd0199dceafe25 : SUCCESS in 13m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/748fe13a9add4536a6f30814ff820933 : SUCCESS in 15m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57442901ef4646859c76023b0f6c314f : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":4},{"id":"e7822c1c82f923ff98e582e32e6d1c63de521382","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 14:03:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e9d96f5d28dd98325abdfee944044f8394b3fbae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 14:04:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a82de380d3c41a6ba8238db46a8bf5d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2662530bdf148e4920b8671a0c31e0f : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e237612fba94b5781050bedc53a6078 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4},{"id":"d732cdee798e2ace904e346782f9941c0f8298e8","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-01-17 13:36:37.000000000","message":"Patch Set 4: Backport-Candidate+1","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0137015e6cd7b6486e2a80f38612e1aa40ab1031","revisions":{"50e26af6c1604984263dd0ba53ffbac38e20e0cb":{"kind":"REWORK","_number":1,"created":"2025-01-08 13:42:19.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/58/938658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/938658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/938658/1"}}},"commit":{"parents":[{"commit":"a5a8e513b2daab8752fdd6ae3c6cf303c8e3d6f1","subject":"Merge \"Fix overcloud deprovision hosts list in prompt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5a8e513b2daab8752fdd6ae3c6cf303c8e3d6f1"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-01-08 13:34:18.000000000","tz":0},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-01-08 13:34:18.000000000","tz":0},"subject":"infra-vm: efficiently handle `hostvars` during deploy and destroy","message":"infra-vm: efficiently handle `hostvars` during deploy and destroy\n\nPass only the required variables from `hostvars` for a given infra vm to\nsignificantly reduce wait time during `kayobe infra vm provision` and\n`kayobe infra vm deprovision`. Time to create and destroy a VM now take\na minute or less.\n\n```\nWednesday 08 January 2025  11:24:24 +0000 (0:00:52.188)       0:01:38.044 *****\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nWait for SSH access to the infra VM  52.19s\njriguera.configdrive : Install required packages to create the images  5.35s\nstackhpc.libvirtvm : Ensure qemuimg and ovmf is installed  5.21s\nstackhpc.libvirtvm : Ensure the VM disk volumes exist  4.99s\nstackhpc.libvirtvm : Ensure the VM is running and started at boot  2.66s\njriguera.configdrive : Create network_data.json  2.03s\nGathering Facts  2.00s\ninfravms : [vm1] Stat image files  1.82s\njriguera.configdrive : Create network_info.json  1.68s\njriguera.configdrive : Create meta_data.json  1.62s\njriguera.configdrive : Setup resolver file resolv.conf  1.02s\nstackhpc.libvirtvm : Detect the KVM emulator binary path  1.02s\nstackhpc.libvirtvm : Ensure remote images are downloaded  0.99s\njriguera.configdrive : Create configdrive metadata folders  0.98s\njriguera.configdrive : Move files to destination from temporary folder  0.97s\njriguera.configdrive : Setup all defined devices  0.97s\nstackhpc.libvirtvm : Ensure the VM is defined  0.89s\njriguera.configdrive : Setup network/interfaces for Debian  0.84s\ninfravms : Ensure the image cache directory exists  0.79s\ninfravms : list all VMs on hypervisor  0.74s\n```\n\n```\nWednesday 08 January 2025  13:39:43 +0000 (0:00:00.781)       0:00:14.668 *****\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nstackhpc.libvirtvm : Ensure qemuimg and ovmf is installed  5.42s\nGathering Facts  2.64s\nstackhpc.libvirtvm : Ensure the VM volumes do not exist  1.31s\nstackhpc.libvirtvm : Ensure the VM is absent  0.90s\nRemove host key from known hosts  0.78s\ninfravms : list all VMs on hypervisor  0.72s\nstackhpc.libvirtvm : Ensure the VM is undefined  0.64s\ninfravms : Ensure the image cache directory exists  0.62s\nstackhpc.libvirtvm : Check the VM\u0027s status  0.47s\nstackhpc.libvirtvm : ansible.builtin.include_tasks  0.20s\nstackhpc.libvirtvm : include_tasks  0.13s\nstackhpc.libvirtvm : include_tasks  0.13s\nstackhpc.libvirtvm : include_tasks  0.12s\nstackhpc.libvirtvm : include_tasks  0.12s\nstackhpc.libvirtvm : Gather os specific variables  0.11s\nGroup virtual machines by hypervisor  0.10s\ninfravms : Destroy infra VMs (loop)  0.10s\ninfravms : Ensure coreutils package is installed  0.05s\n```\n\nCloses-Bug: #2093255\nChange-Id: I15d17f45d15c04aced0a4c0fe97e1d798ff46799\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50e26af6c1604984263dd0ba53ffbac38e20e0cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50e26af6c1604984263dd0ba53ffbac38e20e0cb"}]},"branch":"refs/heads/master"},"c222adc26309c0a169bc14fa57e061c846bf9ac8":{"kind":"REWORK","_number":2,"created":"2025-01-08 14:02:04.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/58/938658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/938658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/938658/2"}}},"commit":{"parents":[{"commit":"a5a8e513b2daab8752fdd6ae3c6cf303c8e3d6f1","subject":"Merge \"Fix overcloud deprovision hosts list in prompt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5a8e513b2daab8752fdd6ae3c6cf303c8e3d6f1"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-01-08 13:34:18.000000000","tz":0},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-01-08 14:01:35.000000000","tz":0},"subject":"infra-vm: efficiently handle `hostvars` during deploy and destroy","message":"infra-vm: efficiently handle `hostvars` during deploy and destroy\n\nPass only the required variables from `hostvars` for a given infra vm to\nsignificantly reduce wait time during `kayobe infra vm provision` and\n`kayobe infra vm deprovision`. Time to create and destroy a VM now take\na minute or less.\n\n```\nWednesday 08 January 2025  11:24:24 +0000 (0:00:52.188)       0:01:38.044 *****\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nWait for SSH access to the infra VM  52.19s\njriguera.configdrive : Install required packages to create the images  5.35s\nstackhpc.libvirtvm : Ensure qemuimg and ovmf is installed  5.21s\nstackhpc.libvirtvm : Ensure the VM disk volumes exist  4.99s\nstackhpc.libvirtvm : Ensure the VM is running and started at boot  2.66s\njriguera.configdrive : Create network_data.json  2.03s\nGathering Facts  2.00s\ninfravms : [vm1] Stat image files  1.82s\njriguera.configdrive : Create network_info.json  1.68s\njriguera.configdrive : Create meta_data.json  1.62s\njriguera.configdrive : Setup resolver file resolv.conf  1.02s\nstackhpc.libvirtvm : Detect the KVM emulator binary path  1.02s\nstackhpc.libvirtvm : Ensure remote images are downloaded  0.99s\njriguera.configdrive : Create configdrive metadata folders  0.98s\njriguera.configdrive : Move files to destination from temporary folder  0.97s\njriguera.configdrive : Setup all defined devices  0.97s\nstackhpc.libvirtvm : Ensure the VM is defined  0.89s\njriguera.configdrive : Setup network/interfaces for Debian  0.84s\ninfravms : Ensure the image cache directory exists  0.79s\ninfravms : list all VMs on hypervisor  0.74s\n```\n\n```\nWednesday 08 January 2025  13:39:43 +0000 (0:00:00.781)       0:00:14.668 *****\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nstackhpc.libvirtvm : Ensure qemuimg and ovmf is installed  5.42s\nGathering Facts  2.64s\nstackhpc.libvirtvm : Ensure the VM volumes do not exist  1.31s\nstackhpc.libvirtvm : Ensure the VM is absent  0.90s\nRemove host key from known hosts  0.78s\ninfravms : list all VMs on hypervisor  0.72s\nstackhpc.libvirtvm : Ensure the VM is undefined  0.64s\ninfravms : Ensure the image cache directory exists  0.62s\nstackhpc.libvirtvm : Check the VM\u0027s status  0.47s\nstackhpc.libvirtvm : ansible.builtin.include_tasks  0.20s\nstackhpc.libvirtvm : include_tasks  0.13s\nstackhpc.libvirtvm : include_tasks  0.13s\nstackhpc.libvirtvm : include_tasks  0.12s\nstackhpc.libvirtvm : include_tasks  0.12s\nstackhpc.libvirtvm : Gather os specific variables  0.11s\nGroup virtual machines by hypervisor  0.10s\ninfravms : Destroy infra VMs (loop)  0.10s\ninfravms : Ensure coreutils package is installed  0.05s\n```\n\nCloses-Bug: #2093255\nChange-Id: I15d17f45d15c04aced0a4c0fe97e1d798ff46799\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c222adc26309c0a169bc14fa57e061c846bf9ac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c222adc26309c0a169bc14fa57e061c846bf9ac8"}]},"branch":"refs/heads/master"},"e1ceac95149ac1a9390f267d17f63c58475ccc80":{"kind":"REWORK","_number":3,"created":"2025-01-09 00:10:42.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/58/938658/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/938658/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/938658/3"}}},"commit":{"parents":[{"commit":"a5a8e513b2daab8752fdd6ae3c6cf303c8e3d6f1","subject":"Merge \"Fix overcloud deprovision hosts list in prompt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5a8e513b2daab8752fdd6ae3c6cf303c8e3d6f1"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-01-08 13:34:18.000000000","tz":0},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-01-09 00:08:01.000000000","tz":0},"subject":"infra-vm: efficiently handle `hostvars` during deploy and destroy","message":"infra-vm: efficiently handle `hostvars` during deploy and destroy\n\nPass only the required variables from `hostvars` for a given infra vm to\nsignificantly reduce wait time during `kayobe infra vm provision` and\n`kayobe infra vm deprovision`. Time to create and destroy a VM now take\na minute or less.\n\n```\nWednesday 08 January 2025  11:24:24 +0000 (0:00:52.188)       0:01:38.044 *****\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nWait for SSH access to the infra VM  52.19s\njriguera.configdrive : Install required packages to create the images  5.35s\nstackhpc.libvirtvm : Ensure qemuimg and ovmf is installed  5.21s\nstackhpc.libvirtvm : Ensure the VM disk volumes exist  4.99s\nstackhpc.libvirtvm : Ensure the VM is running and started at boot  2.66s\njriguera.configdrive : Create network_data.json  2.03s\nGathering Facts  2.00s\ninfravms : [vm1] Stat image files  1.82s\njriguera.configdrive : Create network_info.json  1.68s\njriguera.configdrive : Create meta_data.json  1.62s\njriguera.configdrive : Setup resolver file resolv.conf  1.02s\nstackhpc.libvirtvm : Detect the KVM emulator binary path  1.02s\nstackhpc.libvirtvm : Ensure remote images are downloaded  0.99s\njriguera.configdrive : Create configdrive metadata folders  0.98s\njriguera.configdrive : Move files to destination from temporary folder  0.97s\njriguera.configdrive : Setup all defined devices  0.97s\nstackhpc.libvirtvm : Ensure the VM is defined  0.89s\njriguera.configdrive : Setup network/interfaces for Debian  0.84s\ninfravms : Ensure the image cache directory exists  0.79s\ninfravms : list all VMs on hypervisor  0.74s\n```\n\n```\nWednesday 08 January 2025  13:39:43 +0000 (0:00:00.781)       0:00:14.668 *****\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\nstackhpc.libvirtvm : Ensure qemuimg and ovmf is installed  5.42s\nGathering Facts  2.64s\nstackhpc.libvirtvm : Ensure the VM volumes do not exist  1.31s\nstackhpc.libvirtvm : Ensure the VM is absent  0.90s\nRemove host key from known hosts  0.78s\ninfravms : list all VMs on hypervisor  0.72s\nstackhpc.libvirtvm : Ensure the VM is undefined  0.64s\ninfravms : Ensure the image cache directory exists  0.62s\nstackhpc.libvirtvm : Check the VM\u0027s status  0.47s\nstackhpc.libvirtvm : ansible.builtin.include_tasks  0.20s\nstackhpc.libvirtvm : include_tasks  0.13s\nstackhpc.libvirtvm : include_tasks  0.13s\nstackhpc.libvirtvm : include_tasks  0.12s\nstackhpc.libvirtvm : include_tasks  0.12s\nstackhpc.libvirtvm : Gather os specific variables  0.11s\nGroup virtual machines by hypervisor  0.10s\ninfravms : Destroy infra VMs (loop)  0.10s\ninfravms : Ensure coreutils package is installed  0.05s\n```\n\nCloses-Bug: #2093255\nChange-Id: I15d17f45d15c04aced0a4c0fe97e1d798ff46799\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1ceac95149ac1a9390f267d17f63c58475ccc80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1ceac95149ac1a9390f267d17f63c58475ccc80"}]},"branch":"refs/heads/master"},"0137015e6cd7b6486e2a80f38612e1aa40ab1031":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-01-10 13:48:11.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/58/938658/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/938658/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/938658/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/938658/4"}}},"commit":{"parents":[{"commit":"a5a8e513b2daab8752fdd6ae3c6cf303c8e3d6f1","subject":"Merge \"Fix overcloud deprovision hosts list in prompt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5a8e513b2daab8752fdd6ae3c6cf303c8e3d6f1"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-01-08 13:34:18.000000000","tz":0},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-01-10 13:44:40.000000000","tz":0},"subject":"infra-vm: efficiently handle `hostvars` during deploy and destroy","message":"infra-vm: efficiently handle `hostvars` during deploy and destroy\n\nPass only the required variables from `hostvars` for a given infra vm to\nsignificantly reduce wait time during `kayobe infra vm provision` and\n`kayobe infra vm deprovision`. Time to create and destroy a VM now take\na minute or less.\n\nCloses-Bug: #2093255\nChange-Id: I15d17f45d15c04aced0a4c0fe97e1d798ff46799\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0137015e6cd7b6486e2a80f38612e1aa40ab1031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0137015e6cd7b6486e2a80f38612e1aa40ab1031"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
