)]}'
{"id":"openstack%2Fkayobe~938792","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~I4da3a3739d57817ffbd094caea0fc80c3ad8eff8","project":"openstack/kayobe","branch":"stable/2024.1","topic":"set-driver-info-stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-01-10 09:10:04.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I4da3a3739d57817ffbd094caea0fc80c3ad8eff8","subject":"Get the list of ironic nodes - use correct scope","status":"MERGED","created":"2025-01-09 12:00:52.000000000","updated":"2025-01-10 11:10:55.000000000","submitted":"2025-01-10 11:09:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938792-set-driver-info-stable/2024.1","cherry_pick_of_change":930691,"cherry_pick_of_patch_set":11,"meta_rev_id":"f76bc249b4e67256435b32294ad70f0299ffdb82","_number":938792,"virtual_id_number":938792,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 11:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"all":[{"value":1,"date":"2025-01-10 07:39:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-01-09 16:03:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-10 09:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-01-09 16:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-10 09:10:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-09 12:00:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"state":"CC"},{"updated":"2025-01-09 12:53:03.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-01-09 13:21:02.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"state":"REVIEWER"},{"updated":"2025-01-09 13:43:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 07:39:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"7e269778af04b6c21012538b0e9fd4af77781a89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-09 12:00:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"35e14c58e0a88df8dd5f70564ece43ffd117207a","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-09 12:53:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5ef69df426700da5195f2722a34b373e5e135b5b","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2025-01-09 13:21:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"868886e86414460710ad348162203ad6fa25e588","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 13:43:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/046bab41ea0147f9b164563555f49566\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60a466afcd4c4ca4abe694921e14cb7e : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf8d9501752c4f06a2b6537c8baac3d0 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18af329d3ac5460a8770d3748dcadfc9 : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dfdcbce4a0c94ac49f89107b39d8607f : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/15524a5301e44265bbce8d4fae1a2cf7 : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/28afae4561e14c228621e916e9ab58cc : SUCCESS in 10m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/05291754538347ecb93da37bfcc1ea87 : SUCCESS in 1h 06m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c7ee4ce8ed244f9492dcd32efb3d7059 : SUCCESS in 55m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1369b1def18648248d22309a97b9f8aa : SUCCESS in 46m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3b7be5232bca4404a8f5103118dba775 : SUCCESS in 53m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3e928646032d47a9818c3b901a02b0fe : SUCCESS in 21m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/35184cab13904f50a7844f9b334d3b9d : SUCCESS in 14m 44s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2402ae9ad204d6ea997e9ca2d6b14cb : SUCCESS in 14m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bf0e8a2b86bb443596d8186c4a5659d2 : SUCCESS in 1h 06m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f84180f4a6a49f4996aadb560faa450 : SUCCESS in 1h 41m 22s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/eaff58c14253486ca31cf99daed4e9a3 : RETRY_LIMIT in 5m 45s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a14072f5cf5490e9c6336cd01ffa8b5 : RETRY_LIMIT in 3m 11s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b0abaf03e7984c61b3e976238e09e5e9 : SUCCESS in 12m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3f6c382813e4fd68014d2e423347b15 : SUCCESS in 14m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c28fd36c05cd48beb8f7672b80f60b68 : SUCCESS in 37m 50s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/24625607c47e4af083ad792813ffde84 : SUCCESS in 29m 39s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ced203939f6f4b5780c8b9a18b2744ca : SUCCESS in 14m 30s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6294b3a6e87f475797538082acac74a8 : RETRY_LIMIT in 5m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d9d9f0d754ad477ab6769239fc3b8cfa : RETRY_LIMIT in 4m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/efbf26c57dc3477490219522983e4a48 : SUCCESS in 15m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/54b53e88107f440faaf2aefa05466c12 : SUCCESS in 16m 36s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdbcffc01a7d4ce99573c3e472b6ba74 : SUCCESS in 9m 29s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/95145f94f8cd40b79c0d5493c7b7a3b8 : SUCCESS in 9m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/72eb0ebb13f44c2c8237a2acf71a8427 : SUCCESS in 21m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fef7cb659ce4c5ba572f9fcdc74f919 : SUCCESS in 15m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbc546f054694a52925b6a07deb21238 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c198149e157c45348e3abf5a1cd1ee9b : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02974fbb2bf64ae4b26bea93876fa207 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50d770fd0adb478b92b0f575adbbb388 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e1a7e03df042f75d16fd75d6a2e72f7a73961169","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-09 13:58:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"057af15f1d508c9ad4a742cb7c794a283cbe01c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-09 16:03:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e00ee10ea89ee6de0806707890d63518697d78c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 18:25:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e87d8fcb7684dd4a417bf3cf28c0673\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/226a4443052e431fb2d8d78da01c021f : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34b7fc9637147e58df559215d389b17 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad3608da7bd24f4aac9b68057ba96c98 : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84d997cefe7a443cb4f76d7fec4847c5 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ddeff218daea4aefb2279d8a0b0eea53 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6edbfb2efacb4b5cbbbba629c36cfd5d : SUCCESS in 10m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/68baf640825641609646319dc83375e7 : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/573351b4f74b438c84e75dd612eb6aff : SUCCESS in 32m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ac68a765bc04d07bd1373a5144689b1 : SUCCESS in 52m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/11ca3b6f346f4de1adfe782af42fa89c : SUCCESS in 38m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2299ad598cd6462fb8a5f575e160d29b : SUCCESS in 12m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/733638d3bf9347b199be76b5a4d5be51 : SUCCESS in 10m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f774faab89a456ea2508a11d0134331 : SUCCESS in 8m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/36d34d45cbd24045980db52a15108c65 : SUCCESS in 1h 53m 09s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54f01676232a43969161cd25d65faf0b : SUCCESS in 1h 34m 40s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d98ac15710e140df865a9e2ab3806161 : SUCCESS in 2h 16m 45s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3708a79f25844469d045fcd8fdbf70f : SUCCESS in 1h 57m 55s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/825d42bc91224e1facb47dc504e79472 : SUCCESS in 18m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/584aea420ec34387bc776795037f8142 : SUCCESS in 12m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b603b9bdfaf0445993c005d62bca855d : SUCCESS in 1h 00m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b62cae2af31e49da8ab5d6d61481b669 : SUCCESS in 29m 03s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e52a91f8cfd04190ac23a16f09b014b1 : SUCCESS in 22m 29s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ee10302805a64fbeaa504fc1726e3b4d : SUCCESS in 33m 21s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d9593159e7f492d9e9d432ce03276c0 : SUCCESS in 29m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5c689f8730dc4ec79947795acc8bca1a : SUCCESS in 13m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/fba99908bb3840c9a1ad261240b277a7 : SUCCESS in 16m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/321a87bb1d2b4209a4e39018c57178ab : SUCCESS in 8m 42s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/1a1a4f00b82b4a83a4249ee64a60f18c : SUCCESS in 12m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/76ea364e0c114eea9154d91ca57cf990 : SUCCESS in 19m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14e6e1db21104756961e5b4241f0a625 : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64cb01a39ddb4051850660b8a3bb72c7 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ef689fae72743a7a85675bb2a4330a6 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54658f035d7646b6b0ba43d5f197228a : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0bbdb7c80284ac8976835fe788043c8 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":2},{"id":"308ab4ca221edafd7b1a98b5ada64fb77a37ddd1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-10 07:39:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fedf979bd8bd1615eb74076651a04c480eb3b2a8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-10 09:10:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e0d8d136d10892169121e53df5347d1aacf533de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 09:10:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c344b78a6cdbf14f688136a085552ec6d2fe798a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 11:09:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7082f4a082a84557ad704af4074d5977\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb37f13876b144fa906d0f4b4f2386e3 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/004e18e9240242d5a2b602b432ca1c57 : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ac8f95fd3503459a8d3f2fd252c8050b : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2e2ebb9873b6417fb7882064793a53ef : SUCCESS in 10m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b109085828174cb190f0b040bca5a6e7 : SUCCESS in 8m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/131db5f231e24c409a8544d54ef6de70 : SUCCESS in 58m 34s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6784a30a642f48639ca9cafe4ea5db72 : SUCCESS in 47m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3e45d754333343f8ac2ae5bf7c21d0c6 : SUCCESS in 1h 13m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/255ee837f3cf4f2b9ecede67a14ce5ac : SUCCESS in 18m 09s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95acb23655bf4ed188825a184df57523 : SUCCESS in 14m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ae3adab61c174af19e2f0b94f5e8238d : SUCCESS in 1h 54m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd3e987e73154ec6bdae8668787f4093 : SUCCESS in 1h 38m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c4d654c62f1f498fbee36437d3c42674 : SUCCESS in 17m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2f7839813b5457b8acbbc6849132359 : SUCCESS in 12m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5185e59021f547fb979f5fa9e94adfc9 : SUCCESS in 30m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c0fcc3ee5e148409de5ee5f88f27280 : SUCCESS in 24m 15s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/76e7a1dedfb747709613acb1a444d645 : SUCCESS in 35m 31s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d19d87e94c274711875e408fff8a833c : SUCCESS in 26m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4dd743a431bd4f0d8a1b43bcff91c979 : SUCCESS in 14m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fff225d8bb694d4b8ae3c00b923b967b : SUCCESS in 10m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/37179e36f4cf4221ba4185ea43b99118 : SUCCESS in 21m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67727c7d925a4780bff928b1539dd416 : SUCCESS in 17m 22s","accounts_in_message":[],"_revision_number":2},{"id":"4955bc66d9023c6611fad7fb37ac74839c411666","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 11:09:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f76bc249b4e67256435b32294ad70f0299ffdb82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 11:10:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3160057e86334d458f370f0982b40152\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b16d6ca0163b4132a1385fb63db6a97d : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74937b8ffe9842d5acb2990cee346e41 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"606518baae3bcf8e0f0c111f747bdf9c345a7ce2","revisions":{"b6acee1518d05cb443b8281a2b24496167002a0c":{"kind":"REWORK","_number":1,"created":"2025-01-09 12:00:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/92/938792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/938792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/938792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/938792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/938792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/938792/1"}}},"commit":{"parents":[{"commit":"d20514b7e24505e891da7ec6ea731c43fb96d7d7","subject":"CI: test database backup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d20514b7e24505e891da7ec6ea731c43fb96d7d7"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-09 12:00:52.000000000","tz":0},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n(cherry picked from commit 479b4cdf73fd20d3a957862784481593019e8b8d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b6acee1518d05cb443b8281a2b24496167002a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b6acee1518d05cb443b8281a2b24496167002a0c"}]},"branch":"refs/heads/stable/2024.1"},"606518baae3bcf8e0f0c111f747bdf9c345a7ce2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-01-09 16:03:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/92/938792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/938792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/938792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/938792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/938792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/938792/2"}}},"commit":{"parents":[{"commit":"9da82a952ddfb585aeefc5d4d79933738591fe4b","subject":"Fix slurp upgrades due to 2023.1 branch name change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9da82a952ddfb585aeefc5d4d79933738591fe4b"}]}],"author":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2024-09-27 09:51:00.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-09 16:03:18.000000000","tz":0},"subject":"Get the list of ironic nodes - use correct scope","message":"Get the list of ironic nodes - use correct scope\n\nUse correct system scope - all - while getting the list of ironic\nnodes in ipa-images role.\n\nCloses-Bug: #2083014\nChange-Id: I4da3a3739d57817ffbd094caea0fc80c3ad8eff8\n(cherry picked from commit 479b4cdf73fd20d3a957862784481593019e8b8d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/606518baae3bcf8e0f0c111f747bdf9c345a7ce2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/606518baae3bcf8e0f0c111f747bdf9c345a7ce2"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
