)]}'
{"id":"openstack%2Fkayobe~939780","triplet_id":"openstack%2Fkayobe~master~I539a29c77a0165d43ca2ce359105c5460f450cd8","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-01-24 09:45:21.000000000","reason":"removed on reply"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-01-24 19:48:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I539a29c77a0165d43ca2ce359105c5460f450cd8","subject":"Fix Depends-On","status":"MERGED","created":"2025-01-22 10:56:12.000000000","updated":"2025-01-24 19:49:37.000000000","submitted":"2025-01-24 19:48:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939780","meta_rev_id":"266ec1973033c1798f07f3771915089e2869e347","_number":939780,"virtual_id_number":939780,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-24 19:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-01-24 09:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-24 09:45:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-01-24 09:56:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-01-24 09:39:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-22 12:51:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-24 08:39:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-01-24 09:45:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-01-24 09:56:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2af84a9ab00468a698224e4e08cd2577a4f06d18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-22 10:56:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f54e8281ba016d17b2e49f2d9674619e5de4db2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-22 10:57:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"915f1a1647c7e9c1dca98889c8f53ac9e30fd569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 12:51:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a01dd0e3eb8040c5a2b28b330355493a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1df1e3d108b04f46ba90739933cc0600 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c84b7c91170492d8158216b31af0f9f : SUCCESS in 5m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aed6f4722211430295a515a52d66b7ab : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00e897d70b74f67a5731310e1344a54 : SUCCESS in 7m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8aae082a69bb4b3ca706cb2df13c4982 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e3903cc252a45ceba07fe3c8a11109d : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a95d2a1fb5ac4319b4e41b92f792d177 : SUCCESS in 10m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e5cae4348c9f4517beea6cbaf189bac9 : SUCCESS in 47m 50s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/65efe7f0410b48e780397aa70afbcc20 : SUCCESS in 1h 00m 25s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ad47cb81c294fccbeaa988489d1748d : SUCCESS in 55m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ffbfcc8674c24eed9d7084dbb18520e1 : SUCCESS in 1h 05m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/93f82c473acb4f0eb325a205bd5e11d5 : SUCCESS in 17m 13s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/563a7d2f0b8148bbb42bff330ee05a90 : FAILURE in 11m 56s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4a4035dd1c0441aa7ca28b3d74bf674 : SUCCESS in 13m 43s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/da16aaac04364c5b98262ad102b82e09 : SUCCESS in 1h 53m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f54911f46694ff48a0d4560f580d558 : SUCCESS in 1h 06m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7dd176c3cfe44a31b7662f6fe957c7a3 : SUCCESS in 19m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a65daee13cd843868f8931144831b17a : SUCCESS in 15m 27s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/33f83676e67b4f90954e78ccb0612a08 : SUCCESS in 27m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2bd699368cdf4a0f9a5a65e840e289ec : SUCCESS in 31m 42s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bacda3ac978a43d9abc453d3e2f3e594 : SUCCESS in 27m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e2ea04edf9f34f3f93c0120207c2a63d : SUCCESS in 10m 10s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4e15bf65e4f44144be3407a2a1df4a6b : SUCCESS in 14m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd64737159fc4ac6be2084b992f1badc : SUCCESS in 13m 03s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/8c731ea8872c4de0a8952ce6602336bf : SUCCESS in 13m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/637fd66da5aa423081c632a4448088ef : SUCCESS in 14m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a12614d553d2482e8099b7c164e6bfcc : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":2},{"id":"fcd9b02e34143effe41b7c5ca48da8b7e009c589","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-24 08:39:33.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"06d53ab337d321adeb8cc2b8ccc6284d69cd03b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-24 09:31:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6092eb0f19f8409055a5bb990fc2a8670954e53c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-24 09:31:42.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fd8cad3252a3323cc919e5d56f9f1eb1aef8a9f7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-24 09:35:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b4d22d2a0db60de5ba3c31ca15b3e120d0024ea8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-24 09:38:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2d1e305067b6001fa77a637e25d51240ab0fd924","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-24 09:39:41.000000000","message":"Patch Set 4: Backport-Candidate+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"723bea09baa557d290551841d3f51c870419c755","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-24 09:45:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d182bcf242df649965892abd8789478a40fd77fc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-24 09:56:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a13eb838f42048339afc0c21eab6abef82084ab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 12:09:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2719e22458254a9ba2ea8fbce3645b53\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebf43022973c4361a1f3275266f05ff9 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ccef106b0c446c982bea674e9e1e32 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1214df4fc944f7086e9ed2037ae83de : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09fcc7004f0f4e7e81f42ef1f39f3e36 : SUCCESS in 7m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a56b5cf88b16441f83548f89c2006472 : SUCCESS in 13m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3d8c64f74a8e425fa2c6096fabf93118 : SUCCESS in 20m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f49ad05624646a088a5c738af895401 : SUCCESS in 18m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/14e82fc2a6f04cd7840ba8666ba4ee8a : SUCCESS in 57m 58s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d9bbd31c68a94d0eb9e6a51990083e08 : SUCCESS in 59m 54s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bbc9f4c242d4b988bfcb3e328866ff9 : SUCCESS in 58m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6e423d26662748dda01ed16d14320bd7 : SUCCESS in 1h 04m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dde73af9515743adb33dfd66c07396c8 : SUCCESS in 17m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/209b58dd6c754a778dec28b03e86040e : SUCCESS in 18m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea39e55823ee450a88a83c9fdc63a51b : SUCCESS in 14m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4f5b5386698348d7a4db7b99f49ee12d : SUCCESS in 2h 07m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75a1cc52eaea48138d1e42fbc0cd585d : SUCCESS in 1h 37m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4f3fcc13919840708f44bd7640c3c31e : SUCCESS in 10m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8e38fc89dac49059e6e73d87fd9de10 : SUCCESS in 16m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6a72ab0d028f4808a700c0f7bfd72146 : SUCCESS in 38m 49s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dff0ce464ffd4a798899aaf9872a02c3 : SUCCESS in 29m 38s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee269f1625fa4507882fd66dfed39254 : SUCCESS in 30m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1fc762934235429ba1a8d6abd24ccaa7 : SUCCESS in 9m 47s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/5d9ccf8ff6ec4caf8cb677f165be2618 : SUCCESS in 13m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9772ab23b04b46e486a88534e32c8515 : SUCCESS in 11m 07s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/2bfd08f7e71147b0ab38a0843e5d42e1 : SUCCESS in 13m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea1cf89f188a46b6a8e227aa86f8f7eb : SUCCESS in 12m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/493a6b693e9944688867cbffcf67167d : SUCCESS in 11m 27s","accounts_in_message":[],"_revision_number":4},{"id":"b96d3320e9a631b7169fe1dc1e8f47a7e94e0631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 12:10:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"90b988f9eb637be9d2b4d5e0884c746ddaf69d4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 15:19:37.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd7ac3bd0dfc46178b5563fb9b7c6e10\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b94be545df4ca099def2ca04a69777 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b910f1550103418594a9cf1bce87dae7 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b60ccd6bb9d4076b62d2c2a71a9192f : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/54b5f524d6ac4898ba7bf4f5c00a0496 : SUCCESS in 3m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e2993193f115441fa09f2d7fcc864124 : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/953f7759fd944b4fa351a378194632ad : SUCCESS in 6m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/49ef0e11dfeb48d6be961e6e95695a13 : SUCCESS in 58m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18cf426b0e6d4c5ba5a05c243aaed20a : SUCCESS in 58m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2f6e892c5eae4efcb02692f490685a29 : SUCCESS in 1h 04m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/12a4a726922b4f0299f9a6ec1a8e8bbc : SUCCESS in 17m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7f574eb6cd44cf480ba4019c1f1a135 : SUCCESS in 17m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/395890238681421ba26a37156db08b5a : SUCCESS in 1h 52m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e972cc1a1bcf437d8d4c1cc2c0ed4935 : TIMED_OUT in 3h 04m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e713e73ac3b34962bf113dc1a94ca13d : SUCCESS in 19m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/256fe44725e04cb5a655215bdb4bec24 : SUCCESS in 15m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1dfe4b3567344313bbcd8e11a10b0c03 : SUCCESS in 44m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65b2a1e9724b40d3b5164824d9664056 : SUCCESS in 30m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/64f19782e26e46be96430c11088f0794 : SUCCESS in 12m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2975e2c279684cc3b2ad4cb6c4d5b893 : SUCCESS in 13m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1ffab7172a5949829db3a527f306bcd6 : SUCCESS in 14m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/228b20138d874e0cb1b0c825984f0cdb : SUCCESS in 12m 39s","accounts_in_message":[],"_revision_number":4},{"id":"b635726cac529307ae14255d6987a99385ee7f34","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-24 15:24:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e8ec4315092e08c8ae708271ca7020603fb7d10d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 17:43:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fdf3964a9a44a79abcc7841f1ca30a4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5465bc76fae48f696daa82389b40c4d : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20f335a6336f421f891b693d8a00c8df : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec4249b6e21a46769ec5b5ccca6e464c : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/470ee202c0ac49719101fb2b6b24981d : SUCCESS in 8m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5d2c863432d34eac80ba783ad978b4a1 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd2fa28480f947caa6bfcbd332bb0044 : SUCCESS in 12m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5d4652d74154c7a82b780b4a1428896 : SUCCESS in 11m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/831148e9809e475e957ea6dc8ebd350d : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c69aa0a6934044b9b010b224eb05321b : SUCCESS in 1h 03m 30s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9dbd3ce2cda4497dab001feecf85c4d1 : SUCCESS in 54m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8d68b96ad4cf471c8f4c68a746eddd3b : SUCCESS in 1h 21m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dc30169acf154bed894e760f1c5c890f : SUCCESS in 21m 13s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/98bcb86bedab427d8e07f83ff4b2258c : SUCCESS in 20m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13ad4ff1613e475dbf4de4dc8620f100 : SUCCESS in 13m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fb987e020e03473b92381fd010e36777 : SUCCESS in 1h 55m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a962ff0805f8462b8a69d8305461d7eb : SUCCESS in 2h 02m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ecbd80caad974ddaacad6cfd75fa0f36 : SUCCESS in 18m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a808228575094933a12cdb95c36b73ae : SUCCESS in 18m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1b1eee3ebbc1411d89fc28115d495e0f : SUCCESS in 37m 35s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1c9e560ed34e47a494d17a541e1e1412 : SUCCESS in 30m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/595629d9ef17431fb763b6858b8476a1 : SUCCESS in 24m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/97a00f11f1734bdab7e0429bd5d7d977 : SUCCESS in 12m 39s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/776b208a511a480db8f13019f7e74fa4 : SUCCESS in 14m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9a6b6f7e94a4657846737700b9c13a0 : SUCCESS in 11m 22s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d253db65d26546a9a2cfd43ce52a84ab : SUCCESS in 13m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/93726249e2794671a8d11719db13a7fd : SUCCESS in 13m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a8b2f2702fd4afaa4d0f2cdc3940192 : SUCCESS in 11m 50s","accounts_in_message":[],"_revision_number":4},{"id":"60a2fd086836b4926ca266df50b76b9f1e0c7b0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 17:43:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d7f02b051fa557d7ca08a69f928ac7de1482c60a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 19:48:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a54910631b439682eeb13fc9c204d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f993841ad8945a0bbaa6368ea331517 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f9951f225fe43e28275666d4c563d02 : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c6d9fa647246c8bf2aeae449057006 : SUCCESS in 7m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1413468eee6c4189b4049a8f26a544e6 : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/924ab77ee12342d7bb3dfd25955e4d01 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dfa3834a18dd499786273a526d915fee : SUCCESS in 10m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/29f52f6b3e3f42e095f39318dcaef273 : SUCCESS in 59m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f05685533314445b90348e4daf073e9a : SUCCESS in 53m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3120202a768f4c288636c2b09d6354ec : SUCCESS in 1h 05m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/34974da9ecb04b8baf87ad56f4540d49 : SUCCESS in 18m 02s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a2f9734f2c74b859a9763a096f75ca5 : SUCCESS in 12m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1cf127a2aa2f4f4482836618ad6c7fbc : SUCCESS in 1h 52m 24s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ceb65393bb784788b380fb1749d9ade0 : SUCCESS in 1h 34m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/099488e2eadc4a03b45b297234fcc9ae : SUCCESS in 17m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4b9c2953a5d48278c9046a75c035e3e : SUCCESS in 15m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8183127260884159a0c9531e9d88b85d : SUCCESS in 31m 05s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5605fee5df844b5ae5b34cdb0c444b3 : SUCCESS in 25m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d052ba0b6ed6412b9509d12ee18bdae7 : SUCCESS in 14m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03e9cb44340e4326b654332c93af6f85 : SUCCESS in 13m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4641fa7382204b9eb475cae24e14378a : SUCCESS in 14m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/798113040cb4413693821b13a2339199 : SUCCESS in 12m 55s","accounts_in_message":[],"_revision_number":4},{"id":"57ec130b5aae09e271aaca058ca4a7a7e9af8f83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 19:48:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"266ec1973033c1798f07f3771915089e2869e347","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 19:49:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0569f9f6a3524d89987c03a9e2ae184a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/095a71d22fd245ad91e5161f4379e07d : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0b7d8015251d5f5dc7cba584a8e2ca86a46f758a","revisions":{"93ee05451feb47e40a2c06265bd30896b1934613":{"kind":"REWORK","_number":1,"created":"2025-01-22 10:56:12.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/80/939780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/939780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/939780/1"}}},"commit":{"parents":[{"commit":"1424835969053b8c4d6ea042647c23d074502635","subject":"Merge \"Install setuptools before creating virtualenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1424835969053b8c4d6ea042647c23d074502635"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-22 10:55:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-22 10:55:13.000000000","tz":0},"subject":"Fix Depends-On","message":"Fix Depends-On\n\nUse version checked out by Zuul so Depends-On works and only override\nversion for previous release.\n\nChange-Id: I539a29c77a0165d43ca2ce359105c5460f450cd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93ee05451feb47e40a2c06265bd30896b1934613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93ee05451feb47e40a2c06265bd30896b1934613"}]},"branch":"refs/heads/master"},"8cb8f116341477150d6288c1e1458fe4da1bd253":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-22 10:57:10.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/80/939780/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/939780/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/939780/2"}}},"commit":{"parents":[{"commit":"1424835969053b8c4d6ea042647c23d074502635","subject":"Merge \"Install setuptools before creating virtualenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1424835969053b8c4d6ea042647c23d074502635"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-22 10:55:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-22 10:56:30.000000000","tz":0},"subject":"Fix Depends-On","message":"Fix Depends-On\n\nUse version checked out by Zuul so Depends-On works and only override\nversion for previous release.\n\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\nChange-Id: I539a29c77a0165d43ca2ce359105c5460f450cd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8cb8f116341477150d6288c1e1458fe4da1bd253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8cb8f116341477150d6288c1e1458fe4da1bd253"}]},"branch":"refs/heads/master"},"47e985a27d54af372c1bd5600b123fda523cd8d6":{"kind":"REWORK","_number":3,"created":"2025-01-24 09:31:18.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/80/939780/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/939780/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/939780/3"}}},"commit":{"parents":[{"commit":"1424835969053b8c4d6ea042647c23d074502635","subject":"Merge \"Install setuptools before creating virtualenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1424835969053b8c4d6ea042647c23d074502635"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-22 10:55:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-24 09:30:23.000000000","tz":0},"subject":"Fix Depends-On","message":"Fix Depends-On\n\nUse version checked out by Zuul so Depends-On works and only override\nversion for previous release.\n\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\nChange-Id: I539a29c77a0165d43ca2ce359105c5460f450cd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47e985a27d54af372c1bd5600b123fda523cd8d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47e985a27d54af372c1bd5600b123fda523cd8d6"}]},"branch":"refs/heads/master"},"0b7d8015251d5f5dc7cba584a8e2ca86a46f758a":{"kind":"REWORK","_number":4,"created":"2025-01-24 09:38:23.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/80/939780/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/939780/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/939780/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/939780/4"}}},"commit":{"parents":[{"commit":"1424835969053b8c4d6ea042647c23d074502635","subject":"Merge \"Install setuptools before creating virtualenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1424835969053b8c4d6ea042647c23d074502635"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-22 10:55:13.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-01-24 09:38:16.000000000","tz":0},"subject":"Fix Depends-On","message":"Fix Depends-On\n\nUse version checked out by Zuul so Depends-On works and only override\nversion for previous release.\n\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\nChange-Id: I539a29c77a0165d43ca2ce359105c5460f450cd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b7d8015251d5f5dc7cba584a8e2ca86a46f758a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b7d8015251d5f5dc7cba584a8e2ca86a46f758a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
