)]}'
{"id":"openstack%2Fkayobe~940131","triplet_id":"openstack%2Fkayobe~master~I9fdd017f857e9f56fd4926a3314443348fe53fe1","project":"openstack/kayobe","branch":"master","topic":"kayobe-ci-previous-release","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-02-01 19:00:27.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I9fdd017f857e9f56fd4926a3314443348fe53fe1","subject":"CI: Set previous_release for Epoxy","status":"MERGED","created":"2025-01-24 13:30:02.000000000","updated":"2025-02-01 22:34:05.000000000","submitted":"2025-02-01 22:32:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940131-kayobe-ci-previous-release","meta_rev_id":"c46d3354fb6a3b20a8bade1bfe12c2524ae8b511","_number":940131,"virtual_id_number":940131,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-01 22:32:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2025-01-29 10:14:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-01-27 20:20:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-27 09:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2025-01-29 10:14:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-24 15:39:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-27 09:51:33.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-01-27 20:20:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-01-29 10:14:30.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"6c0d2160458d1a213e387e577f61c44a1d7b3aae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-24 13:30:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"566110b25e9ae7de2985fa6579dedea3992f416b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 15:39:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b9df0ad509a4114a9aecd6a8d42dc7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4f9fa8226ec46f3a16a67e7d8cb7a1f : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/188716b387dc491bbd1d97f37eae2411 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79f961956d3d484dae08d68b9ecfff71 : SUCCESS in 13m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01067a2836ce40dfbfc75d01395048fd : SUCCESS in 7m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8a42da7bc282469084c092e371775ccd : SUCCESS in 34m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/eea7fca3718a424eaced8314d2a782ae : SUCCESS in 58m 06s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0d8f999f7e743b7a00d6a57b6b0de61 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4af9874e821149599829d541182f9182 : TIMED_OUT in 2h 03m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6b4eaf71060e444e9087b79508a43623 : FAILURE in 10m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f4b0b405d1cd42d49726f1c8e659ab2e : SUCCESS in 17m 42s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9507463284c947998a39e740a6b1f35d : SUCCESS in 14m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/08eca23ef69e4b019f27a9c94125cc68 : SUCCESS in 1h 57m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6872c3c7216e4a2e8bcea65e8b012db7 : SUCCESS in 2h 05m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dd19e306d2f841bbac6ab71857c6a0be : SUCCESS in 20m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f7f0cb3975b4972bd1316efb066d905 : SUCCESS in 17m 27s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bd09fd47c46f474d92ac554dd7ce4ae0 : SUCCESS in 39m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/041cdf5bab204e55b08c4012a382941f : SUCCESS in 30m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0afbee5d949a4b13be70e24de8cb2da5 : SUCCESS in 32m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cfe11bdc94e4458cb1a7bd5a9fa36436 : SUCCESS in 12m 57s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/fc41a5124fda443f8fc9570cff37b8b0 : SUCCESS in 8m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b8d7aa6917c4b29bfcd3324e5390b54 : SUCCESS in 7m 12s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a9cbb19ac6d04da9859c1ebb1a5b56d8 : SUCCESS in 13m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c5fa7ade3bcd40468619c6e348bc4cb9 : SUCCESS in 18m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/937e1aebbdf74f0697f028a91358adaa : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d1545f8a216e2859742ed880f8ca107329bd217c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-24 20:53:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a09f69e11032c55bcb532280f117a700fa59387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-24 20:53:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f6d93d77efea8b66efb7f9950ea45543b8b022ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 22:47:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a30b9f13be5438c8b5bb1d154c75c57\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0458b29eff9e4721949f5e006231112f : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d8b40a89c74468da2758a6348e4da64 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f892053ccaf45398a23c9e75bd8f907 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec34d975e4f40d7b3e3a41f6207fe95 : SUCCESS in 7m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/147d54d94a7c4914bdf32d78957fee6a : SUCCESS in 1h 00m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/61789095e0c9468e87322b1ca74d75f7 : SUCCESS in 1h 10m 09s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33259983207842d4bdbf9d809ce22da3 : SUCCESS in 58m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ee2c6f5725184d7795ff376e17e97d48 : FAILURE in 1h 17m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/921192a6485d4f6cae02b7676240393c : SUCCESS in 18m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a0d3238fd304488c97029de45b758793 : SUCCESS in 18m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90afae0b309b4a329af92af1be010c9d : SUCCESS in 14m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/68f8322a37984986a91bcc8d0bf2d533 : SUCCESS in 1h 49m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a44da444b0a34b9096610a12f0dbd36e : SUCCESS in 1h 46m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/81863001bba34cc1aa7bfbf11734a5cd : SUCCESS in 22m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c8e27e4ce2145c7a82d1f6c3bf7a483 : SUCCESS in 14m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/23b2f190ce1641ff85c4a6443408549e : SUCCESS in 47m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c2edd7da332b44afafdc50342f6ab572 : SUCCESS in 28m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06424bbfbb4c427294fc3447b43f7347 : SUCCESS in 33m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/824f8947393a4ce4a4a78631e3cdcbf9 : SUCCESS in 13m 54s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/0dea7f3e163549918b8d7052fbe2baeb : SUCCESS in 11m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee064aa306a446e5bd768337b0bc29b9 : SUCCESS in 11m 53s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/e5caf6c186f345679c512db137069987 : SUCCESS in 12m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cc96df9ca13842809fb37781a209b5e0 : SUCCESS in 13m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/738659101a7c47e8a2b9759cbcbba705 : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":2},{"id":"78eb13079e9531838a46083f3a3c58394a849d7d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-25 21:17:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7cba4a4515a7d74a341d7053dc0c9be9bc1f06c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 23:18:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b96dac12383749e8a2778fb8af5bf9a5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22fa9087b5b7496eaec3a6fcb9eb061e : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eecb1b83e8c48228f40a239554db69c : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e56ee37042a84218bdddd4a2bc71b9c5 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf5b01df0374dea9370f443820ddc3c : SUCCESS in 6m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/644342e078b34e78ad80500f8424b2af : SUCCESS in 55m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/62de6921a09c439999c9f422d606c964 : SUCCESS in 57m 03s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4f00c00ec854ce39de350b4496e20c4 : SUCCESS in 55m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/00215e1a47054e1cabd82d4be60e8e06 : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/af07c1066cd74c1b89ebf8ed0b1f726a : SUCCESS in 17m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/15ac67e5dad84c9d9922159763dddf6a : SUCCESS in 18m 15s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4641c43fe644250884bb8b69405ed42 : SUCCESS in 14m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ab34f71174b34d5aa9879ac00628e067 : SUCCESS in 1h 00m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bc9785e7eef459eb389e07e2b570ea2 : SUCCESS in 2h 00m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b18e6e18f2334697a918580977b8bfde : SUCCESS in 13m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2aa1f792688463ca38374b01f105491 : SUCCESS in 19m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/50a23cfb7c344fe5ae7b97d7ceb6f3b4 : SUCCESS in 21m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0870d3310ea84da89a64a78aae2ceeb9 : SUCCESS in 23m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f68feb8be2f94b52899592be4905a3bf : SUCCESS in 31m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b57352976d404a2f82859c6db9c34a1e : SUCCESS in 8m 02s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/43ffda3b2093495799cc024e634a3217 : SUCCESS in 10m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be2ad053700f4ca19f7e440e0cb9a5c6 : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/e0271091ef664ad689c4739b9019db9a : SUCCESS in 7m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ed190750583a4282a004e03e2025e041 : SUCCESS in 10m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3c18f91361c429896903aca4aa63365 : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ed5fb322cea2ab892dacdb6c0b0827d21c13961b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-01-27 09:51:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1a8c0e88ee1686b9ad4c4272d1f67ee111d24d5d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-27 20:20:27.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9bf45638214ebbdf6b4a30f0966f1855f9927573","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-01-29 10:14:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6f17f1521ce48cb285d275dfc821c83f13bd2334","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 10:15:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fdb5fb392177ced35baaa63d55b27946a0e57ae8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 15:27:48.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88521983753e4b72a35e5273dcb75729\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa4b1a0bcfac42b38f371ce33574e3cf : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd4577eeaddc4c2dbcdce11a2ec3787e : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c42141c8b104453c97888e70d841fa55 : SUCCESS in 29m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c289eaafbfb549c098faf366a44dde47 : TIMED_OUT in 2h 03m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb1d57e5f3cd4b0c90ed8f72949e9d02 : TIMED_OUT in 2h 04m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9fda35880d064bb6aeca10050b314ce7 : TIMED_OUT in 2h 04m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b12487bf21124528817aa31676dffaff : SUCCESS in 21m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/659fe083124043d49bfb70bca38ca26f : SUCCESS in 14m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c29ccd396bb84d0bab554604988cd01c : SUCCESS in 2h 01m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e488215e41514753a2ae08e674f6e5a3 : SUCCESS in 1h 01m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2b5ae4e9d38143fb857815ec062a2d0c : SUCCESS in 17m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f367c711cf30499da840ce6f262b585b : SUCCESS in 17m 51s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dbaf4463dfa64592ba565564b4cbc074 : SUCCESS in 29m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10e6d2e39ba04639bacd1c163024f153 : SUCCESS in 27m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/61ece6eea58045d9b2cc061d55980b7c : SUCCESS in 15m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e5cf90ef4f84768b43c021b2c93edef : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0c439e51268b4bdcad68e8b4fdbd1a92 : SUCCESS in 13m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1202f07ba14b44aca965af379c18c381 : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":2},{"id":"43c670bf9434f93344e81771dce0a5c15b269ed8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-29 16:53:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19d4396156b7b34b3747ac4df0849be1f45c94e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 20:10:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0501b174d8904ecdab9a7d4893351966\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/423ebf028a474bf5af87ba6d5703d215 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/214480d26d464be9816a1ecb7c8783a9 : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31b72ee77424462d907022569c4e1b7c : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01fc03179fc842e0a94f38f832857c70 : SUCCESS in 11m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0d1da8f6346a4d1e91e11e1b9405911b : SUCCESS in 58m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cf86b2a1097e4039ba79627b9ed4e37f : SUCCESS in 1h 07m 24s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0af6704ab074a978fe8599603a61b75 : SUCCESS in 55m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fcebba47bdda41cea33cae49ef0e9faa : SUCCESS in 1h 12m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7e45d5ad9b88492a829826e1a0948c3e : SUCCESS in 20m 51s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9a1177176fdf42de8d50ce1aa026f2a0 : SUCCESS in 22m 13s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40aed95183a84e699ec631e27656707e : SUCCESS in 18m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/467c9db1c7064fe98cdf99f0c32291ba : TIMED_OUT in 3h 03m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67396734e16346bbbcc49c25e6069252 : SUCCESS in 1h 02m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8f416573188e41e1bd613341ae35db1e : SUCCESS in 18m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cd772c1294642ccae19547da4ee405b : SUCCESS in 15m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4d9d4eef22f94f048bb748db014eaece : SUCCESS in 39m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c3d778dadabf4b9c807544a94484a5f9 : SUCCESS in 31m 53s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5de4d2cd7f8840109aa6f5bf07708d12 : SUCCESS in 28m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4e5f927a052243e0b556a95a491f23d6 : SUCCESS in 14m 14s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/417b4db0cf324ca797ad567500f8595d : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed2b75042ef04838a770e234544c0394 : SUCCESS in 11m 20s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/087d3ab52a414d18a19789f3d2ea5e49 : SUCCESS in 12m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/83da13b3cdcd4171bafc4120c82d61f5 : SUCCESS in 17m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90cecdeda09e42c997eb5310a81e4adf : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":2},{"id":"9dca4491f9d12a4ad9ffb849dcbc22b73075e990","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-30 08:20:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9554cb62ea38d2d8f9ad4ac58b5abb4794dad4cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 09:51:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8899815c8ded4885bfc0047f47cd4186\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/141a3654c8fc4b96bcbae21cf36bcbbe : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4cd7a72ad324e86a8effe49ab130ac5 : SUCCESS in 8m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5df1823164bb433092907dea13907770 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2085bd954229427c94f6eeb8f0496e1e : SUCCESS in 12m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9e240ed182ee4ccbb9938584854f1462 : SUCCESS in 53m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5dff5cdac3654f7e9eeb420eb5a93c05 : FAILURE in 40m 45s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c38a73d4b8ec47cf88847f3a79e57225 : SUCCESS in 57m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/aed621eebcdf44c19c2786d08513ccaa : FAILURE in 39m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a166b86b8ddd42dfab17004c46ad2db1 : SUCCESS in 21m 41s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/185dac9d6f5e4fdb9a43c6cbc42e6011 : SUCCESS in 22m 00s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/287c646cbf9847338427e4e8a5599138 : SUCCESS in 17m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e00cbf88b22c484c84f27a1acb76ad76 : FAILURE in 39m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28157eb27bb6427b8f8097dea9df1ddb : SUCCESS in 1h 03m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/51abdbe44b254395bd0a8daabcd73cd7 : SUCCESS in 26m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6c64621e7f743c4abfe360e4b47618d : FAILURE in 41m 29s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/05acd84148fc4d01a17f4fb3c7d04217 : SUCCESS in 55m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/97720b2f7367479d944a33b403c91844 : SUCCESS in 1h 26m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b489269568834ac2967207de03205b1e : SUCCESS in 1h 25m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d4227cd3a74e497d99f041e1390ae71a : SUCCESS in 15m 03s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/97acc43d576a456fb158999d4f9dd538 : SUCCESS in 16m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f06f1b955714921aae6d70f52ee4275 : SUCCESS in 14m 17s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/f916b81355084cfaba26d564be4013e0 : SUCCESS in 15m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2287b2366e8a40ef8760038c538d69b2 : SUCCESS in 15m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46613af7592641dc8bbd3bf3c9610b4b : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":2},{"id":"0c9265c6ac806e5e343a65150e65221cd3579099","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-30 13:14:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4eef524f6a26d5e75fdcd3529baf467c537255f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 15:54:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56534928aa3a45799b237fbf247540da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f703cd28c40438994c010d2dedfa364 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbdc1596ab7846ad894985d683dc6ea3 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8efa51e2b9ea47fd88af382f79bfce79 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a119509a8e409d892268fbcf6d7441 : SUCCESS in 7m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d05a473a2d93499db6c3c0c35bd76452 : FAILURE in 1h 03m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c1b25f9ebbe747848043c40c9d184b71 : SUCCESS in 1h 02m 02s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f46908f687f4a0189c542f95750b328 : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2d1f5014e0c543759295d53a40340d68 : FAILURE in 56m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c9e24ccdff71414098753aaf0607b5ff : SUCCESS in 20m 11s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/13a660cfb74e43f598c4501d51fbc2b3 : SUCCESS in 18m 11s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1dd5155d12844bed8f99d92bffb15341 : SUCCESS in 13m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fa54681e88684b81b0b58390d715358f : FAILURE in 1h 12m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cdf23ed519e43009240e8e54c138192 : SUCCESS in 2h 05m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2e6459232c5e427e8f8656f6d29b08c3 : SUCCESS in 16m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0305416ca62840bd81da8b361c9742cf : SUCCESS in 18m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/745deda8deae4da9ab7b41c050d4d6d7 : SUCCESS in 38m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3e424e4201e0455e9bf03f005d023c31 : SUCCESS in 27m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16676de3c7554ec496304ce61669aacd : SUCCESS in 27m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/81499ed30dbf4192975616a04e3cbc4f : SUCCESS in 14m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/97d652631b6748e8ab917f99d29ab3ae : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f455ed4ffba460584de1d847c2ae6c9 : SUCCESS in 12m 04s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/2d73c409bb7647c2b7dabeccb7daf317 : SUCCESS in 12m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e236fda7d8fc43418b912619cb30c5f7 : SUCCESS in 11m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5392ba4fd064543a785e8ab9bc95f10 : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":2},{"id":"6502e684a2749dfa10471f7226d28133bdc7dbb5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-01 19:00:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6354acff544bf6c9f0b651bfa27d0a1cbec58834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 21:04:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e337f577a965410286ae41bc382e8eeb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60bfb741cd8348e3b4f9941a825795d9 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5348528a75c8403ca2a0373baec326ab : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1b4084f5f4d4b0db364ebc68ce18bbe : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c3c97ef196a42a69b50a4c72f78d1d8 : SUCCESS in 4m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3d9eafff63a940668d2a7daed7a6395c : SUCCESS in 43m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a3b5f35ec5854d95ad558b398a6bc653 : SUCCESS in 43m 33s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35c50003215949ceb598aa6f46e080b2 : SUCCESS in 55m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0939a855bb1145359e86bc7e14dc649e : SUCCESS in 1h 03m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/538fa74073114251b6c578e87f9e912b : SUCCESS in 16m 54s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f0f7c51f88e24e8badf5caac808ffffe : SUCCESS in 14m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65ead592e5e541558033ae7187627cab : SUCCESS in 15m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/48476ddeb91947b3bad4b73d5d8d87e1 : SUCCESS in 2h 03m 20s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf74091e805f4e4a8d190ad27fe2a7ad : SUCCESS in 1h 28m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3fab3d02a7c64edc8f1d4f48725a7aa3 : SUCCESS in 12m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/328e6fad9ed449cbbdcc0acdcc208141 : SUCCESS in 18m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a79e52af00d94b5b8c3a2927000cd43b : SUCCESS in 42m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4a7e432b063c4493967bca004174667f : SUCCESS in 33m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fa29c3d78494b6ea85288c84cbca8da : SUCCESS in 32m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c9caea23dee84dd38a0858d8753caca5 : SUCCESS in 15m 39s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/5d2e2ed2ea0a441684314dd82876261e : SUCCESS in 10m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f080e566663d42eba400ef837e36abc0 : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/344335d4eda74af5af629342fe345856 : SUCCESS in 9m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6844302ca0c94886b3061468ff477295 : SUCCESS in 11m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39487d866a5b4823a78550f03a9f894f : SUCCESS in 10m 27s","accounts_in_message":[],"_revision_number":2},{"id":"ec9bdaf5c2b3958be5f1d82846738d7d36aebf20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 21:05:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a2c7f6018b602c84ae25db3a80da87e70692bbe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 22:32:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b45701945f2643598a79ef96becd5736\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d6ef25c477470f9d97e8bb9a1daae4 : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70acced58ffc4d2f9672cfc8e6c05ba6 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c6b0b62b4c4db0aa67fffaea4f2978 : SUCCESS in 6m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9654b473fec8415997ff874c1b63a2c4 : SUCCESS in 45m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6b213e51f8a4b8dab475b9a01ace673 : SUCCESS in 49m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/eba95a8f08804adea7ccbc14623d337b : SUCCESS in 50m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5be42d17568549499f575aa98aea816a : SUCCESS in 14m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e35edf005884ea9a62f96f925d00e8c : SUCCESS in 8m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fc4fd21a710540aa94bca91aa8587706 : SUCCESS in 1h 26m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5c7c11ef0d84f78947ee2a30ca2d813 : SUCCESS in 1h 02m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3a6fd3e46fd140e290a2405a02a3ca51 : SUCCESS in 13m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fbe178992364ffe9b8137a9218794de : SUCCESS in 8m 36s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7b561223bc5a49839b639c4d9c20b5ed : SUCCESS in 23m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83cf78054d244b3f8acf0e9d4358986d : SUCCESS in 17m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0f85d3849024473f99f4602b77133fe7 : SUCCESS in 8m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb512708bea34489bc5ffaaf463d59d4 : SUCCESS in 7m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/394b1907861c4552a436b04df5bdd917 : SUCCESS in 7m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c953fe37e76a4643bd579b53b6d1f3db : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":2},{"id":"37152a11adc4928f2a341b91fa6b961c19cb04a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 22:32:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c46d3354fb6a3b20a8bade1bfe12c2524ae8b511","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 22:34:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5893dc11311f47ccb1aea11c74bac6de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b822088a0fb141d0915a9de72bcc6c2b : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3133805e3d34dbe2f7eee44b7d6a73f656a62cee","revisions":{"18b30df3afbdedd12ecb417bcee681434776df9e":{"kind":"REWORK","_number":1,"created":"2025-01-24 13:30:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/940131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/940131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/940131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/940131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/940131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/940131/1"}}},"commit":{"parents":[{"commit":"1424835969053b8c4d6ea042647c23d074502635","subject":"Merge \"Install setuptools before creating virtualenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1424835969053b8c4d6ea042647c23d074502635"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-24 13:26:44.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-24 13:28:26.000000000","tz":60},"subject":"CI: Set previous_release for Epoxy","message":"CI: Set previous_release for Epoxy\n\nThe previous release is 2024.2 (Dalmatian), or 2024.1 (Caracal) for\nSLURP jobs.\n\nAlso switch Ubuntu upgrade jobs to 24.04. SLURP testing on Ubuntu will\nbe enabled once support for Ubuntu 24.04 is backported to 2024.1.\n\nChange-Id: I9fdd017f857e9f56fd4926a3314443348fe53fe1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18b30df3afbdedd12ecb417bcee681434776df9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18b30df3afbdedd12ecb417bcee681434776df9e"}]},"branch":"refs/heads/master"},"3133805e3d34dbe2f7eee44b7d6a73f656a62cee":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-01-24 20:53:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/940131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/940131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/940131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/940131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/940131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/940131/2"}}},"commit":{"parents":[{"commit":"0b7d8015251d5f5dc7cba584a8e2ca86a46f758a","subject":"Fix Depends-On","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b7d8015251d5f5dc7cba584a8e2ca86a46f758a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-24 20:53:12.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-24 20:53:12.000000000","tz":60},"subject":"CI: Set previous_release for Epoxy","message":"CI: Set previous_release for Epoxy\n\nThe previous release is 2024.2 (Dalmatian), or 2024.1 (Caracal) for\nSLURP jobs.\n\nAlso switch Ubuntu upgrade jobs to 24.04. SLURP testing on Ubuntu will\nbe enabled once support for Ubuntu 24.04 is backported to 2024.1.\n\nChange-Id: I9fdd017f857e9f56fd4926a3314443348fe53fe1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3133805e3d34dbe2f7eee44b7d6a73f656a62cee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3133805e3d34dbe2f7eee44b7d6a73f656a62cee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
