)]}'
{"id":"openstack%2Fkayobe~940441","triplet_id":"openstack%2Fkayobe~master~I1179362e7f2b42c638ceb037bfceb3386daeed85","project":"openstack/kayobe","branch":"master","topic":"ansible-cfg-pre-upgrade","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-02-19 08:48:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1179362e7f2b42c638ceb037bfceb3386daeed85","subject":"CI: create ansible.cfg for pre-upgrade deployment","status":"MERGED","created":"2025-01-30 12:25:52.000000000","updated":"2025-02-19 08:50:10.000000000","submitted":"2025-02-19 08:48:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940441-ansible-cfg-pre-upgrade","meta_rev_id":"c6a4aa7352ae9f1dc042127a9bed9e88e9b6d6a5","_number":940441,"virtual_id_number":940441,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-19 08:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-02-18 18:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-02-19 06:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-02-19 06:55:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-02-19 06:19:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-30 14:13:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-18 17:24:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-02-18 17:24:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-02-18 17:24:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-02-18 17:24:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-02-18 17:24:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-02-18 17:24:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-02-18 17:24:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-02-19 06:55:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"392d0687df7a4523895bca288f2f706db4bedd87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-30 12:25:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a09e3882149eac520523dc78b0a9bf8b4d540909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 14:13:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9ac7d42f5234db88f43ae95a17e6d7e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28e3fde5131c4aca94da41b9985c9d8b : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db33a51058fa4c06bf64ea201a8fbf1e : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c861cb93c000491699beb4238f16a445 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf8158bacb9545c682cc6fc0d74f6f7c : SUCCESS in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/167db3dc3f6e492992de1b1d06bf00a4 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fa9caa128ba44acbff79cbad409a3fc : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01559ad5d8394ae18a007a76d339e13a : SUCCESS in 9m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/15e2adbb276646e4b91b2d7a2f6839ea : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e532e4a874e14b33a0d77f540bfd16d0 : SUCCESS in 56m 35s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06fd3a56cabc4b00ae759cf909cbb073 : SUCCESS in 1h 08m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/635a05d995234ba5b11aec39f8721c48 : SUCCESS in 1h 06m 41s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/02841f964a3342619c7c7d5e7f5ae137 : FAILURE in 10m 51s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c21e93b80b8b4ee783ad4a1212c2e4fb : SUCCESS in 19m 56s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9de7e0d00564c8793567f78e4828b15 : SUCCESS in 8m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7d84b8eaf6fe448c80842a2189b983a0 : SUCCESS in 1h 42m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d78b2be197943188a6ea15529195428 : SUCCESS in 56m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6a912398bec74db590f7dc22f9142267 : SUCCESS in 23m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba2d0f9824734107a8e0ab091954eaf4 : SUCCESS in 14m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/241be198af4c427fb23a471ea45aa8bf : SUCCESS in 46m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ca840b0d88af4b5284b2a5440fa8dbc8 : SUCCESS in 18m 34s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5509c5c6278644ec859ffd0410f3a269 : SUCCESS in 33m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5ad8abeea0754d109d39e594f8c23107 : SUCCESS in 13m 51s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/9fbe4475be2341e8835cbe41d0991cd7 : SUCCESS in 13m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94db0a6181bd44d185a2db3602c06f77 : SUCCESS in 16m 48s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/0bf085e472a047098654014deac2b2fa : SUCCESS in 14m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/78cd7d9f05dd49b99f2caade0e0b5fa6 : SUCCESS in 13m 51s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26aa13850e43406ba8de8b4ad3b487f9 : SUCCESS in 15m 30s","accounts_in_message":[],"_revision_number":1},{"id":"42e25aa2d6ba4dc37f16ea3c88fdf4184fa434b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-10 13:57:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d30dbe6827a5dd9e2c11e055d9a94c5edebee651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 15:50:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7769703578e40ecaf41033f347fdb88\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfb328305fbd4898be192e11189b1431 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/763a3e0f205b47f3ac704f2d17f1e72f : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3f9d18686b742129e784643f9869696 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d30626f70e24ddfba4398509bb2a0da : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3777b7c4bcb146c8a8a7d6733f65d449 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc3f2957f2304ff8ae8b922bd9d485b4 : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/410cdd31afb948a3a7cbc76f4d6e851c : SUCCESS in 10m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8b7c2cb1c8674d8091b6f34cbe831de0 : SUCCESS in 51m 24s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0b67c2a0d9084bbfa78cc1c43d0e1b65 : SUCCESS in 1h 07m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c3a2bfafcb424b6eb40d68148438e360 : FAILURE in 10m 17s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f83a6dbd0ece44619944417a0ba48cc1 : SUCCESS in 51m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/62057044ddff4d40bc4895a3adb937d1 : SUCCESS in 1h 03m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8885dcecc55a46b38b49349d571d4f20 : SUCCESS in 1h 02m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4da9d3b6e4ed4ce2bc12fa401ef6fd1e : SUCCESS in 21m 05s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/24f30bd8d38f4db7ac5ad7fab3f4671a : FAILURE in 10m 59s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/510e86244b95467e92263178ebec4d02 : SUCCESS in 15m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f746765b8ad24f7786a2aa1c0038077e : SUCCESS in 1h 47m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33b6f7d1ea8a4728802a2d63566987dd : SUCCESS in 1h 32m 21s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/da5e7298b6f64a41836313dc110e1a40 : SUCCESS in 1h 42m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2589a60e408042aa989baf8a67ab4deb : SUCCESS in 23m 42s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/2659834cd6264212997c22b1b5143407 : FAILURE in 19m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fd6ea2c6ba842b1b2278b87ad4be6dd : SUCCESS in 23m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aadb4af316934eb88bb280e3f934f172 : SUCCESS in 19m 51s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d03eabff216745a18a639ca0c40938a9 : SUCCESS in 42m 11s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/b809ff9d8bf74e0889a2acc2e7f1f6a7 : SUCCESS in 1h 13m 37s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/07658d21b4bc40c38f7a2951d656de86 : SUCCESS in 27m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c52208129f804c8283f019848e848d9c : SUCCESS in 33m 56s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/419b51a04d9b47738248fcf1516d39c6 : SUCCESS in 36m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/17dd4930b3b04faf9ab6bc4e0e0a7180 : SUCCESS in 18m 28s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e23fcc4fb28b42d5a7227658e024deaf : SUCCESS in 17m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42b68a9454664dad946747d0bf893f43 : SUCCESS in 16m 31s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/cea511f4738042d0b82b38313fc676d8 : SUCCESS in 17m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2b9f35f1184a4feea68a7879979f292c : SUCCESS in 16m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2054f40c4ce43f9ad02016f0bd46a5f : SUCCESS in 16m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4715b38df1da8dc0ea1fe6a1230328c69db3c13a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-10 15:51:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a22ca196b6284dfac928fd1ea45440cd2150310e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 18:37:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75d8207600994f028fd26b7f0fabdb67\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca78b33eb08749f1b920d380d50b265b : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/817696eb4d5941428af8fd3fe6b3b9fa : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b7d00fc0b994b7bb3ccff71fe71b563 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd6847f895744b72ab281f31dc60b4d2 : SUCCESS in 8m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c4db7cdb7f41427fa78cb8c8d7d69c1f : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/32126b20ae8d401ab5417844bf7171fd : SUCCESS in 12m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e59fe658ab14322add6406630d8ab93 : SUCCESS in 10m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d0ed9de9b8184ac0ad5f81b3e962099b : SUCCESS in 1h 13m 09s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/b558dab3d2ba4b86b67e7f05570060b0 : SUCCESS in 1h 14m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/eb412f49f19348029a4f38f0dfaf7ab8 : FAILURE in 9m 39s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aab84c1fb4054261abd72de7692a9dc2 : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9d4b64c70a774aa6a867fc038377df05 : SUCCESS in 1h 08m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6f4d87b9c22642808ef4bc8ed3d4d2bb : SUCCESS in 1h 09m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1309b329bbfa4bb58338bebb7e0005e2 : SUCCESS in 17m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/083e79f409364713b65118161949827d : FAILURE in 9m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65ca659c6d8649178e6f4c0dd30e51bd : SUCCESS in 14m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7552d5bb267940559409fecd5f9eda66 : SUCCESS in 2h 04m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68b4996ab0794dd79d5ea0769cca6969 : SUCCESS in 1h 45m 11s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/154aa837593e4dd58cfc4884adced4a6 : SUCCESS in 2h 18m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5bc8b64248b34d8d98056b1d4f77063f : SUCCESS in 20m 11s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/dd46112705374b87ae87249cb14c0133 : SUCCESS in 21m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89089fe6e8fd4791a48af8174b264a06 : SUCCESS in 21m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c3d988f7a770420989b0c5084e86723a : SUCCESS in 19m 30s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f499c8fc537b41dc9806901afb3e55e5 : SUCCESS in 54m 58s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/1a2f0741ee48474aab1486bbc776c432 : SUCCESS in 1h 02m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/02cb4621925b4109b5827c61e27709d9 : SUCCESS in 30m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d50daa1e22754ad19519016269977c69 : SUCCESS in 36m 35s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/10651d0d0a954f63af9e10c6c5e1d1d1 : SUCCESS in 29m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f875a6c815594a59aab4f4ed7fe1d773 : SUCCESS in 14m 07s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/29f3c22cd7bf406a8b275137393949a6 : SUCCESS in 18m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5f9467ff91743339926b737e7719dda : SUCCESS in 17m 21s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/741e44e1d64042318cb61f8d194ef6d0 : SUCCESS in 21m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/36034908fc834cf4abbe3aa48de5c7b6 : SUCCESS in 16m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63353de4ebad49b288d84999ce2aac13 : SUCCESS in 20m 26s","accounts_in_message":[],"_revision_number":2},{"id":"7ee7a19e750474b60cc29633d39f10042fcaa996","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-02-18 18:01:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"22decf52926147a4b6fa5666afd65ce9a29d2413","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-19 06:19:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"49e6342346bd9b45eeb08135d497ada44437454b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 06:19:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"699830dff6d4a5c89854f304b21175f95737931a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-19 06:55:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"440bb44ee13a80f1fcbc679ac16557208b627492","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 08:48:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb4710a1fb64771837590be7405d62d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/052ae54406604e9c9d5441d578fe3eea : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d12d065eb3cf4de1bdd32ad06835bf8c : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440e81c8bdc148f6b6932ece94921b0d : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bfb322decef045bda107274d404747a8 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8e6031af9db94e97b8d34c396d3170ef : SUCCESS in 6m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1919dd68104f4235af2aa93f84d05d20 : SUCCESS in 6m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/966a41bc30bb4e398e8830a9348e07b7 : SUCCESS in 1h 05m 55s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/130c1930b58f4829a31fb1e469fdf63d : SUCCESS in 1h 03m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/172cf62efce242de9ff65380d290cd90 : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cb9f75d8a52049bfa57b810517044ce7 : SUCCESS in 1h 00m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/23e7341656de4ae9acee80e1a4086746 : SUCCESS in 1h 10m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b50c3aaaf55a4bf5b258925d970ddc34 : SUCCESS in 18m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a9fd21fedcd4dc4b8df94d782715ed0 : SUCCESS in 14m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/721a9c5a6b1d4ae9a6280e120911e318 : SUCCESS in 1h 46m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ce32109e4404b5886b97743cabf14dc : SUCCESS in 1h 54m 20s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bde4ca348be64255a0d9664db0c34b09 : SUCCESS in 2h 23m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3900582993344dc49d74e491e8f929a4 : SUCCESS in 22m 05s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8bd7712b9a054e53a87ebae0182bc631 : SUCCESS in 21m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/479131901e054204aa5710717b9725d1 : SUCCESS in 19m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dfb0f89521c04c4d8ba6b5ad64d5a754 : SUCCESS in 32m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e41e16880751443a8614857082450f95 : SUCCESS in 25m 54s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c6972fd072494ba1b2c519a1dd4a172c : SUCCESS in 32m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/004b0b5877aa42f79d45881db271deb2 : SUCCESS in 18m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de5b6698b137439a8b89d1f2284d3cd1 : SUCCESS in 15m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bc4845a94d6b4d8389e36e46a4c9481a : SUCCESS in 16m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd6eef389b924f6ea13820d541adad7f : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":2},{"id":"2e3f87559e2b573766117949fc691d445ad02ffd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 08:48:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c6a4aa7352ae9f1dc042127a9bed9e88e9b6d6a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 08:50:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a598ef6fd1c0411b9062f2fd98522b29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c2bc69a0ab4349b013300c99da859e : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b7586c8e97ceb69fd22fd07a5b6f51c71d6fee1b","revisions":{"6cdad023fe5d0ea3cd89329602685d039c971c9f":{"kind":"REWORK","_number":1,"created":"2025-01-30 12:25:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/940441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/940441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/940441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/940441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/940441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/940441/1"}}},"commit":{"parents":[{"commit":"0b7d8015251d5f5dc7cba584a8e2ca86a46f758a","subject":"Fix Depends-On","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b7d8015251d5f5dc7cba584a8e2ca86a46f758a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-30 12:24:49.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-30 12:24:49.000000000","tz":60},"subject":"CI: create ansible.cfg for pre-upgrade deployment","message":"CI: create ansible.cfg for pre-upgrade deployment\n\nChange-Id: I1179362e7f2b42c638ceb037bfceb3386daeed85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6cdad023fe5d0ea3cd89329602685d039c971c9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6cdad023fe5d0ea3cd89329602685d039c971c9f"}]},"branch":"refs/heads/master"},"b7586c8e97ceb69fd22fd07a5b6f51c71d6fee1b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-10 13:57:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/940441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/940441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/940441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/940441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/940441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/940441/2"}}},"commit":{"parents":[{"commit":"8dd05487d4ef872e7d3467e74fd6ab5141361d68","subject":"Replace pause with chronyc waitsync in ntp sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8dd05487d4ef872e7d3467e74fd6ab5141361d68"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-10 13:57:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-10 13:57:14.000000000","tz":60},"subject":"CI: create ansible.cfg for pre-upgrade deployment","message":"CI: create ansible.cfg for pre-upgrade deployment\n\nChange-Id: I1179362e7f2b42c638ceb037bfceb3386daeed85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7586c8e97ceb69fd22fd07a5b6f51c71d6fee1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7586c8e97ceb69fd22fd07a5b6f51c71d6fee1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
