)]}'
{"id":"openstack%2Fkayobe~940646","triplet_id":"openstack%2Fkayobe~master~I73e7f2a3258e5427ec16b01e7482093d7bfb33f8","project":"openstack/kayobe","branch":"master","topic":"ntp-force-sync","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-02-07 08:24:27.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-02-07 10:47:42.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-02-07 09:07:44.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I73e7f2a3258e5427ec16b01e7482093d7bfb33f8","subject":"Replace pause with chronyc waitsync in ntp sync","status":"MERGED","created":"2025-02-03 17:21:39.000000000","updated":"2025-02-07 22:28:56.000000000","submitted":"2025-02-07 10:47:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940646-ntp-force-sync","meta_rev_id":"f2d9abaa98f5e57b4e246cc847d15d243f719b5a","_number":940646,"virtual_id_number":940646,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-07 10:47:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-02-07 22:28:56.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-02-07 08:24:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-07 09:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-07 09:07:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-03 19:44:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-04 08:59:57.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-02-04 09:06:28.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-02-07 22:28:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"2ac053694d7d0f50fa9c18c8112754d37c9485a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-03 17:21:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4054e36bf1c12984125c3f8c2dea2697fe2255fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-03 17:23:20.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3e5eece119277c8bb07c7d5a561ea823e37a84fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 19:44:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1be9fd1701a945ab94f65b44bb3aecc6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f023b0eeacd41f6aeaf66ad0a6eae45 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c2b841acb1344a991c4d1678dd5b906 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e52f7f452e8d4037af7e700ca0755c5c : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b89ebd67e2ec438b9a476866a294f6af : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da6997e405d947a19ec804f4f6198887 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6601dd9b991249a3bd3b8028baa74562 : SUCCESS in 6m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c46ecf356321442da066430e1b79e9cb : FAILURE in 7m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a16d5520f5004c73a7716f4de1af91a4 : SUCCESS in 1h 05m 15s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/deb7a2e9e12447a8b5d858eea79ebc5c : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ad27d70f639e4fb09da1e36b9e407653 : SUCCESS in 48m 20s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c81c4d52ea94505aeec9a6df973a6d9 : SUCCESS in 1h 08m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f84ee4b3079047e2909ba5f78f2964ab : SUCCESS in 51m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a0b203367a6f474da341f93affc0c60f : SUCCESS in 55m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/94e388b9b1e447c0b4bc9c94529c2309 : SUCCESS in 18m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/38563ebd017b4bac8dd27485aa34be40 : SUCCESS in 17m 05s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb0817b687964da1aa1cdcf92ec37e27 : SUCCESS in 14m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4175a32942d94a65887ea3776bef918f : SUCCESS in 1h 30m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45a20f1191e44c8eb377140bf1943e9a : SUCCESS in 1h 45m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/eb62cf9c7d554993b3c4f0da96d9c460 : SUCCESS in 15m 10s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/7fb19d1e2efe4a499037547ee7b07886 : SUCCESS in 14m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f8985303f8d4d22aeb70e108fa13cdb : SUCCESS in 14m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/357e7fd434364d26af273fcd2cf4d4af : SUCCESS in 15m 11s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7f9cbe982c1f4bcf9aa4a12d55e579e3 : SUCCESS in 42m 50s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/877fca8b0c2c4d169683826ae19a1141 : SUCCESS in 1h 16m 54s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/90874f17576444408e5cbdf80a506f19 : SUCCESS in 33m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad5d7e43cefb493ab7efd9f1950848a9 : SUCCESS in 26m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a786e30942904199a83dc425e46e0688 : SUCCESS in 13m 57s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/c6d9244f71fb434bba7c44ef6e145b2d : SUCCESS in 14m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/140956c840214e8ebc7748b1cfa7760f : SUCCESS in 16m 52s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/7a55674aeb754da7975255418bf825c0 : SUCCESS in 13m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e6418a15e168407cb461210852c31a18 : SUCCESS in 13m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15106c8668d44745af48d4b6ba609d92 : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a2050fcdf2977266b870fea411988c28276a9ced","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-03 20:32:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a4539e61d536331739c7e56279022d67fbd9084a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 22:28:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a66fd9eb0a64e6585dad9b204bac119\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1f6cba0496f4a1f91796e2dbc85da3c : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e261c2fde5d54975829d637a884f7cf8 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77cdb6f6b6a94959af8108b5bc9dcb6c : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7382860935964907a5b361b930a106b6 : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/69770d91710a4aaf96f4730cbf7ef567 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b98c7f2cdeb94a88ad75772f841e90d1 : SUCCESS in 11m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f254354be794404b75562723c53f8ec : SUCCESS in 12m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/52b9e51cb6bc4ab2aeea7e2941ad4ab2 : SUCCESS in 1h 01m 47s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e5741a888a0e4e51bba6a30dcc0cae7b : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/99e43c03492146d18618078f6c973055 : SUCCESS in 1h 04m 35s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf9e427377e249df8f0231ed621381b9 : SUCCESS in 54m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e96362f326444c5997708c091c203b35 : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c78fb9891caa4d589a3f2e35b60087fd : SUCCESS in 1h 01m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/295c8b6f45374af99eb6bcbe807f0281 : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f324b564b7ba4ce5a0d6f6132d974922 : SUCCESS in 17m 15s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6139ac4e2eb54b4dbd081c74bb0b45bd : SUCCESS in 13m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d87f8c0c306f46b5ad89328c33b95d48 : SUCCESS in 1h 46m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a802b2e9a4054853a87457c29330caa2 : SUCCESS in 1h 49m 50s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4f22a57becb54c508e8591c14dff4344 : SUCCESS in 1h 35m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e5d1599cd915497fbdb083c33ada1d31 : SUCCESS in 16m 44s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/c724022f547e44fcb4dc48ca0bf47d0c : SUCCESS in 9m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f14f2b205f50425d9f6889b8d249f26c : SUCCESS in 14m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c45b1f385cae4e56b63474fe043deab4 : SUCCESS in 9m 31s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fe09c7b686ce4392ae1cc64c5d59a747 : SUCCESS in 41m 32s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/c7f8ab6489e448cf86e037d68d1b319f : SUCCESS in 1h 05m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/793d96f3d11d424fa30d72761f5e49df : SUCCESS in 25m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93d734385854457595a7d8c0ddaf3268 : SUCCESS in 26m 15s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8f8cb962c23f4e2da6720d9b820f8795 : SUCCESS in 25m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bb5e9a631a1c473eb4180b6d7aeed7f6 : SUCCESS in 13m 33s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/748c32fac2544d40a839cde0409d4305 : SUCCESS in 14m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7032baad892843be9daa0815f8a40b03 : SUCCESS in 8m 24s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/4dfde12ba9d74e20aa72f15ec26ada91 : SUCCESS in 14m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fb02e1df62e14ef397aaa71ca8d8c9a6 : SUCCESS in 14m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c305a995aa194345ba833ee601151477 : SUCCESS in 11m 04s","accounts_in_message":[],"_revision_number":3},{"id":"48cd3ec643c9c336feb20bf8d1002d609351c78d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-02-04 08:59:57.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ede51035d7d08db4a59ff555701f0c1ea1ef6bc6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-04 09:06:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8d69fdc0a88479826b1508d315523f6b6a39b57c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 09:06:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c924dbf844db70cf663e3fb5dc36f0b9f63c6b4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-04 09:19:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"852b978f596b1217950196cdbb6296ece2db7dfb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-04 09:19:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c59b6e91512d24c135916bdddbda406a005888a","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-04 09:22:06.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0982e977148549647756f7aba0cce2fd9b6186bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 11:21:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45a7e4efb0064a7ab9bed4a0a0835c86\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb2db6522c18490bba61e20b796c2bc2 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dac69770cba947b7b8b78cb877488616 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ce6f259442642b395450805c56d8c50 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c7c8ad7dbe5427ea978c368a42b6797 : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ade4c41efa824c50b8160ed61df6fc14 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7c4df02fac5433299a84ddd0a9c3f45 : SUCCESS in 6m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/84ee95cf5c0e40dcb876381acd2e5267 : SUCCESS in 12m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/49cf1eea49604a9087ab34c0aebdde0a : SUCCESS in 55m 19s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/44114638953b4c75b180c53b759736e6 : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c840bf9c420c436f86ec0a4a4317db42 : FAILURE in 9m 37s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b06df202afd34d288128bfbbf875baad : SUCCESS in 54m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/08263e0cb03441faa9d2ed8930272220 : SUCCESS in 34m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/50f8ef5ccc214c308b2ee6cba4ad5d3f : SUCCESS in 1h 02m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e0502c306e0d4d8d99b130944610f171 : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/df666e53f06549bd9aa006cdc0361b97 : FAILURE in 10m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/239e15b294c944c1ba3ad2a5a2d4f214 : SUCCESS in 13m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e3b1535ebe184dd6b0e47a63768e9bcf : SUCCESS in 1h 46m 28s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/961f2594504b43ea929ac524f51718e2 : SUCCESS in 1h 44m 12s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c2480847053d42bdb045aa98ac5417e9 : SUCCESS in 1h 48m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/830c1b60f7e54625888b74a4890ebd9a : SUCCESS in 22m 58s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/3046cb88c2424db99fa01985a37da9d5 : SUCCESS in 16m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/232a963f9fd54f988e627a3c8cff8fc3 : SUCCESS in 20m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c8f23c2aa5b944ccbf4b90fad1d8c9e8 : SUCCESS in 8m 51s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6ad4c69af2c946e1be82e369c59a380a : SUCCESS in 36m 25s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/9690e62238394e5e9254d3ec4417c2ab : SUCCESS in 1h 13m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/56209deafe244e33abcb173606a2a8dd : SUCCESS in 29m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5d4c00d0cfd4b2a97a706b38a4e135b : SUCCESS in 35m 25s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/04f7702fcf62425ebb24fec70358a34f : SUCCESS in 30m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eea903fc187d4a7296da5ebf70b9dfe0 : SUCCESS in 13m 36s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/101707e530ae48d78215e45803ef8f63 : SUCCESS in 12m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f67a3909b4034ea6a18837bb0171b448 : SUCCESS in 12m 12s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/63388c02a9ef40d3a5773457b79b3f2f : SUCCESS in 10m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/83b83e26d5da44a2becdf1dcb7f84fa2 : SUCCESS in 13m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/434eabc8c98444fb8d5915b29e376640 : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":4},{"id":"0b2215be756c68539f94f49568035e37fcb6e3b5","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-02-04 11:24:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7ac9afb006f45f6377b4014ccf557185e9029ed4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 11:25:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8164892ef0e39d53ada966bdf36a4fadfe5819e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 13:16:02.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f0e35bace1b478a9579b2b73ae72019\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfdb922583294e0cb41c6afc34072123 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/846c76bbd74d4c54ac93740ce336e8bd : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96cf3b01b4694bd4b0e24fa5159f2f2d : SUCCESS in 6m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c5f0c753d723405ba081e67ec87fa268 : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3b0b9e788a06452aa98ca6189410d19e : SUCCESS in 9m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b60d5b62197649d58940ef1b13d4bf25 : SUCCESS in 11m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6a5f7ace35a143409384e9c724c48f87 : SUCCESS in 1h 09m 22s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f58b3008a39f4046987e7957f820ee9e : SUCCESS in 36m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08deeae6fcd644f5ac777cdf4e9fd1d9 : SUCCESS in 32m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2e30f2a0159141eea2ef6ed8e7420cd7 : SUCCESS in 33m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/aa11fb79c50f4a5687d23660dffeaa95 : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/94825bd511804edbaa76b72802998161 : FAILURE in 10m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56fd5b1fa9f4492bb2dbe25b2f6ff8ae : SUCCESS in 13m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/07200fed9b554e078897cd4314501914 : SUCCESS in 1h 01m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25c807b29bae4beca5fc50d1ac0f5605 : FAILURE in 1h 43m 10s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6c050f156deb4f2a87fb2d9a79f2faf0 : SUCCESS in 1h 45m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2cd8229f0a4041dd8fae74e157eb6e01 : SUCCESS in 15m 51s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/495c6c44104d46f5a93c0d3ebcc11ac6 : SUCCESS in 16m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/173a6678d1e448a2b1260a0bb52889a8 : SUCCESS in 15m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e2030999dcf84138af1574c8f1fe0d83 : SUCCESS in 26m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23a4b4b562494932a775970ad4135e23 : SUCCESS in 28m 51s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6c189472be034e3abc70614db114a5a7 : SUCCESS in 29m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6882d61a43f047f1bde9f239bb463d83 : SUCCESS in 13m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0231c304fde0493d842e2e64d08b831c : SUCCESS in 13m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/38a8a5d9f52d41f9975bfd45ea37478e : SUCCESS in 12m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b29042d0fd414f2487c5da0d680021a6 : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":4},{"id":"845cf8645337cc8303491979ca05d4977bf986c9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-04 13:22:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7f81045e3ffe585461e98053f23c161cd5608ca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 16:21:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a46a42555f04e6780392460ef4466c8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b13d6ec9d7834b208349645d73c68282 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1445d5dd15814e25a70e59a4bb1b1da8 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95f7b58003e3486185ae45a0dd455519 : SUCCESS in 14m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e49f9ad2ab64eab964f213a8cd6e677 : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e9f4358e9074408ba36d6372d8382051 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ca9360db9db497d870741979bb515cd : SUCCESS in 6m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a2fe81a2b3c4fe3969e163a6a523e69 : SUCCESS in 10m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/959edb518ea041e99d55c3896c691fcb : SUCCESS in 57m 17s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/149be90e5ca24cc88e300b746c168c26 : SUCCESS in 1h 04m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1243b856d6674f848d82edecf4d14be4 : FAILURE in 6m 15s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52a95db9c3dc4443b57bb46ca3c0a78f : SUCCESS in 1h 03m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f218783f1c484b2591fca1e0c7c709ce : SUCCESS in 2h 03m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/28e60941da394a76bb2840654565d080 : SUCCESS in 1h 02m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/62c592f035b54d188bccf1d73088d278 : SUCCESS in 18m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d5a72bb3d87a4f6a949486e4cffcfed0 : FAILURE in 11m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9091ba4f66a14d39b8d397067d1b686c : SUCCESS in 30m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c10f8542a7bc497884b68c62e35cefe0 : SUCCESS in 2h 51m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ebfa3cc4717468c848a5e6218d97c85 : SUCCESS in 2h 46m 45s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/30f9d005a98843139d7d73eb0b71d58f : SUCCESS in 1h 49m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ebf9b68798424d5581940e4480c13026 : SUCCESS in 17m 40s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/37f8a01596344b94b27ae0a49809273e : SUCCESS in 19m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6842448860c24f478d96ffaaa87be621 : SUCCESS in 22m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/16e54a61a6d74e95bfe99f73e221e4d5 : SUCCESS in 43m 37s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5e4300148f0e45509d985d45bd46963b : SUCCESS in 41m 53s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/72ead7d8a71646bf8e064d8a0e2126b8 : FAILURE in 13m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d71373cde1eb4855b1e4de3999f0a0b3 : SUCCESS in 29m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b96760d535804d269c0e66855133cc0b : SUCCESS in 1h 05m 15s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e1591f1176b649f88d1a3ac55dbe7554 : SUCCESS in 26m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/29798260bccc4fb1890040e51add18f7 : SUCCESS in 16m 16s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/bd0b508dd464474e947af05248193f36 : SUCCESS in 17m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/157a5495e0f747b8b2657130deb88ed2 : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/e9c9611bbea04cdfb8d4ea5cc1784314 : SUCCESS in 10m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/36aa93a276534183b396dc57d4d20029 : SUCCESS in 13m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bb8c05779c94cae89ba92b27b502089 : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":4},{"id":"bb2d5528ed03d482fcd1035b40635e3e92cd70aa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-05 08:25:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f528178254408214356cf09f95e1204dbc85bd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 10:37:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51d3acd458d742b99dbf4a9d4322960a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6208e34ab54b4334b8116294389ce27d : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df5106da0e7044b98149fd9ca1efa977 : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79d0504d2ac14d78931b470a1c595c0a : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd6b5a9408464c66a9258933afdf21c3 : SUCCESS in 6m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1af4f576547a46efb7fb25a57e571bd7 : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/65079a56ad744babadcf60db7a24b77f : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b34b9c4aef04a0db266364aff7d5e42 : SUCCESS in 11m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6da3c0703f394bd19824cc47122e8d5f : SUCCESS in 59m 53s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/b51f3ddeffc54524ab6b876035b19a35 : SUCCESS in 1h 00m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/945d09c59bb64062b284455e1698fc68 : FAILURE in 10m 14s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfeaa80ebeca42ce85f71fbbf96ae3b8 : SUCCESS in 1h 13m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/708121e2118f4ef7965545c99face267 : SUCCESS in 58m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/17d720f4cc5b40188abbcef6b26f7c17 : SUCCESS in 1h 01m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b72add03e335446f91d17a1bdc3b07c0 : SUCCESS in 15m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3332d6f6549f49358edae8f062352295 : FAILURE in 13m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbfa253beddd480d834218d51e41b539 : SUCCESS in 14m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5579cd6a9da54d9aa9f40da80a844857 : SUCCESS in 1h 37m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/074db344bba24eed8033a7763a85abe7 : SUCCESS in 2h 07m 45s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b72e337e07a84a30ab39933447294cae : SUCCESS in 1h 48m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/06c7d75a32114e5ebaa2363749d49ed9 : SUCCESS in 16m 44s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fcabdaf4e4a14adb8387b1db045ec583 : SUCCESS in 16m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ffb2fc7b7b34bf6a432ac40baef694c : SUCCESS in 18m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bc732ada3a1a4b06bd3599a7e4bc986c : SUCCESS in 14m 07s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/76b6166cde764d6ea2ba825f0008754b : SUCCESS in 49m 08s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/299ea122e594405b87a57c1623927dd8 : SUCCESS in 1h 20m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/186e87529c6349c99243fee64b98fb1f : SUCCESS in 19m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84e1ba0485c44f43894c6e65beeb0377 : SUCCESS in 27m 37s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7c403f9cc4aa45dc8d834c9e666e8ccd : SUCCESS in 31m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3aac299b00534a6e9cd0274ef5387745 : SUCCESS in 14m 56s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/5df59a4b32fd4d22bde44421f7707f65 : SUCCESS in 10m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21e2f4f610404584a02cecd4ca89408e : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/fa76aafa20564e5a884fb7ffbd307246 : SUCCESS in 12m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6ad5e68eafdb4ed6a2cf5b328580cf77 : SUCCESS in 12m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e0b8098b1634b259e0940ca68bd8ee2 : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":4},{"id":"5987eefc90a7703ad15e36ae19b4a24be3a253a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 10:38:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"421287e9019c74b898692d42c0c4c44cf203a6bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-05 12:11:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"42dc52095293d3b0c1d1b6aca789d06c6c6826db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 14:27:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24d5cf979c2f4e3986232b84bbde5509\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20df782c254e45888b4ad4458875b7f4 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3f9f22b6efd4a418c206b283750c0ec : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c98464b2ff7f470f98635e7bd8bea3fc : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e59bc8ab124a32bedb45b5fd018a9d : SUCCESS in 7m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8bf0c616fc9b41dfa5313791ea874f19 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b81b3b056dd04ee4bf1f6d9bccc4943d : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca9bb28d208242cbb2ff86146148aded : SUCCESS in 11m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/56f244fa08df4a0b89f8fb6992324924 : FAILURE in 57m 40s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f1a146f1e1bc4f5bab5999633cee3510 : SUCCESS in 59m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/95e5c49c832d4823ab2f21fd6d8dd9ee : FAILURE in 5m 40s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd06e3e5679b4445b1b4255f13ae4212 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba4a4c45f3874550b9e9a06832f24ae1 : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ddca14f591fb401096c19a030b27b274 : SUCCESS in 1h 08m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fbc03d83ff774f18ac54b0f24f4a1d9e : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e64105c6654947cb8135dc57b5092c17 : FAILURE in 6m 21s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07cc93b17b5043ddb59b48d3b3f6f9bf : SUCCESS in 16m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4ed3f6925e1e45798042bea8fb166619 : SUCCESS in 1h 45m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c3e9b8bd46448d9a63d0426010623b3 : SUCCESS in 2h 00m 06s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b60db99861a14f1b9f93ddbfd0c2d1b8 : SUCCESS in 2h 04m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2ae567ed70544f4bbac10cb9353c8120 : SUCCESS in 18m 16s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fe041d8cf4094e16bc43f572eae47cd5 : SUCCESS in 9m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c91d60ee42bc4ec48e3beba3413b1339 : SUCCESS in 17m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bd18ece1dfd44cdb98309d737f792b6a : SUCCESS in 13m 39s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8c284098cb45409987f563a84f422605 : SUCCESS in 40m 53s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/debfdec5bf484622ad837af88293457f : SUCCESS in 1h 12m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/acd9529ef00d48d983078f5fe72f739b : SUCCESS in 29m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e59b34e08944ac393fd3b4ea88e2547 : SUCCESS in 30m 16s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4470706ef9a94036b02bc1593cdc4a71 : SUCCESS in 29m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea3a443343434af1a1a17a5925faef08 : SUCCESS in 14m 28s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/eb4d7d9f2a094c569d67d3cc2496a5d1 : SUCCESS in 12m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db341aa8183d48029756ac7b42bdc098 : SUCCESS in 13m 52s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/c29973f0006d4ab3b7bf7c5212e6ed3c : SUCCESS in 13m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dd669b6a2bd248d7a4b1bfc8f2dd9374 : SUCCESS in 13m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/773c2a8828914b2694f1ae076f52fc83 : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":5},{"id":"908a032d53dc41f3d9ac8fc879e00318a6156a3f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-05 14:36:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6172c68191504f4843d1d47133ba33e81808bbe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 16:33:36.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cde227f645c4692977efb875171bd7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ee548c9710a40cda9edc385373f2713 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c99c127827a4a2db537e0ff7227aa55 : SUCCESS in 5m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c6eb29a12ac41a39ad1361a9172675c : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7d87e1ac5ec4a1e9963b8f3f9250e49 : SUCCESS in 7m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ff17f9d54a046c9b51677defd405eb7 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd84419d546b4b4aa01eac97c31b5183 : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba9d8c21a0bf4416a42eb0fc1e4ee05b : SUCCESS in 10m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/de51eeb9068c48578736a8b5b358d532 : SUCCESS in 59m 12s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9bea7a80f7604aacb1c1053a5bbd820e : SUCCESS in 58m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/16250c0b5a184dac96a59ba088bceb88 : FAILURE in 8m 58s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de2139d075314e33b3e7ffe360d4d10a : SUCCESS in 54m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/baa6c08c47244ba4834f1079dd01a8b3 : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b78a64cc6fc44b9fb647c3ec6d3223b4 : SUCCESS in 1h 09m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7654daf3d5f8464cbe4b41b6338c7c70 : SUCCESS in 23m 50s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8735e2ff78344b028361cbe5ed5a3cc4 : FAILURE in 7m 48s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc69718dcc9149b1b8aae9537ba886fb : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1b5db3099d8f418ebd693477a0a1bd53 : FAILURE in 1h 48m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f25cf4d90c5b4665a99057b4219d4ca5 : SUCCESS in 1h 42m 03s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/43091e05904d42d89d294e778b994331 : SUCCESS in 1h 40m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/14595e47b25c4c84ba0245380e83d8cc : SUCCESS in 16m 51s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/7c29b31863334b87b0dd45142ab1f905 : SUCCESS in 11m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21d910c202aa43438804256cc40c53f5 : SUCCESS in 16m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/48be311338144ee18ba2efd677bf8b83 : SUCCESS in 17m 35s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f131d42925314947b0e411c9f65aaaa4 : SUCCESS in 37m 30s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/5275d33d446a4c31b63cbc25d3fa7a83 : SUCCESS in 1h 17m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/75cedaf1929b4015a5e232c0fd3c8a28 : SUCCESS in 28m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fa3198e44224c32ab12fc90c3205b87 : SUCCESS in 33m 11s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ac44e315eb6042a99af1254c28bbb529 : SUCCESS in 21m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4b3ab5cf01434127a6ba59e60da0f06c : SUCCESS in 15m 07s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1e9f561d850340098af8a8a7133efd94 : SUCCESS in 15m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55cd8052a54a415dae4332e89bb03dc2 : SUCCESS in 13m 44s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d827eda866db4bb89387147ee8cbf4e4 : SUCCESS in 14m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/183cd5c275ee4497bce7f26d29689406 : SUCCESS in 14m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd686f547af54233802c1e65e2aa4fe2 : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":5},{"id":"b80c6281f9aa8797ed2ca27e1130bea78767b6a2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-05 16:39:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b88bdc584ce0f0447368cfff7709d3f7c5a5488b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 18:38:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e26345b12d2a4897a0495501a6d32f00\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6ebdb50f967463cb457b7c373307879 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96089b73def74a288d51dbedebe3a422 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6539251c9464450ba577f6957364ef88 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c7f7a748a4a4e678158abf055e4fd79 : SUCCESS in 6m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/04f049243aad4069a9af5c824be97f28 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/64549b0094644c8380374e6357304975 : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/718764399ed74a7d961f21b783c86b02 : SUCCESS in 8m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9650a0cad30842ae94a262d33d2b7940 : SUCCESS in 34m 36s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f0ae1fd975e543ef820a834f2aec198c : SUCCESS in 56m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/21239080b292488c9f6fe6e4bfef6272 : FAILURE in 5m 58s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f978b50a891945cab070f867979c9aad : SUCCESS in 54m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d780719984204f57b8aa7bb57f1d35ff : SUCCESS in 57m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d5a7195a69514e499484237400708265 : SUCCESS in 59m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/50782ec644b54044b1686b7705540d69 : SUCCESS in 19m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7b2eaaf55f6645f5ab67d01f4d3a1a0b : FAILURE in 6m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebc3dafc0cd44dfba2dc1f6ff73be22d : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e248735277f8489a9e1f0b09a7ef8f03 : SUCCESS in 1h 35m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecf49a57703e4a44aee040041bb1093a : SUCCESS in 1h 44m 50s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2c6b3f0f1bdf4c37b1e5f141bf8cd855 : SUCCESS in 1h 58m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dd7f083d2b8a4ea291a56ee397e8c746 : SUCCESS in 17m 45s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1959a896e65c4ea2b3bdeede3b700972 : SUCCESS in 17m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac3cd964da3a4c4083aa07517ed2d7e4 : SUCCESS in 18m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c59aadc5d28c4f679f9e67670e73bac7 : FAILURE in 11m 43s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/52982c6621ec4a468fd619e43d058237 : SUCCESS in 42m 12s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/44ac62657d6b4f0c9fbb955fe7d5faec : SUCCESS in 1h 20m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1c657b8a82204cf08faaba119d9dca27 : SUCCESS in 31m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92b088bb51434e529b52a4f082a2207f : SUCCESS in 34m 25s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e62de71bbb5647d08688ee0e54fea0a0 : SUCCESS in 30m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f74af5a84e57480d859ae8ecffcf24b0 : SUCCESS in 16m 12s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1861e4f39c2d469b9b228b729ba5cbdb : SUCCESS in 14m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ceeddcb09fd7459197b47d458c10b62e : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/55eed1cc1370427cbdd8aca9f7f19f02 : SUCCESS in 11m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ab603d3782e2477eabe9832216142d63 : SUCCESS in 15m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cc3e9efa8794ce38d0ed2fafd5bb390 : SUCCESS in 13m 10s","accounts_in_message":[],"_revision_number":5},{"id":"0f5f7900badfdc22f33b7df8be0e67600b7d364a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-02-06 07:58:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"3d1c12acb23e3f1a40248868f017d82e7cba66ec","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-06 08:55:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d9e89e362548fa87d140f1e26b95c34ed72eb763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 08:56:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"35983d94318af44d1ef4de871e5f877c5aff3369","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 12:05:50.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98ddce55e13549a5be1993745667a6b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e1e9f61943442f0b0f67d5fa5a56393 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3909d303037749c8bdd1e6a08caa0f39 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa6b187aae44b24a0dd97c2ec2eba00 : SUCCESS in 26m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a75b902822d3414580274797230e6e9e : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0ff07b9cacf54a32aaa01e763786df5f : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7bdaaeb74c3842f6b11334d90ca8895d : SUCCESS in 15m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d88225aa02e44365bceb98c0fc6ac7ad : SUCCESS in 59m 05s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/d7e58cd8cb06482ebf44a16b43d2f599 : TIMED_OUT in 2h 02m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1e6b881adc34e27bc0b5e9dfa0d6b3e : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/39cf8527e279462c9f1ac6bb7a53b311 : TIMED_OUT in 2h 02m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a8f08559e3444c8698bcf8d48c342267 : SUCCESS in 1h 14m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/911409d13114429394e0a63a737ccc57 : SUCCESS in 22m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ca41d8dd3e64c1cb3b9e1061c924673 : SUCCESS in 13m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f692e48c0771485da837cb7bbc62bc45 : TIMED_OUT in 3h 04m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5e87177162b4187bf52dfb25c063355 : SUCCESS in 1h 51m 34s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b4e1575b7a4b40ee99244913d468faba : TIMED_OUT in 3h 04m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c389b1ebc1e84482954659e1fe0ec1b2 : SUCCESS in 52m 33s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/b855e8f455674c8387d0747dffa29d41 : SUCCESS in 44m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7ccce93bb444d73b50f0bc534087267 : SUCCESS in 56m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/81047b6b9b2e4abaaaef39aa8551cb63 : SUCCESS in 30m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cb90fe363cc487584f3125127f7672b : SUCCESS in 28m 14s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/aa2fc539d9d54e02b5ff7ca913be3021 : TIMED_OUT in 1h 32m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7d7c51eddc214543a71b02dd0d0ace6b : SUCCESS in 13m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d78a15bbfc14cb88266302802663b38 : SUCCESS in 19m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9ac48ba2348646509392388f9ebdcf03 : SUCCESS in 13m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2aeee10b0b7f474da8a703b66fcea1a5 : SUCCESS in 18m 31s","accounts_in_message":[],"_revision_number":5},{"id":"78e8590ce0c00524e85834145417fe094336e160","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-06 12:08:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"584729fe41c35547c7d8a79e826180ecb0ec580c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 14:23:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/861911a5e7ed4fdcb8554631a7a66b88\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bb13552e67545b18c55f142f8c14993 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13370b8490e04b12aa033f138c1f8b3f : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30c3746d1a894586ac44fecf60df74f7 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/748f6ba42d924c229a88c2cd11b685cf : SUCCESS in 5m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e70ab15218874f0ba92bc06bad899aff : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/184f493cd9554bccb629a6a4d41c058e : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2052f01ff7c6489298cfe87eb7879083 : SUCCESS in 6m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e6a875e6472740ebb27441f05a12b100 : SUCCESS in 49m 19s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/78687a6a35cd4686a0c8fe566ac4680f : SUCCESS in 59m 51s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/171aa54e11594208bd2534890528306b : FAILURE in 7m 13s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ba12ea8f1ef4556b74f6f4351fda40b : SUCCESS in 55m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3dd2d191a1f644dfbc4bb11941f74926 : SUCCESS in 58m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/517ccb331ca44e2aa7cea4ec24931c8e : FAILURE in 58m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/840287922ccd484b936973453c3c59fc : SUCCESS in 18m 53s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/243efa5bd1104d7d9ea83081aacb9a98 : FAILURE in 8m 16s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73a21fb540f746aaae221c5db81713b5 : SUCCESS in 17m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/879d073dabdb400f8549255952db33d3 : SUCCESS in 1h 59m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b0ea91b31694bb3a2f0206f5710a636 : FAILURE in 52m 06s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b7139558efe045df953fe307ae99c879 : SUCCESS in 2h 06m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4c6f238371ef45a5af92a69e1048b9d5 : SUCCESS in 13m 59s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/9e96e524c13d4a8485d39c1fc75af1fb : SUCCESS in 16m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aacf56c581024929a1958d9d9540327e : SUCCESS in 9m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/17950c0858c24ae08057a821b5c1d9a7 : SUCCESS in 12m 31s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/dd456271e1eb4daea27cbbcf44213268 : SUCCESS in 32m 07s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/c7b601b2fcfc4236a26d7108e8dd9b66 : SUCCESS in 1h 12m 15s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9cca08d443cf4be99249a3cf1be8ac60 : SUCCESS in 19m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/101f958f51df470fa2fc24c9d653a0f0 : SUCCESS in 22m 28s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/be956e0552f9428999694c81c09b34c7 : SUCCESS in 36m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7e26793aa838496097eb52839f8bea1d : SUCCESS in 8m 49s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/78dade098f3e4fb5868caca2489edf37 : SUCCESS in 11m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/799c2ca0ab5f4d21a354db2d43dfcc4c : SUCCESS in 14m 50s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/99113a4c0d5a4064b3e81d08f10a833f : SUCCESS in 13m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a13c002b8b3048479588796de3095af6 : SUCCESS in 8m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f96c1a880fe46668a52e3e0293f271c : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":5},{"id":"1ad0a791faa756c5a2807af18d3f7e4de1c0c2f0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-06 15:04:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"da847e63175f9acf072d75701c0f17cf1f9b5486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 17:10:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/962b0026182f491eb6bba9f5e4268812\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b77f337af99f46f8930c9ea5eef51350 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2581d783a374b849a5f7009cbd97e54 : SUCCESS in 7m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d044e6064ec64ae6a7deccd505b6eb6b : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d1235c997649bc8e0fbfeee7ceede6 : SUCCESS in 7m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a054614c7c1422ebe68b5e07b59ea1a : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/333f158efc464f61a96a9e24d5a10a6a : SUCCESS in 10m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/723b2fa6e62c4a7f8ba37daf7347ab1c : SUCCESS in 10m 38s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4df8865a0a8f4418ad1b1c4061eeebd3 : SUCCESS in 35m 23s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e85589bce03c4f5eba6b8511daae6090 : SUCCESS in 1h 00m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c1fbaf83d5bd45bcb939cf27a35bb5f8 : FAILURE in 10m 09s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c66f937eb57d40e6bb54e752cf81c9da : SUCCESS in 53m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/38a1f69f035d42ed822ffad7a130ed9d : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c31d4670648440e8b03c68ffe97c4157 : SUCCESS in 1h 09m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/316c6b9e285449b8aa24e0040c68911e : SUCCESS in 18m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/05d73d633e1b41448c466d2a00a8e19d : FAILURE in 6m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5db50c58b884df4970458451b50d816 : SUCCESS in 18m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0960a6e780194449a66c1444ecc2e7c0 : SUCCESS in 1h 57m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ee715e563c142c3aa5e50390e64add8 : SUCCESS in 1h 42m 20s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/73a4c408c92843e48db670cbffadfc01 : SUCCESS in 1h 46m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f8c544958d564ad1a88cb50a53cc9cab : SUCCESS in 16m 29s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/0c38c482800945ea80158dd5ef695a11 : SUCCESS in 18m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49695525c6fa41f0924ed3865ea8096b : SUCCESS in 15m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/43d765dc2e3849119a29c2bb01b85e00 : SUCCESS in 19m 25s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e39ac567be9440099ccaea7648c19370 : SUCCESS in 37m 40s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/1b327d2215d246889b7f9f4387e4f0ee : SUCCESS in 1h 11m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cc4e433536f34f609e88125df9356ae2 : SUCCESS in 18m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1f41d4fa4434134bf29033fc753df0b : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7135746c00eb4b9e80e8913cba4b4a58 : SUCCESS in 32m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/80d18a754fd8442ca638bd02257b7a06 : SUCCESS in 15m 40s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/bfc9e92d8c1745d2a799856053446942 : SUCCESS in 11m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ce138b937d54d6085701bc0c17af16a : SUCCESS in 7m 42s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a23d066d8dab40fd8b00162332ef1319 : SUCCESS in 11m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e3951f59fa2f434581288afcd9414456 : SUCCESS in 12m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cecaf23be074a298881b0aaf78a4397 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":5},{"id":"976d3db3fc9e791fc476a0ef13c6d7791b70f519","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 17:11:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a1d2f9a16718d39a49210affa3542ef79e554ce8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-06 17:57:19.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"c8eb43e090d40f28c477d9cc5aaa05f6c3e44279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 19:56:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5621bc58c0f34ada8fcaa808c027e93d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bd63d6633f245fe8be6879686bce5bf : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e16a02b06864cf9bcf56743b5f599be : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1def20a15644f43bef5a1f8430f2509 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b9c6fb07b84136b678ac3dcb67bc5b : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dbf6e2b13e3c4f8ea26c8e23efdd05f7 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4334dfe2d285410bba6a6cd6afbf5679 : SUCCESS in 10m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/524aba6075324da0a827a05090669b39 : SUCCESS in 7m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/93056d163cc84b78a6b6b0e0a1899266 : SUCCESS in 49m 42s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f0c402f0b7aa48238c06d025f178513b : SUCCESS in 1h 08m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a047cce8b552440cbc4c7a2156a760be : FAILURE in 6m 49s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b0568c004e84953991e48115baafcd8 : SUCCESS in 54m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/29402a9cff2848cf8177fe0df2135d83 : SUCCESS in 40m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9c9c028b16824c4a96881ce4e4dbecbb : SUCCESS in 42m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/35ffba3b6d7e4140b4a00e9b4355141b : SUCCESS in 15m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9f1e5a89b6f5431d8165904dedafc6ed : FAILURE in 9m 59s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7630a6fd6c8461096bc2f1736a73fd5 : SUCCESS in 14m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f7ba3d6729a9488da8195f1fb09ed3c3 : SUCCESS in 1h 53m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfecdd17b0b2419bb21d79e84c371ad7 : SUCCESS in 1h 39m 57s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f53ce3318bc84565b9de77b5535eafb3 : SUCCESS in 1h 43m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4176d73c7b864669861bf7d3ce304bf6 : SUCCESS in 24m 25s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/4e22ba1577be468da86c5bfab2af1126 : SUCCESS in 20m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47cc156a6061448d8e5cbe4347051eaa : SUCCESS in 21m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9c71f8ba371a44c1a9b5fc595c4b320c : SUCCESS in 17m 25s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/80c59cde0cca47618bccd06edc4197d9 : SUCCESS in 47m 23s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/c1484799ea274b97ade3f85e88df55e6 : SUCCESS in 1h 19m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fb9a87ca6cf141d18ae42dc5d12e4e2d : SUCCESS in 32m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fed27cd8af8d455ab7ff92f46abf877b : SUCCESS in 35m 19s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e27a10a2cc27498ea74c49902905e18f : SUCCESS in 36m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/221d06d34eaa4e8eb12f1a8bbc4f9f31 : SUCCESS in 19m 49s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/d4ac451d88564fd6b2e10e3afa655979 : SUCCESS in 22m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6230a625bad84598b4ee71463002615e : SUCCESS in 16m 17s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/99d884c2ff204fff93909f02ec28b1ec : SUCCESS in 16m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c9efc3e8b98446ac917fdd86e83f5ad9 : SUCCESS in 13m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93eeebe01752414b895d9b742af7c03e : SUCCESS in 15m 56s","accounts_in_message":[],"_revision_number":6},{"id":"a8e8d56be2db39a7bea5aa52a0cae5e7be4cdbf2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-06 20:23:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"faaa6358ffa786ec37d11999539ca0d9593ea53f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-02-07 08:24:27.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"10a066afbabe7c7e86f97fbe144336d4189c5ad3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-07 09:07:44.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5a89374927d088d5f10c74991f54345c7121c74d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 09:08:09.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7084cdec87a0f9de310eba0e609c8d2455b2ba08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 10:47:42.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a42067eadfab42da9465a15d2fb89ab2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e825f14dde74132b871609a22253845 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8320e0f29c9c4aa980d3ccea0be3d67f : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e4ee74ae144f7790eb669441975b76 : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f66d2e6d36c4573b39ed40e386029de : SUCCESS in 7m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1234ae0c21f943bcb415aa6804c9abb7 : SUCCESS in 6m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/233615aaae6e475bb28f124b75900310 : SUCCESS in 9m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3d58e5252f2241c4a7d4b186519138e5 : SUCCESS in 1h 06m 05s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/078b344e37e147cc851dedf34f097214 : SUCCESS in 1h 10m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2484f1269ab843278d43f0f7c00ed343 : SUCCESS in 1h 02m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2687a2581fa84d0c9f0d4d0816f197f6 : SUCCESS in 1h 02m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/31bb8d45d659485ab21e7ab8ba74b827 : SUCCESS in 1h 10m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2b5446c8c6484101b433ee9b7c0341f2 : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f5eb82549554848aa9b6c296a652960 : SUCCESS in 13m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a466ad9ea8c54ea99a70d94b9a1dcadb : SUCCESS in 1h 39m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c1e0f68ef8f459093641196f952d275 : SUCCESS in 1h 28m 31s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5a279c49f500454eadaf7b2161da466f : SUCCESS in 1h 35m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/40f8bf620e5643cd99ee29dd11475f7c : SUCCESS in 19m 00s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/418c6f312b42470d8898baed0a332906 : SUCCESS in 15m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cc5d4f0d1d646aabb203df0f8c685e9 : SUCCESS in 18m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7c781d6d148547f99fc4a0c7939655f2 : SUCCESS in 18m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/126cd51313234f32b49d907563bf3916 : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6839dad33deb48f68c8555ed2d0562a8 : SUCCESS in 30m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/22432d6086e24f0199b7ebac603234a1 : SUCCESS in 11m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7aac6da64564415ab0835cc42e4678f5 : SUCCESS in 12m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/86de2903acdd48ab9590556929c907b8 : SUCCESS in 15m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8242bca14a2c4ecc84098624c962073f : SUCCESS in 17m 47s","accounts_in_message":[],"_revision_number":6},{"id":"7b9ae6225dd8e70f54eacee85dbf095534800a5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 10:47:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"94bcb10e91c4e72a74372806519dfa8461a08a07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 10:48:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/558436bbb07f42c483f9085d702a25d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2c4186c4f524a538ff214d63ba5a47c : SUCCESS in 45s","accounts_in_message":[],"_revision_number":6},{"id":"f2d9abaa98f5e57b4e246cc847d15d243f719b5a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-07 22:28:56.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8dd05487d4ef872e7d3467e74fd6ab5141361d68","revisions":{"2a1a4583dccf8fa72b3f0ce1f7a332a605c226da":{"kind":"REWORK","_number":1,"created":"2025-02-03 17:21:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/940646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/940646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/940646/1"}}},"commit":{"parents":[{"commit":"9130dae3b83168c692b9ae9ba04964c0123660b3","subject":"Merge \"Add podman support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9130dae3b83168c692b9ae9ba04964c0123660b3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-03 17:20:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-03 17:21:24.000000000","tz":60},"subject":"Replace pause with chronyc waitsync in ntp sync","message":"Replace pause with chronyc waitsync in ntp sync\n\n`waitsync 60 0.01` will wait up to about 10 minutes (60 times 10\nseconds) for chronyd to synchronise to a source and the remaining\ncorrection to be less than 10 milliseconds.\n\nChange-Id: I73e7f2a3258e5427ec16b01e7482093d7bfb33f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a1a4583dccf8fa72b3f0ce1f7a332a605c226da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a1a4583dccf8fa72b3f0ce1f7a332a605c226da"}]},"branch":"refs/heads/master"},"d72eaa9a0c5612912d09407c4b9b5576925346c6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-02-03 17:23:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/940646/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/940646/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/940646/2"}}},"commit":{"parents":[{"commit":"9130dae3b83168c692b9ae9ba04964c0123660b3","subject":"Merge \"Add podman support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9130dae3b83168c692b9ae9ba04964c0123660b3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-03 17:20:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-03 17:22:27.000000000","tz":60},"subject":"Replace pause with chronyc waitsync in ntp sync","message":"Replace pause with chronyc waitsync in ntp sync\n\n`waitsync 60 0.01` will wait up to about 10 minutes (60 times 10\nseconds) for chronyd to synchronise to a source and the remaining\ncorrection to be less than 10 milliseconds [1].\n\n[1] https://chrony-project.org/doc/4.6.1/chronyc.html\n\nChange-Id: I73e7f2a3258e5427ec16b01e7482093d7bfb33f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d72eaa9a0c5612912d09407c4b9b5576925346c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d72eaa9a0c5612912d09407c4b9b5576925346c6"}]},"branch":"refs/heads/master"},"ccdbb71cd7d90b9145675e855e3988d4507adf49":{"kind":"REWORK","_number":3,"created":"2025-02-03 20:32:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/940646/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/940646/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/940646/3"}}},"commit":{"parents":[{"commit":"fad3b7f1da951f85b214ac9b7a573f6ff11c608f","subject":"CI: Add Rocky Linux SLURP upgrade jobs for 2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fad3b7f1da951f85b214ac9b7a573f6ff11c608f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-03 17:20:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-03 20:32:48.000000000","tz":60},"subject":"Replace pause with chronyc waitsync in ntp sync","message":"Replace pause with chronyc waitsync in ntp sync\n\n`waitsync 60 0.01` will wait up to about 10 minutes (60 times 10\nseconds) for chronyd to synchronise to a source and the remaining\ncorrection to be less than 10 milliseconds [1].\n\n[1] https://chrony-project.org/doc/4.6.1/chronyc.html\n\nChange-Id: I73e7f2a3258e5427ec16b01e7482093d7bfb33f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ccdbb71cd7d90b9145675e855e3988d4507adf49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ccdbb71cd7d90b9145675e855e3988d4507adf49"}]},"branch":"refs/heads/master"},"ab26d63b6b935a5f1c351a2f9116f0084a2c04aa":{"kind":"REWORK","_number":4,"created":"2025-02-04 09:19:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/940646/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/940646/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/940646/4"}}},"commit":{"parents":[{"commit":"fad3b7f1da951f85b214ac9b7a573f6ff11c608f","subject":"CI: Add Rocky Linux SLURP upgrade jobs for 2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fad3b7f1da951f85b214ac9b7a573f6ff11c608f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-03 17:20:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-04 09:18:38.000000000","tz":60},"subject":"Replace pause with chronyc waitsync in ntp sync","message":"Replace pause with chronyc waitsync in ntp sync\n\n`chronyc waitsync 60 0.01` will wait up to about 10 minutes (60 times 10\nseconds) for chronyd to synchronise to a source and the remaining\ncorrection to be less than 10 milliseconds [1].\n\n[1] https://chrony-project.org/doc/4.6.1/chronyc.html\n\nChange-Id: I73e7f2a3258e5427ec16b01e7482093d7bfb33f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab26d63b6b935a5f1c351a2f9116f0084a2c04aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab26d63b6b935a5f1c351a2f9116f0084a2c04aa"}]},"branch":"refs/heads/master"},"77bbc925bd569f6a6299fbe51a09b8cbaceff199":{"kind":"REWORK","_number":5,"created":"2025-02-05 12:11:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/940646/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/940646/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/940646/5"}}},"commit":{"parents":[{"commit":"fad3b7f1da951f85b214ac9b7a573f6ff11c608f","subject":"CI: Add Rocky Linux SLURP upgrade jobs for 2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fad3b7f1da951f85b214ac9b7a573f6ff11c608f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-05 12:10:44.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-05 12:10:44.000000000","tz":60},"subject":"Replace pause with chronyc waitsync in ntp sync","message":"Replace pause with chronyc waitsync in ntp sync\n\n`chronyc waitsync 60 0.01` will wait up to about 10 minutes (60 times 10\nseconds) for chronyd to synchronise to a source and the remaining\ncorrection to be less than 10 milliseconds [1].\n\nForce system clock synchronisation in CI jobs. This allows us to run\ntest_ntp_clock_synchronized on Ubuntu again.\n\n[1] https://chrony-project.org/doc/4.6.1/chronyc.html\n\nChange-Id: I73e7f2a3258e5427ec16b01e7482093d7bfb33f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77bbc925bd569f6a6299fbe51a09b8cbaceff199"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77bbc925bd569f6a6299fbe51a09b8cbaceff199"}]},"branch":"refs/heads/master"},"8dd05487d4ef872e7d3467e74fd6ab5141361d68":{"kind":"REWORK","_number":6,"created":"2025-02-06 17:57:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/940646/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/940646/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/940646/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/940646/6"}}},"commit":{"parents":[{"commit":"fad3b7f1da951f85b214ac9b7a573f6ff11c608f","subject":"CI: Add Rocky Linux SLURP upgrade jobs for 2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fad3b7f1da951f85b214ac9b7a573f6ff11c608f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-06 17:56:32.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-06 17:56:32.000000000","tz":60},"subject":"Replace pause with chronyc waitsync in ntp sync","message":"Replace pause with chronyc waitsync in ntp sync\n\n`chronyc waitsync 60 0.01` will wait up to about 10 minutes (60 times 10\nseconds) for chronyd to synchronise to a source and the remaining\ncorrection to be less than 10 milliseconds [1].\n\nAlso flush handlers to restart chronyd before synchronising the clock.\n\nForce system clock synchronisation in CI jobs. This allows us to run\ntest_ntp_clock_synchronized on Ubuntu again.\n\n[1] https://chrony-project.org/doc/4.6.1/chronyc.html\n\nChange-Id: I73e7f2a3258e5427ec16b01e7482093d7bfb33f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8dd05487d4ef872e7d3467e74fd6ab5141361d68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8dd05487d4ef872e7d3467e74fd6ab5141361d68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
