)]}'
{"id":"openstack%2Fkayobe~940894","triplet_id":"openstack%2Fkayobe~master~I3e750e38f7b4bff6981b44a3ee03cd2595e51e69","project":"openstack/kayobe","branch":"master","topic":"multiplatform-capabilities","attention_set":{},"removed_from_attention_set":{"36781":{"account":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"last_update":"2025-02-10 11:29:44.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3e750e38f7b4bff6981b44a3ee03cd2595e51e69","subject":"Add variable to allow image builds of different platforms.","status":"ABANDONED","created":"2025-02-06 16:13:36.000000000","updated":"2025-02-10 11:29:44.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5cb98c187993b4367f88d98591150a5cfdf8b02e","_number":940894,"virtual_id_number":940894,"owner":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-06 18:22:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-02-06 16:50:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-06 16:50:30.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-02-06 18:22:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-10 08:51:39.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"0ef88b38fdca45e17bcdf5046f98a83b58a28e8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"date":"2025-02-06 16:13:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b28f387c16fc2e49996d83531b1ac16d899eaa20","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-06 16:50:30.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a5ef37ceeabc958c20da56b10620ac1343d03bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 18:22:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8c480ff45df49aca6404d3738bdab39\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a61748da86574f848c951dcd690df246 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50e2c194f2d498d9700e8fd36595a0a : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94f8b0f42c384c9daeb3761767852849 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3e57e0aff24569810ff78ca33c324c : SUCCESS in 8m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2afbf22ab6f644cc80f26ff8a833bbb0 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/65773cfda129495e879b249ce8c73f40 : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1056d50bb90b45f99533ff4a03861358 : SUCCESS in 9m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c0aa7fbddd1945eb9346dcda798385ab : SUCCESS in 1h 01m 15s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/3eddd55e291e402d8715f626240005b2 : SUCCESS in 1h 09m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/52265208a4cc442d8dcec94c733c051a : FAILURE in 9m 13s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/752bbcdf65094b9e862048605e9e4ac0 : SUCCESS in 57m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c5a3ceeda8f744b385c5d00adf343c7d : SUCCESS in 58m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4d943e0b55d946e59f1d768f162fc0e2 : SUCCESS in 1h 12m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a48feff22755449687debfa24f19489f : SUCCESS in 19m 12s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c4b819e9538a4edab98e06e45a057822 : FAILURE in 10m 00s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23393c3257ed4fdda1e16163b5ebad8f : SUCCESS in 15m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f3e090b1381c4a3ebacce5e1dd237dc9 : SUCCESS in 1h 46m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b95b1f26d7fa4d1b9774952f103e81e9 : SUCCESS in 1h 44m 51s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/76bb18decbc445aaa3b0fe37b8e473b6 : SUCCESS in 1h 54m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5965566409da426d9e5af1684eaf2a30 : SUCCESS in 19m 22s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/93921304c24b412aa206366547d41f1d : FAILURE in 14m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d121bf0acb7484b9ef47753e6f5551e : SUCCESS in 15m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b06045309dda4398b6103d04b22b490c : SUCCESS in 17m 59s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b2773ce2471c4e689d58e675f13c4413 : SUCCESS in 38m 36s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/fa0eb09c391a4fbd9d4f72aa04371011 : FAILURE in 14m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1167fdc13d5549ab89dc04684cc76615 : SUCCESS in 32m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e64e7c1b86a403b8b1a6e6b99f503da : SUCCESS in 18m 06s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f68d1cd9d75d4a509dafc45fcdd7a753 : SUCCESS in 30m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3bc06649c87e46749fbd3c981b22e60f : SUCCESS in 13m 05s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e2d089db289b42c7b4f4b5dd43873ed8 : SUCCESS in 12m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/814e357ec62a4057b21184751b167dc7 : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/f0ef7565980341b49a985f4630071faf : SUCCESS in 11m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5b44314311414a83848d5561a88518bc : SUCCESS in 13m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ae40bd211ae426eb151b1629c82c8ff : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5cb98c187993b4367f88d98591150a5cfdf8b02e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"date":"2025-02-10 11:29:44.000000000","message":"Abandoned\n\nFigured workaround to avoid changes to kayobe","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c9b0c336d2460f495200ac2ee02ad1dbc2d0208","revisions":{"2c9b0c336d2460f495200ac2ee02ad1dbc2d0208":{"kind":"REWORK","_number":1,"created":"2025-02-06 16:13:36.000000000","uploader":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"ref":"refs/changes/94/940894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/940894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/940894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/940894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/940894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/940894/1"}}},"commit":{"parents":[{"commit":"fad3b7f1da951f85b214ac9b7a573f6ff11c608f","subject":"CI: Add Rocky Linux SLURP upgrade jobs for 2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fad3b7f1da951f85b214ac9b7a573f6ff11c608f"}]}],"author":{"name":"Massimiliano Favaro-Bedford","email":"max@stackhpc.com","date":"2025-02-06 15:54:23.000000000","tz":0},"committer":{"name":"Massimiliano Favaro-Bedford","email":"max@stackhpc.com","date":"2025-02-06 15:56:28.000000000","tz":0},"subject":"Add variable to allow image builds of different platforms.","message":"Add variable to allow image builds of different platforms.\n\nAddition of this variable allows users to configure the platform architecture for image building.\n\nChange-Id: I3e750e38f7b4bff6981b44a3ee03cd2595e51e69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c9b0c336d2460f495200ac2ee02ad1dbc2d0208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c9b0c336d2460f495200ac2ee02ad1dbc2d0208"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
