)]}'
{"id":"openstack%2Fkayobe~942344","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I1179362e7f2b42c638ceb037bfceb3386daeed85","project":"openstack/kayobe","branch":"stable/2024.2","topic":"ansible-cfg-pre-upgrade-stable/2024.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-03-11 15:31:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1179362e7f2b42c638ceb037bfceb3386daeed85","subject":"CI: create ansible.cfg for pre-upgrade deployment","status":"MERGED","created":"2025-02-20 09:21:59.000000000","updated":"2025-03-11 15:32:41.000000000","submitted":"2025-03-11 15:31:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942344-ansible-cfg-pre-upgrade-stable/2024.2","cherry_pick_of_change":940441,"cherry_pick_of_patch_set":2,"meta_rev_id":"91529f12a32360f0f24733e2adcbe645f718cc15","_number":942344,"virtual_id_number":942344,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-11 15:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-03-11 13:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-02-20 09:39:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-25 11:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-03-11 13:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 09:39:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-02-20 11:36:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-25 11:00:58.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-03-11 13:39:22.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"f6fc238539515fae6df8a100844cfd9b360912f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-20 09:21:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"30e8a85ed390c7bc4f2058261484af52ba9fecaf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-20 09:39:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fbd23552ce84ce8b184fbb39d8e7e451bf45fc66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 11:36:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d84cba55e441e0b35e085c19a0a52c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff3e745c7f904cf48bd1100bc66753fc : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/123d3dd77cbd46039dc1d389b7993b27 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a8b0c20d34d498eb17925a8f0397ba8 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2085ae94b27546cbbe155ca0d963caec : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d76d392ed3ea4ac3b9bcdd1dac38e931 : SUCCESS in 7m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8da0058ad0ec44209ac33ca5f66d8ae5 : SUCCESS in 4m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ac5605452999437e90659908c3e03760 : SUCCESS in 8m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba72d5204260413093ac23c5b26f282e : SUCCESS in 9m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/22e4332f150247438f551c45a8cb1081 : SUCCESS in 53m 05s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2519a10e835c4df5bc362b4351b7b041 : FAILURE in 11m 25s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ada23176275e49538d41518cd2e47599 : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ed8d0a76b19b4ad198aa72d6d7f7fd2c : SUCCESS in 41m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fa4206b6a6c1487ab45deb321c9f61e7 : SUCCESS in 16m 53s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/08ee7efb16244a4e85d2a9b6a45b3253 : FAILURE in 9m 53s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa746e4e43344829a64e11a28908d1ba : SUCCESS in 16m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6127bf449ee2452fa43f74c7caa2fd8a : SUCCESS in 1h 53m 34s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5afdfe8fcbe94d9791eacf56e1f2aa81 : SUCCESS in 1h 35m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ff9f356fd62c4f4393f61dda89fcfe7e : SUCCESS in 17m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/564b74c0fbde4ca7a5ca959fe8c6537f : SUCCESS in 18m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/cab02b30f01241dfbb8427980eefc99d : SUCCESS in 43m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bd79d3cbb5594012bc60a19348282c7a : SUCCESS in 28m 42s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c54eeb24a374392b51b1c9c0267bc39 : SUCCESS in 29m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/44b0199a6dca49cbb5937a0cb547f0c5 : SUCCESS in 14m 43s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/cee539b034564b57890bf4ec717d1d9a : SUCCESS in 12m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33a16bc3de6640d098e2e654069292c3 : SUCCESS in 11m 58s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/822fc41d77614db78cd8275fec17a7e2 : SUCCESS in 13m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ccd118c7e34943b895fe31eb7c5adc49 : SUCCESS in 21m 27s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bc812680f224e409f982a84ec3c6c9e : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a6298295d272b476de1ff4f5b02e1c7236a18bcb","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-25 11:00:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a2c485e49b6b8165aa976e1f44de9d8db2b82152","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-03-11 13:39:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d14f269c13f0afb88d2013f2510346a857f8d752","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 13:39:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ca40cba453c8750edac16d80a49ac4d2915bbde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 15:31:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92695c96b23847ab909b964d25e645ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bab719dd3c64b059bad5426782c8abc : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/598712f21dbd490eb740a674e3c49d6d : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00181803e3694fa8954f0f6ad08f28e6 : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3da2fb6150fe4481abf41fc0da8abde5 : SUCCESS in 4m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/76f9abf7dfcc48d7abe72dacb3da3622 : SUCCESS in 6m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/814cdcdbe5a742cdb6dd22661e299f92 : SUCCESS in 5m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/16955fea539148fd950362fc8e612c7f : SUCCESS in 1h 01m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/007cf6cf72974b1ba0582460be335cca : SUCCESS in 1h 03m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9b4a4a02ed6044f881a4bc8bcd91f6b2 : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/523b4041b2dd483e8800ca31a3e202df : SUCCESS in 18m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08def423175b43188da2c63e8cb38d7a : SUCCESS in 15m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/68c7cab587f9430e9e57cd2692015c98 : SUCCESS in 1h 45m 26s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3db07920df0b4b38a513ee9e689effd5 : SUCCESS in 1h 33m 45s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/25f1d5eb7aa44d30b24011608f569b53 : SUCCESS in 24m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b102e5657074886918ffb7696fd1b98 : SUCCESS in 40m 31s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9e9e79b8c7314a3a9332e5fddd169cf0 : SUCCESS in 53m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a332a9d621e6431495311ba178bb2751 : SUCCESS in 53m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ae416c091f84725b4bcbe31d592de4f : SUCCESS in 17m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bbd12652a80414f8e514c83e849c490 : SUCCESS in 24m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7faa5b577b07498fbb57723a863ed7a1 : SUCCESS in 19m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18cf7ccc4b5e4fbe86ba856345100dd6 : SUCCESS in 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a992a321df3be2acc51b564f9dc177eb4b209202","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 15:31:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"91529f12a32360f0f24733e2adcbe645f718cc15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 15:32:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb75b26ce431423fbf12fb18d21fec3f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/828817e7f4db465a94797ac7f2252655 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64d62b7f8b2915f1fda9e2ba8963311935b76867","revisions":{"64d62b7f8b2915f1fda9e2ba8963311935b76867":{"kind":"REWORK","_number":1,"created":"2025-02-20 09:21:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/942344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/942344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/942344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/942344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/942344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/942344/1"}}},"commit":{"parents":[{"commit":"eb618e8d137920220b700bddfd5d853499d23c35","subject":"Merge \"Ensure hwclock is installed when required\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb618e8d137920220b700bddfd5d853499d23c35"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-10 13:57:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-20 09:21:59.000000000","tz":0},"subject":"CI: create ansible.cfg for pre-upgrade deployment","message":"CI: create ansible.cfg for pre-upgrade deployment\n\nChange-Id: I1179362e7f2b42c638ceb037bfceb3386daeed85\n(cherry picked from commit b7586c8e97ceb69fd22fd07a5b6f51c71d6fee1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64d62b7f8b2915f1fda9e2ba8963311935b76867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64d62b7f8b2915f1fda9e2ba8963311935b76867"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
