)]}'
{"id":"openstack%2Fkayobe~942346","triplet_id":"openstack%2Fkayobe~stable%2F2023.2~I1179362e7f2b42c638ceb037bfceb3386daeed85","project":"openstack/kayobe","branch":"stable/2023.2","topic":"ansible-cfg-pre-upgrade-stable/2023.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-03-11 16:18:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1179362e7f2b42c638ceb037bfceb3386daeed85","subject":"CI: create ansible.cfg for pre-upgrade deployment","status":"MERGED","created":"2025-02-20 09:23:16.000000000","updated":"2025-03-11 16:20:53.000000000","submitted":"2025-03-11 16:18:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942346-ansible-cfg-pre-upgrade-stable/2023.2","cherry_pick_of_change":940441,"cherry_pick_of_patch_set":2,"meta_rev_id":"55db3cedbd9d80069a53f5cb6b2ba6eadb2a04c9","_number":942346,"virtual_id_number":942346,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-11 16:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-03-11 13:39:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-02-20 09:39:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-25 11:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-03-11 13:39:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 09:39:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-02-20 10:42:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-25 11:01:17.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-03-11 13:39:43.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"44258cb4a310d40265d14fe47cc7e715b487b008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-20 09:23:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3d93782561960cb180adfd2dca264459b2d1f5a4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-20 09:39:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0c80abbec1f3e4105eda90d6e7494e9fffeaa432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 10:42:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e45001cc81b4a52a666dbe15163ccfc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65249ae4c766475cb8133ba672f090f5 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e720e20ccf494d0b932d231c18a97179 : SUCCESS in 6m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5b135ff4bb64ba793a5073bcd4d81df : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/702034fe3efa4578844b13c1e3a5742a : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1398eee666fd439ca64cb6c51f26f013 : SUCCESS in 9m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/91b9e7340850408caede65a37fd3fd98 : SUCCESS in 58m 26s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/82a83e7f4bba4e3895122be830978623 : FAILURE in 8m 03s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d46e509c53a43cab980e292cfc3e0ad : SUCCESS in 48m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9c32f4488e4b42a78f3c86788d1688e7 : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6b60c55224894653b4734ad59b971ea1 : SUCCESS in 14m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ef94f17f32224cea92e305ad9a42f435 : FAILURE in 9m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e98c5c0777946a0bd83cd218c296295 : FAILURE in 8m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7aa36fa41de644ef84068c132e98791d : RETRY_LIMIT in 4m 08s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/641744b3d77a476e844810fbee1ffeea : RETRY_LIMIT in 3m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e3dbe47d015d4f31ac2144bc402dbb09 : SUCCESS in 16m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05eca56df2a14dbb85ba045ad3af6bee : SUCCESS in 14m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4b169d4b355b4d5990a4647e093bc4f4 : SUCCESS in 37m 12s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/94732985a73c4925bc5713ed45581712 : RETRY_LIMIT in 2m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8592dad50a6486f9271982ae03c6158 : RETRY_LIMIT in 3m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/44a28ef7562d4b3893acc47438666d55 : SUCCESS in 13m 50s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/a6e8fa7cec794a619402e8a5df770425 : SUCCESS in 13m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e38659f1f76c48358b8c62c3b6d48d39 : SUCCESS in 10m 12s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/46993190036f4ef3ac006869c4d9f9c3 : SUCCESS in 12m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fbb15937c0a64add91c960e1d2722aca : SUCCESS in 18m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df6e71ab26b1478295557d4d7506bb56 : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa8ea39b371f4cd3b020f21a5d1520b2 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bd535aad9e646a4b8a0f902e02c15f9 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22c0ca176ef740b394c0307f7bf3f42f : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7be0fbaea00f4ad9b8d01d2a3eda9623 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"373392484df3a9fe21b25df33ed827813dbbf222","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-02-25 11:01:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f32f45f5dfaeeefe638b4c30b8cbf80c1a1691d3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-05 11:03:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a98a579615869b9582dc02e4b181ea142ff4ca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 13:17:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6039dcbf55794c0885422dd8edd89cb5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02a9be03561a4404b66fe3381ee32dc4 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15da45365fb14f1887d52e03f3a85d7c : SUCCESS in 7m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/40439e61cdaf420fb6b74de5b931af7c : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3f9f29ba837e405085ef6aef582cddf7 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b0cfecdeea3c4bbda7dba1ed718d0a33 : SUCCESS in 9m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/241c429bae304f319f6872912731470b : SUCCESS in 58m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7fd65e73e8e148139f36b516f70a8f65 : FAILURE in 8m 20s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bdf1247e8e6f445f87bd352b094154fa : SUCCESS in 50m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e3c8cb6d3a0b47de914053b6332a1795 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/81fe3a50d47a4dc2a8fcd2cc4b37ce22 : SUCCESS in 10m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/de43e8605ae8472f9bfba1d7cf5345f0 : FAILURE in 7m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a71b401969848e1833f020c3cb9f061 : SUCCESS in 13m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f4670d1583f84bad83b03115bb4102e5 : SUCCESS in 2h 08m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3bc5a8da0d64afeb7e1cd8ffa7588d5 : SUCCESS in 1h 53m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d7be638e9f1844948427255896149b1e : SUCCESS in 16m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4abea8e0722744f7ac1754d16582daa1 : SUCCESS in 15m 48s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/049ee50cbd2a45a0a3a231e0414d2c50 : SUCCESS in 38m 14s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6bcad474673849bd83e7bb47a5ab5a1b : SUCCESS in 28m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/190d74b8fe6a411a9a1351aad3f85ac9 : SUCCESS in 28m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4953f70963e949a2bde71322ad727a13 : SUCCESS in 13m 46s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/00ced3dc86cd4270b865f4bfe65ddb09 : SUCCESS in 12m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adac30780b714ca1887a4275fd3700d2 : SUCCESS in 11m 31s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/8800c6b7f9034e33b035bc68f5749300 : SUCCESS in 11m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2acae25639514a21804cbd29517e6633 : SUCCESS in 17m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/740d2f318c7e41ccb3c7f9f283a6f11a : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e58f03e61d444f148966202e8c6321ff : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aeef6db17ae34efaa390a3ecd22c6aa2 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f910a0ffb3e4e9bbc71ad9a3896c962 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/547d9e8896404dc28e22e82c4bea6b06 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"709dd3eb3a9df17a44b1baa279979bc3599c4723","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-03-11 13:39:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"825267c3eec077714c73a40738710e4308e69592","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 13:40:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d3e9da14603c29fa74abaa179e6609882a9266f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 16:18:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5c873e9c5024271870bd14c675a358d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fa0808066c3478ea9500edd19f0c8d1 : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fd3a9e57c14b4283aa3341289ecf5a85 : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81f8604f87f747139654a3d33a382ca5 : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c8104f3ed4e4e63979dae5e89188a2b : SUCCESS in 9m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/14596dbf146d4c2a90508d5b9f503c01 : SUCCESS in 58m 16s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14320d040f9a48f69e6b300cff7f99e4 : SUCCESS in 46m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7ba8cc9d6f9a4a69b43160bc195257c8 : SUCCESS in 1h 03m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/74019c2f6f8f407db39c80b66c706fb8 : SUCCESS in 15m 30s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f82285aab6204fcbbb171652d6606e50 : SUCCESS in 12m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/39cf99d17b6b4cf8a35486d92940e3a0 : SUCCESS in 2h 18m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e802024127504985b6ce5063ec8165a5 : SUCCESS in 2h 01m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/eaf9d28e3f64409b88fdc739a624f69e : SUCCESS in 17m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2c18d9f53244ab6aeb67679cfbf0b74 : SUCCESS in 14m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1a92a7902756482eb5909af768bc3d17 : SUCCESS in 29m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0220b6250d0a446889730e5d0e452a08 : SUCCESS in 37m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/99a00ebfae0d461abc10d2a51edf0ff3 : SUCCESS in 12m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9c324597d9a4e7da512fb66824699ef : SUCCESS in 10m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0eefe102b7804b3facf09227d28ebd48 : SUCCESS in 19m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56506a4a34e341ab9f757463d408058f : SUCCESS in 16m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ee8fc590c4387268a91bf09806087b3f4cd47edd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 16:18:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"55db3cedbd9d80069a53f5cb6b2ba6eadb2a04c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 16:20:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef0341cd7e6841908cb59a0f94ce53d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33efb180a6b04274800f7402bafbafde : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebb51f57091635fa8283a740365b78aed15be32f","revisions":{"ebb51f57091635fa8283a740365b78aed15be32f":{"kind":"REWORK","_number":1,"created":"2025-02-20 09:23:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/942346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/942346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/942346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/942346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/942346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/942346/1"}}},"commit":{"parents":[{"commit":"1bb0209cd5d121d2baf6dd551fa13c29ee19b9c2","subject":"CI: pin to ansible-compat\u003c25.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1bb0209cd5d121d2baf6dd551fa13c29ee19b9c2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-10 13:57:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-20 09:23:16.000000000","tz":0},"subject":"CI: create ansible.cfg for pre-upgrade deployment","message":"CI: create ansible.cfg for pre-upgrade deployment\n\nChange-Id: I1179362e7f2b42c638ceb037bfceb3386daeed85\n(cherry picked from commit b7586c8e97ceb69fd22fd07a5b6f51c71d6fee1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebb51f57091635fa8283a740365b78aed15be32f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebb51f57091635fa8283a740365b78aed15be32f"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
