)]}'
{"id":"openstack%2Fkayobe~942579","triplet_id":"openstack%2Fkayobe~master~Ia87582a06a241b1be3822af3ba5165f9adf0cd2c","project":"openstack/kayobe","branch":"master","topic":"ci-slurp-overcloud-bm","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-03-11 22:13:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia87582a06a241b1be3822af3ba5165f9adf0cd2c","subject":"[DNM] Test kayobe-overcloud-upgrade-slurp-rocky9","status":"ABANDONED","created":"2025-02-24 12:24:03.000000000","updated":"2025-03-11 22:13:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1b8a7d833a5a18d8e2d9954e5ecd296d2d9c7704","_number":942579,"virtual_id_number":942579,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-24 18:48:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-24 14:26:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31a2138b9ff47868b17867e881a7b318c427f77f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-24 12:24:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b2da71e944e4cbec107d01265d688bed7e42eb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-24 12:26:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a9488c744b020e43136175c43a385149e9a4e5e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 14:26:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c411429c71d439ba879862a56f8f926\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77a28ff3a6d14b4f9c925dc8c899f04c : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/996309013b134b819ff381a85efbebb6 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2117b3d0a09245dc9ca8e2e6b3626673 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bedb0cc0234647249a90c10bbfd349b3 : SUCCESS in 7m 14s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1fefdacd892c46909b47383890f75dd4 : SUCCESS in 1h 58m 27s","accounts_in_message":[],"_revision_number":2},{"id":"80732430d25a016d4c83e96faa29972d205d7457","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-24 15:25:16.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c831d439c896354235818e308ade617f5d900fc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 18:48:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3fa2cabb3f14f0bbf7c7d2efe72d82b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0419ef9fa25b477481d5dee83116d36c : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72e4d5275f6f4e44ac386c030fe1c014 : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b1b8215c5664f23b78e65118032f082 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d089ba66c4b461e934c1b07e72c29c6 : SUCCESS in 6m 52s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/02bf2f7171cd4debb97b67589e5a208b : FAILURE in 2h 47m 59s","accounts_in_message":[],"_revision_number":3},{"id":"1b8a7d833a5a18d8e2d9954e5ecd296d2d9c7704","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-11 22:13:03.000000000","message":"Abandoned\n\nJob is passing again","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fb0b0c91270385f8f49791b883be451f4e57286f","revisions":{"6cbe521bb8b69c0afe2f12521f8d20eaaf6a081e":{"kind":"REWORK","_number":1,"created":"2025-02-24 12:24:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/79/942579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/942579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/942579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/942579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/942579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/942579/1"}}},"commit":{"parents":[{"commit":"cbbd23fa6b3dcfb7a726f6026c1814e46e6a57de","subject":"Merge \"CI: create ansible.cfg for pre-upgrade deployment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbbd23fa6b3dcfb7a726f6026c1814e46e6a57de"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-24 12:23:48.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-24 12:23:48.000000000","tz":60},"subject":"[DNM] Test kayobe-overcloud-upgrade-slurp-rocky9","message":"[DNM] Test kayobe-overcloud-upgrade-slurp-rocky9\n\nChange-Id: Ia87582a06a241b1be3822af3ba5165f9adf0cd2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6cbe521bb8b69c0afe2f12521f8d20eaaf6a081e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6cbe521bb8b69c0afe2f12521f8d20eaaf6a081e"}]},"branch":"refs/heads/master"},"0a0c8998d0de4d0101611ab8dcb11fae12ce75a1":{"kind":"REWORK","_number":2,"created":"2025-02-24 12:26:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/79/942579/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/942579/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/942579/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/942579/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/942579/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/942579/2"}}},"commit":{"parents":[{"commit":"cbbd23fa6b3dcfb7a726f6026c1814e46e6a57de","subject":"Merge \"CI: create ansible.cfg for pre-upgrade deployment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbbd23fa6b3dcfb7a726f6026c1814e46e6a57de"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-24 12:26:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-24 12:26:30.000000000","tz":60},"subject":"[DNM] Test kayobe-overcloud-upgrade-slurp-rocky9","message":"[DNM] Test kayobe-overcloud-upgrade-slurp-rocky9\n\nChange-Id: Ia87582a06a241b1be3822af3ba5165f9adf0cd2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a0c8998d0de4d0101611ab8dcb11fae12ce75a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a0c8998d0de4d0101611ab8dcb11fae12ce75a1"}]},"branch":"refs/heads/master"},"fb0b0c91270385f8f49791b883be451f4e57286f":{"kind":"NO_CHANGE","_number":3,"created":"2025-02-24 15:25:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/79/942579/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/942579/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/942579/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/942579/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/942579/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/942579/3"}}},"commit":{"parents":[{"commit":"cbbd23fa6b3dcfb7a726f6026c1814e46e6a57de","subject":"Merge \"CI: create ansible.cfg for pre-upgrade deployment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbbd23fa6b3dcfb7a726f6026c1814e46e6a57de"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-24 15:25:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-24 15:25:09.000000000","tz":60},"subject":"[DNM] Test kayobe-overcloud-upgrade-slurp-rocky9","message":"[DNM] Test kayobe-overcloud-upgrade-slurp-rocky9\n\nChange-Id: Ia87582a06a241b1be3822af3ba5165f9adf0cd2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb0b0c91270385f8f49791b883be451f4e57286f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb0b0c91270385f8f49791b883be451f4e57286f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
