)]}'
{"id":"openstack%2Fkayobe~942980","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I1086dc38e5137de3d1c03ba61088f0d04ef1f1d4","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-02-28 16:56:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1086dc38e5137de3d1c03ba61088f0d04ef1f1d4","subject":"Pin cliff below 4.9.0","status":"ABANDONED","created":"2025-02-28 08:53:40.000000000","updated":"2025-02-28 16:56:11.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7bbcd97fa9083bae2590813af75b43a17b3f615f","_number":942980,"virtual_id_number":942980,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-28 13:06:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-02-28 15:21:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-28 10:39:58.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-28 10:37:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-28 15:21:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"72494af2399b7cbfc1276ba8a8d1de1fa913aa94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-28 08:53:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b76ed09a0d6fc025abf9941bb8f27c70cdbcff74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 10:37:55.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/942979 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"aaf489bf144968f8d3e593406f825ff418996442","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-28 10:39:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bff26ffa200be7666fd1d8379a52f30f2892feec","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-28 10:39:58.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"48dd1d58793fdd5b177bc51a9994b8c2deb94cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 13:06:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80c8a9af54884d92974d3176bd264482\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8efddb87a5bb48e9a8526537b5bae8e0 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e633d34005d47768c294de4bced49c4 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/adfb1e78ba8842b1ab9367323890c283 : SUCCESS in 6m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b598d402a1004a61954812e6eaa32a70 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad4e4dbdade147f4975c132f90cb7277 : SUCCESS in 8m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/87d1c1664e9f42659a1026849406ab91 : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0408823414a34891a6d92ce29afc494c : SUCCESS in 10m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04e75b9466c44515b69af5ab2fe08329 : SUCCESS in 11m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ad249e3d22644077806a2985e85f859a : SUCCESS in 1h 08m 08s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3ba546b2a6464691aa08d676e7c07fdb : FAILURE in 9m 08s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b449220bfc745d6b39ec26b1073e92b : SUCCESS in 1h 11m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/964218d92d3843c19c77129088863f2d : SUCCESS in 1h 20m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9693e30e0d5f4c7c8b133815c4cbfa66 : SUCCESS in 23m 54s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0e48a21b685d48e98e63e2e9c4ece022 : FAILURE in 9m 40s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73b7044f85e8459da23bbde2ade77ff0 : SUCCESS in 19m 05s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/277bbcbeba8548e8b4cb4dab787af2db : SUCCESS in 1h 47m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/945dec56ab044b7d9972ced5a30825fd : SUCCESS in 2h 23m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/132f3849a0b74c4a9d3305b2bb22ad0d : SUCCESS in 17m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3da8ebb160e468baef92442dd6a8ebf : SUCCESS in 24m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/50f5a66bfaf34c218458b06f947b307a : SUCCESS in 40m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e1a23c7cc9b1435fbe8ea027b6d88b47 : SUCCESS in 35m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b09b3f93e72b4d15a450133aa32d93bd : SUCCESS in 27m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6f338e459d1b485085bbbe3c70250202 : SUCCESS in 20m 18s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/2f922199a75848dc9201057b15ea5918 : SUCCESS in 16m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73ded25f4dac4b1db33d0450fc4b1e93 : SUCCESS in 17m 06s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/f33914ae842142a9ad62ebf0deeecd1e : SUCCESS in 21m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/552c14479a794178890dc7b6353d13f1 : SUCCESS in 23m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8985d3f080844cc98bb6aa0fe5da1c99 : SUCCESS in 20m 00s","accounts_in_message":[],"_revision_number":1},{"id":"25a1d6f055ebfc45124e21e9415efe1b5764ceef","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-28 15:21:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7bbcd97fa9083bae2590813af75b43a17b3f615f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-28 16:56:11.000000000","message":"Abandoned\n\ncliff 4.9.1 was released, which fixes this issue","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"375ea12ea0eeabddf930be86b8368856f740994c","revisions":{"375ea12ea0eeabddf930be86b8368856f740994c":{"kind":"REWORK","_number":1,"created":"2025-02-28 08:53:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/942980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/942980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/942980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/942980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/942980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/942980/1"}}},"commit":{"parents":[{"commit":"a1490bd592079971d0a4ab83cd252ecd4a2b2e51","subject":"Replace pause with chronyc waitsync in ntp sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a1490bd592079971d0a4ab83cd252ecd4a2b2e51"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-28 08:53:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-28 08:53:14.000000000","tz":60},"subject":"Pin cliff below 4.9.0","message":"Pin cliff below 4.9.0\n\nThe 4.9.0 release of cliff breaks all kayobe commands with:\n\n    sequence item 1: expected str instance, NoneType found\n\nPin below 4.9.0 until we fix this issue.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/942979\n\nChange-Id: I1086dc38e5137de3d1c03ba61088f0d04ef1f1d4\nRelated-Bug: #2100579\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/375ea12ea0eeabddf930be86b8368856f740994c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/375ea12ea0eeabddf930be86b8368856f740994c"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
