)]}'
{"id":"openstack%2Fkayobe~943333","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~If7c727b98c3b821a40062520e6c8e111d44ac1c0","project":"openstack/kayobe","branch":"stable/2024.1","topic":"bump-michael-rigart-interfaces-stable/2024.1","attention_set":{},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-03-05 13:35:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7c727b98c3b821a40062520e6c8e111d44ac1c0","subject":"Bump Michael.Rigart.Interfaces version to 1.15.3","status":"MERGED","created":"2025-03-04 16:14:03.000000000","updated":"2025-03-05 13:36:40.000000000","submitted":"2025-03-05 13:35:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943333-bump-michael-rigart-interfaces-stable/2024.1","cherry_pick_of_change":943238,"cherry_pick_of_patch_set":3,"meta_rev_id":"1dc5b136015a7e25ada01adb4eb68cbc50448ae8","_number":943333,"virtual_id_number":943333,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-05 13:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-03-05 11:54:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-03-04 18:49:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-04 17:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-03-05 11:54:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-04 16:14:03.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"state":"CC"},{"updated":"2025-03-04 17:03:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-03-04 18:49:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-04 19:08:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-05 11:54:10.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"0a7e760fc40cc080169c64a5820555f7f05cca42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-04 16:14:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"56002a643bb9b629511ba30b2b51a2d7add27c2d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-04 17:03:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df16a14ed6c8ab92c1d2a7b8080ba5dd322f7050","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-03-04 18:49:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e431a7c747bc9f7f417ba33f4470ed33b0e824c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 19:08:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36bc404eab8b49c7b1a18849f90ffdcb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/618e039d94664660ae131532d0adc7bf : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad45b3ab3c04855962436796d98f251 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bceba2c7e1945999be533ffc5f6eac7 : SUCCESS in 2m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c9c993d8f6fa41b3a04cf07c7cb0c35b : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/714bde28159b465b93147659e7af7894 : SUCCESS in 13m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bfbd2bd120db427d8ff6a1be5fd4f72d : SUCCESS in 10m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4143dd10b16a44d38087d13652785217 : SUCCESS in 58m 18s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/64a5d397404e4f0e813b09c2fcdfd47b : FAILURE in 8m 54s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d9418a37ff2407b886086259fbd650b : FAILURE in 52m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/783fbdb698904ec6ad3d88c3afa39abb : SUCCESS in 37m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/88884fdfe89643878c4a61cc0e0fcde4 : SUCCESS in 17m 46s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2c7e2d769e9a4fecbaa04ae4da1dd34f : FAILURE in 12m 00s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb7c576923bc4d04a06653f64dfd1b8b : SUCCESS in 13m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1940d40c308c42a6a3d712b2aea105cc : SUCCESS in 2h 03m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27260bcb520f4dc6b9287ff4bec00daa : FAILURE in 1h 40m 07s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3c20edc992a24ec5b2a65ed3ea169f20 : SUCCESS in 2h 23m 26s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b059807c8484457b87a91e43004f7d1b : SUCCESS in 2h 08m 22s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d7ea45ae7b5e4b719c7933a62eeaf52d : SUCCESS in 16m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44c2630f7f6d43a5871d230fcad97526 : SUCCESS in 13m 23s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2267b064ecd54b4fa7ce53a4ffa67012 : FAILURE in 25m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d9c8426c162d403391dd868ab4232518 : SUCCESS in 28m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91b58470c22c4255ba17fee83ab2d2f4 : SUCCESS in 24m 15s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/fbb9ac3b3b7c429c8df85e11aae9dc43 : SUCCESS in 30m 42s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca1ba34863584b86b2c60066ff114398 : SUCCESS in 17m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ed313c7db8b4ebb8b90048863dc1f66 : SUCCESS in 15m 08s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3dec6c092fbb4b30a5c7bdbb53e23a93 : SUCCESS in 12m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/663215cec9744488b33feb1ca72fe20f : SUCCESS in 12m 24s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/9fd64826a26441c98fd175e760ebe921 : SUCCESS in 12m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4f3e5e90c81744898b3c36c8abd60e69 : SUCCESS in 18m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c9a45de567e437cbe1d42f790a6371d : SUCCESS in 16m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d4fd08e7dd1442ab43c78b63db1a7cc : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a2a5625490242799faa00d26057d69f : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ee5bfe72bc4424c9a69c52dc8a06654 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a6066323812418486311e8cb66f9af5 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"94008caa40eb55d5c5593de871328730ed87de46","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-04 21:16:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f4f1ac4fbf072614d040cad0bbbd5d8aee950ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 23:32:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/867eda9837b54686a9ead21e2faf56f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcd20f3e1e304f17827bf87c49e39b24 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff30cfb39e3c420bba0f3a4576c0b572 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64fb2d01eddd44a4b7554a8c00550e16 : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ce5c7daf3b94f578b133378da89c728 : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c157170761fa4f5583bedb8db12bcf80 : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/773952add51e4ebfb25a17a75a1f33e1 : SUCCESS in 10m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1c981e34958b4647844e6551496b024e : SUCCESS in 1h 05m 26s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9aec3651b71643e69130fe4d71fbeb9d : FAILURE in 10m 44s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/347581419afe4c12a8bf472f0e856a81 : SUCCESS in 56m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/88e30fe4c71b46b78b50aa93977741ad : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/31ee33e3ae9e458da601d0d25c883f73 : SUCCESS in 17m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/47a862e344b24b268417364423ac06b7 : FAILURE in 10m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e0eaa1a54564b0c9779c9c8e65a33f4 : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1abb7a2353534c3ea589c8b945ea1c2d : SUCCESS in 1h 55m 11s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa52d1a4a37a4989aabe270c57d00a6c : SUCCESS in 1h 22m 21s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/aa93736ccf7849e9a699a995973180f7 : FAILURE in 1h 37m 03s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9035584225b144e2ad547f72696ac820 : SUCCESS in 2h 09m 35s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8184c6868117466ca20a869ae4d9a493 : SUCCESS in 18m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52f8f741e1d0495e86bfa1cb2bfe544a : SUCCESS in 18m 04s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/172f724bae3842cd9eb60eeda397468d : SUCCESS in 38m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/269e5f98ddb045f6825b4d8e684ff7a7 : SUCCESS in 23m 04s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ada6977275ee4516985ffc5a17b2e707 : SUCCESS in 16m 17s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/116f9db1bb40428285accbda297d15a9 : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cf1dd68dae2495ba12cd6dfd86d203f : SUCCESS in 24m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f796120306d446c7af84b25d548c69c3 : SUCCESS in 11m 40s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/522f82af199d46f58ed3a1a54d3cf038 : SUCCESS in 12m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3357ca1a64054f4195a1fdbb068f043a : SUCCESS in 10m 05s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/6d2fb3b836fd4d64aa799f59b5c9d9f7 : SUCCESS in 9m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ed929862dec34765b0c16179d2be5784 : SUCCESS in 15m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0728a7b923654af3bbd78aef4fea5e0d : SUCCESS in 17m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7707894fc37f4be4ace616ab24d52b71 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f228de5b2864e2d93497f4d6ce2b879 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83e6264af8244b12ab2d084ff222593a : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0035ed366f4b47d088f11d867371a53e : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"29605187bec7a2cba6aad6073bd8de878d4eb1fe","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-03-05 11:54:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1787f9b3c6425d7257f8494febe4fd6396f0b464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 11:54:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1a898831a0c5b31fcc738c8a36ab54e97f77f13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 13:35:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd583a8a9d744702a3226c63fe8cedc1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac57cb1a583f40c6b8527445a63878e3 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce82c12be834408382c7840ca2477586 : SUCCESS in 2m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/efadad1c3796404c906094f0d36271d9 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9689983b7e44507bb12d91ef4e85f80 : SUCCESS in 12m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bb8a2823fc2e418aa0dd20131d850bac : SUCCESS in 8m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4f501db010c3453d8602e8f302217fbd : SUCCESS in 1h 05m 26s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d312cdd08fc437d8afcf61af20a5675 : SUCCESS in 49m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ba2429e1f4c749329d5ea30dd47f67b7 : SUCCESS in 1h 12m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7699fbc99b82448aa473f0cb5517ea40 : SUCCESS in 14m 26s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/875b3b2a46e447b797dcde3ffeedce0d : SUCCESS in 14m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d9116615c4a14287bf4c526f00c07a18 : SUCCESS in 1h 11m 19s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e11f2504997f4b838d22bf3115262974 : SUCCESS in 1h 40m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/18083dce1dc044ef91182ac156bd47c0 : SUCCESS in 15m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9209120060d5412aaf566773df91deb4 : SUCCESS in 16m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/45b545ee55fc4d718692b72bda1d78f0 : SUCCESS in 21m 34s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7cf37fb5c604030b9f1bd645eff871f : SUCCESS in 30m 25s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6f4e179b0b654cfe992875a28b978fff : SUCCESS in 35m 56s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b87d9b6519146439118d6084bb82538 : SUCCESS in 18m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/693797b0abfc42ae98b039b1516fc2a1 : SUCCESS in 17m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73d38789098f40c8bb02f98669b5d236 : SUCCESS in 15m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7727648813a94e5c829684bcaffccf3d : SUCCESS in 23m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81347de253c14c62891b7d6618536032 : SUCCESS in 24m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d45e3cde83a5937d09c70bbc1c9b5c9e2abc750e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 13:35:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1dc5b136015a7e25ada01adb4eb68cbc50448ae8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 13:36:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb32c79f693d42bfb53bb26894fb150e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aec1336cf004db8b15710585b16d61e : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a071aa73c6594b6da17363dfe0040b1b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd92924f04afcba0475613988c51e8b890c8b1b8","revisions":{"dd92924f04afcba0475613988c51e8b890c8b1b8":{"kind":"REWORK","_number":1,"created":"2025-03-04 16:14:03.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/33/943333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/943333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/943333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/943333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/943333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/943333/1"}}},"commit":{"parents":[{"commit":"e57200ece671be69ab7e45da2aa27afb904727c6","subject":"Support forcing time synchronisation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e57200ece671be69ab7e45da2aa27afb904727c6"}]}],"author":{"name":"Dawud","email":"dawud@stackhpc.com","date":"2025-03-04 10:57:53.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-03-04 16:14:03.000000000","tz":0},"subject":"Bump Michael.Rigart.Interfaces version to 1.15.3","message":"Bump Michael.Rigart.Interfaces version to 1.15.3\n\nBump Michael.Rigart.Interfaces version to 1.15.3 to fix an issue where\nNetworkManager was not restarted before bouncing interfaces.\n\nCloses-Bug: #2100792\n\nChange-Id: If7c727b98c3b821a40062520e6c8e111d44ac1c0\n(cherry picked from commit 80494bf7da3a060194370b9782b54851cb89f135)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd92924f04afcba0475613988c51e8b890c8b1b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd92924f04afcba0475613988c51e8b890c8b1b8"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
