)]}'
{"id":"openstack%2Fkayobe~944142","triplet_id":"openstack%2Fkayobe~master~I777e3caa782a3a9b77699a668b27b4334040efda","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-03-21 12:53:21.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-03-21 12:53:21.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-03-19 14:35:02.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I777e3caa782a3a9b77699a668b27b4334040efda","subject":"Bump version of stackhpc.linux collection","status":"MERGED","created":"2025-03-12 17:38:31.000000000","updated":"2025-03-21 12:54:29.000000000","submitted":"2025-03-21 12:53:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944142","meta_rev_id":"b4c084aed9f8aa47020cb762a5300c206621f4c9","_number":944142,"virtual_id_number":944142,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-21 12:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-20 09:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-03-20 12:49:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-03-20 12:49:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-12 18:41:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-13 08:04:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-03-13 11:39:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-20 09:40:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-03-20 09:40:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-03-20 09:40:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-03-20 09:40:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-03-20 09:40:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-03-20 09:40:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-03-20 09:40:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"befaa647b24b241a3163475f155003762e595208","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-03-12 17:38:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22353df82711aeb6afa0b8be7f28b7245b89f1f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 18:41:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b22264e4a647411cb871d3f179676ace\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c59b6ff28e047f39259e4e4c85f925f : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e876860429b64533a172af514eae328e : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a95e3d47ceb74dc2aa6f1b7e4109418c : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/034429a2a68e422da10f6fc5580e5981 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e155abdee0e7422882ecbb67d970c013 : SUCCESS in 1m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e98ed924af2344fb9c89cbae53820808 : FAILURE in 3m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/288d018b7bfe4dc5b6307886fda4abcc : SUCCESS in 13m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2264813e6bd848aca57e7e56a92b1e06 : SUCCESS in 8m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ebf59397f11a4adb9f62dbffd1c04b44 : FAILURE in 7m 27s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/31b094cf024f46109ee74418008cb905 : FAILURE in 4m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2aadfa397d0f4977836720abe81e8c8e : FAILURE in 8m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2f2e72b866ca42b688a85104ce4a88a3 : FAILURE in 12m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7c86d56647774d03aea9a1a2075e96a8 : FAILURE in 7m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c34b21785e144f0d9e215620e8ce6d65 : FAILURE in 4m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47374d6904cb4d10b13b09bf54c84410 : FAILURE in 8m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c073cf52d77f41f3b7f1c22b4ace0d5f : POST_FAILURE in 1h 01m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/962355a3da134d9abe8b8873f634f9b3 : FAILURE in 58m 34s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/21ed8d86b3634ff4af7d1aab0786148a : FAILURE in 1h 00m 27s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f691694a0eed40de88e6ec46a07e4c49 : FAILURE in 7m 54s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/c0d347a0e6c04d67814991fe4ca82dbc : FAILURE in 5m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae44b468067545acae2241ae6a595500 : FAILURE in 4m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b047295202ee4a189991a9ee8c413447 : FAILURE in 4m 57s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/49d63d26f234484397105a716d351c7b : FAILURE in 7m 30s\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/e003fdc72b4841e6b13fa957718a8051 : FAILURE in 8m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/55d9f155be52421e84291a3fcac02bfc : POST_FAILURE in 27m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc0e244f0f1743d595ee610fa9c91867 : FAILURE in 15m 13s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/fc1e994ec9c243fda1f125e156043333 : FAILURE in 16m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/95d134650af24065a7e7d3d47b68d842 : FAILURE in 14m 52s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/2a1d78b0f2674fe0aa237605662c425f : FAILURE in 7m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f37291042603448d9ea6f63096561605 : FAILURE in 3m 48s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/95ee06e64fb241a69d6bba56978af98e : FAILURE in 6m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5403154933d2480eb7d90b5cbab28d92 : FAILURE in 7m 27s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/662b277bc0304d4b9b8390c0032cf77e : FAILURE in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1d92cd4d246ea06c4552936434dc47f5c4f5b171","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-13 08:04:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba9ec488b93f73599a3a5e3dcd8afdcbecedf4f0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-03-13 12:35:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e8eaa654dbb0320386447d692852ed949c32b70","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-03-13 12:36:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c655a5294df9c6d07428ff590f524f4bca73b70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 14:59:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bf9486c7341465baca3d8287f40b453\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6f83fba2d7e424fa020a22b45f230ff : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b180ca1908a4f409c1a60d8bf0e0e7b : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/002cc43a56454720b220ff5872aed293 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f774cd3cea44258bde28e1580043510 : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0abc71759194688b29c4761f43a25ab : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d6f1803575d249a191e7d7cbb8b1667a : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/692f9be5ba8e4075b40729f20a7cb823 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dd69ed56a9ab4106a7e011dd4ce7ef30 : SUCCESS in 12m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/32b51f959e6c4b29ab1a72ae307cfdec : SUCCESS in 1h 12m 54s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7c325245a1f64ce8a688223628a75dfa : SUCCESS in 58m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daea0d6622c64dee81c4cc648ed0b186 : SUCCESS in 58m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8c6cf0e0e87f471ca922faf168ee23cd : SUCCESS in 1h 21m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/be123566ae5844f1b94ff7f3c99742ae : SUCCESS in 1h 10m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/679a7422d45a4d318344654388c5b0f0 : SUCCESS in 18m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47844546960d444594bbfe2592ec8313 : SUCCESS in 14m 11s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dc72b47ac99143e3bf0d1236401949cc : SUCCESS in 1h 48m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52fdb00e0c4444708da75ded9ce8a22f : SUCCESS in 2h 08m 06s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/fe2da8381cc941bc81b2cf23d33072de : SUCCESS in 1h 53m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fb59a958471b4a7a9c35f452f6e33622 : SUCCESS in 8m 22s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ca3f815cc1134db99787bf5d47307601 : SUCCESS in 24m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff585b11f6484a42bc2b8592375600d2 : SUCCESS in 28m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1d5ebd4ca12e47fdbaf9e80072dc51d5 : SUCCESS in 21m 04s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2a350d3d2bf44bf48ef136469d10630d : FAILURE in 34m 37s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7e0f2c7d3b3c4275b7656deb524b1cba : SUCCESS in 29m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03ba6ba1b62e4cb786fb0ddaf08f26ee : SUCCESS in 35m 11s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/391510695fb54e0e9213cd2cfc5d48fa : SUCCESS in 29m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b9670177f4d649c0a24fd448dc7c2e20 : SUCCESS in 21m 34s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/7ccff9692c88481080fb6feb28da7204 : SUCCESS in 19m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5148e5f17c3846f68d16e03f360a58e3 : SUCCESS in 10m 11s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a30cbd40e8d94125a921fc457ec00477 : SUCCESS in 11m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/af813f3b81e141ff963787b7307a9d41 : SUCCESS in 14m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73c89ce1431a442aa16db507621e7daf : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"be5188dcda7462e504492ee8220c50e45a488195","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-03-17 09:22:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d479bd885be78913af806e820ffabb075bc3f46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 11:13:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b82e12a9597c42c4a56db5f68175cd85\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/655027aa3413409a94fa96b1db80c284 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/832c45a8018b4757bfbe8d33b3d04085 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c429370f5e9345a4ac2f695dcb7c38ea : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bbebfdabc1345cf962eded379c39ff8 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43cc40ad64ac4d02a80b59cc61bdb315 : SUCCESS in 2m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/57db5d01afbf401ca12f7a5b4adec642 : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b45676d1f05547098027a6ae17f36111 : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/87cc9682d8604addbf2cad3dd30ddda6 : SUCCESS in 8m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c0e45d5defe243e08171d42fbfff1a61 : FAILURE in 48m 32s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a346e4cb6c1842d1afe80e23dc75072b : FAILURE in 1h 05m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b835bec30c5343e48a978263c8708a70 : FAILURE in 44m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/86f10a20198d45d3a4da63969b56b464 : FAILURE in 47m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2d0cb184c2104a62b07ac4baf9166953 : FAILURE in 1h 07m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/93c437023ceb4d7aa0d003ba2e22840b : SUCCESS in 14m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d331e481074e4d85b15cffca2414d653 : SUCCESS in 11m 56s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/65c431f061c04de883e49fa584d35c21 : FAILURE in 1h 08m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f85273d228d4938a74fd7d67699447c : FAILURE in 1h 49m 52s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/edae3f9702524556b8ec1aba1b7daf25 : FAILURE in 1h 10m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d0c974642b4345f7a9f6e66e3af01efd : SUCCESS in 21m 06s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/a42a60cdbc604a20bc42f6d9be5ee823 : SUCCESS in 24m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56df2075cc2f47708c18370b1ed719dd : SUCCESS in 14m 02s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/793e0d0d5bc045d683435e1f7f38bb48 : SUCCESS in 14m 06s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4a347fa690524d5eaf3e52cdc1f8e10f : SUCCESS in 33m 51s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5ab22f8810b341d094c3c5950ab88c81 : SUCCESS in 29m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1bf8a0431f946e7b7422dc193735368 : SUCCESS in 33m 31s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f51b26f35c2944cab4e7cba5c74f79f5 : FAILURE in 34m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7cbcaf763c2744c4a82a772db5adab60 : SUCCESS in 18m 39s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/6788d37a6a034ee6902bdf0c6c8eddee : SUCCESS in 11m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/599a931100184296be2c464a86648eda : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/da020f8e607443acbba622860f5e0890 : SUCCESS in 10m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dc8582c11e934ada83896abf77c97311 : SUCCESS in 18m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7aa2cadf5bc845a3a6781664fb455653 : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"dac26caee17e861c39e0b50d8f2332c3472d4f22","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-19 14:35:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7981fc170e0c3a90fccb7814376854a8dbee9c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 17:46:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e570d18e78ff4124bb28dab3072b8859\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03245fc7eaee40b4a67a1b90ba785dea : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f75d0d0fabc4ff3a58c7662159584a2 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f147184bbcfd46a7883a4a690036be2a : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c4dbc79ccb49d9a10330882d3bc186 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bc62e2637d74863ad6d8aa2f89f2422 : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/91beb676e3ab4d28ba5f3bfcfb5cabb8 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/645b2a00cd7546339a2276fd08ed83e4 : SUCCESS in 20m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/834b932740aa47058d5eecc19e93e6ea : SUCCESS in 6m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2cb6c621a59843bbbb48d35756fd5d34 : SUCCESS in 1h 04m 28s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/83b1edc5334d4ee8bb8e3c48202418d5 : SUCCESS in 40m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f3ab1436ffa45c59e5d62430f545330 : TIMED_OUT in 2h 02m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/45be850e526144bb89ccb5f395a89dc3 : SUCCESS in 54m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9b4238764bae4affbad2062b1b103358 : SUCCESS in 1h 09m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7fe0740a52674395a46fb23c8cc92f25 : SUCCESS in 17m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d3484eb04854b81a6939dbc388f0b5a : SUCCESS in 14m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/85dfa3bf9df04679b230b84d648fd152 : TIMED_OUT in 3h 02m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f2de947466a4227b6c95ff377e9acf4 : TIMED_OUT in 3h 02m 51s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ee3eade253e1404eaaeb4928110e1716 : SUCCESS in 1h 57m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3dcc83bd71204c0eab08da874c5037da : SUCCESS in 20m 14s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/4eaf3c18146347698dab46df22c5f9b8 : SUCCESS in 19m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9696f0d332de4e3aaabcebd6fafd5f80 : SUCCESS in 18m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4af6832304754431a1f98278f70dca16 : SUCCESS in 18m 14s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/752f5065e06f40e18c4c8f0b1bb91871 : SUCCESS in 41m 39s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ca34207a3562413dbbc6582249c374ab : SUCCESS in 34m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11274dbc7664499ea400827678a86b0d : SUCCESS in 33m 12s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1550330bbf5e4c6a9474edcb462dfb26 : SUCCESS in 36m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0d328e183a514ed1bc9e764e672376dd : SUCCESS in 19m 43s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/5020916076994bc7820de4d9cee5c91d : SUCCESS in 17m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b09e30e9570f4e4a9d6b3d3c37917381 : SUCCESS in 16m 39s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/9d71457c6f3d4adc8230e00991dcb8a8 : SUCCESS in 17m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/deb816d117f447b899ba69a92006cc6d : SUCCESS in 17m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f63cde065144e7f86c5ec246a1483d8 : SUCCESS in 24m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8cd7980877a9324c760f4109d047318e31250849","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-19 20:15:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f14c346048b740b8853fdbffc7af05a842f00a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 22:26:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9a4c45149914811bdfebdb6f6437321\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83834a3286ac4bd6bbcc1c7e45f84255 : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af26bc659204be7902ee6622456f46e : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e48de73eeb3433e9edbeb778988a139 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16017e5148a242c0911838ee4d8f195f : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c245bc276345451daac43805e27443e8 : SUCCESS in 2m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a34ea779846448fa6e40cec7c755791 : SUCCESS in 2m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ac485ca7c08b42fe80b339037343230a : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d380ea84fe2b495297a3d881e4acb574 : SUCCESS in 5m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a68c881d94eb406586d3ab3d40bef210 : SUCCESS in 34m 34s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0823d7972d02493f987438e9457077b4 : SUCCESS in 1h 10m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d77dcbbe40a64ed8aa640838d4a7c3e5 : SUCCESS in 33m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1704867bb41b4ebdb817b1c419c0bdb1 : SUCCESS in 1h 00m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f8fb956baab34a169a75bb0af69209b8 : SUCCESS in 1h 08m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d7a9f3c47bbc4e5aa305abed6986a9f4 : SUCCESS in 18m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74de80952b90413b8a763b934b280229 : SUCCESS in 14m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e2a95e69cf9c4410a15625d5ec605c70 : SUCCESS in 1h 48m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18bc2e113164415386151f6669e740e6 : SUCCESS in 1h 57m 36s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/38dc9ffff50b4f6295f7da50f7690a17 : SUCCESS in 2h 04m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e5afc7b97b2b4165ae0eac8980a733de : SUCCESS in 20m 10s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fbbad7334b074e398624b73e81e3e721 : SUCCESS in 21m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/741216a7cada49988356879a97a87b7e : SUCCESS in 17m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cecfd95a9cbc4ac28706f203c10df1d6 : SUCCESS in 17m 24s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a03a609185084fd895bccdca4b76d4c1 : SUCCESS in 37m 37s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a860d6f64b1d4bcd864cab5cb1f97279 : SUCCESS in 40m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f67e650289e944b398a27459e61ceea4 : SUCCESS in 29m 02s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f41d812bbd5f430393362df6d1029c49 : SUCCESS in 27m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/69dc6af3df174362bd2f3d1c1134b68e : SUCCESS in 15m 46s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e52ec689d0be4c17a42be71ff5741555 : SUCCESS in 16m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/301689166e2040bd8da8e5728ad684a0 : SUCCESS in 13m 55s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/2f9f9355eacf4e978725e9ec77eb127e : SUCCESS in 9m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/23efe3b0690d44779345e85935f7bd56 : SUCCESS in 11m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e521ec3ce93f43cca19882d72f887368 : SUCCESS in 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8aa86f37bd5edc9bc0e653dfd59b4b944f1f8d2d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-20 09:40:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6124eda73cc2d001bc3770a068418c5299974ac8","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-20 12:49:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb013780903ee160d9cfa39fe3c87d72cf23b30d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 12:49:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca08294b96c6463e1d0978d689a4a91233e3db42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 17:40:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7449745bd4f84180b1435b3dc7f0683c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/488a743cb88946d696b4c4b79a0c1a41 : SUCCESS in 9m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df210ec486934b6a86427c9d5c79a67f : SUCCESS in 2m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6be9305fb8274126b36f64734bcbc3c7 : SUCCESS in 19m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8073211395b34ac3adacd04718158336 : SUCCESS in 2m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a2e14dd92c374271a0351b581df15090 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/528e3f10cdb94b568f4191ee4cf024f7 : SUCCESS in 6m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb042fa622dd451bba4819ad9d1c7d0a : SUCCESS in 10m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/409d3e6b23b34161a5dc6271edc7e5c7 : SUCCESS in 1h 21m 24s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f002e1e8182c4fa7b1d97144a60a0639 : SUCCESS in 1h 06m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3085922678c44a41b04cd4873a47f3d8 : SUCCESS in 40m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dd357008fbbc453ca01203750e94504d : SUCCESS in 35m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5e031f8700cb4e4d8e60a3e12de1ed0a : SUCCESS in 1h 09m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d33e444575424de28e6de8af94e630ed : SUCCESS in 18m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2b39bc546d7403d9cae4a1d3c1e5aa1 : SUCCESS in 19m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/aece996eb2664db88a4c1863e8365d19 : TIMED_OUT in 3h 02m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df90ec07c5704bf3905e3dd1e2e91119 : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/608d0d9c2f95463f96c01dd1f78eb174 : SUCCESS in 2h 03m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9afe7abcf38a4274979a6e37a1892836 : SUCCESS in 13m 19s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fa0f873c329a41ffba3d89c118f74880 : SUCCESS in 20m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3baf026318d14897be4d4e71e19bb27b : SUCCESS in 33m 38s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/baf49dae73694f86a88f7baa712850a9 : SUCCESS in 34m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8085aaac05f4440bb0b8174ff1a487c1 : SUCCESS in 36m 18s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/aef895d6aa86409fb6feb9da073e97ee : SUCCESS in 24m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2a8a69e7678842bcba1778ced8d4e479 : SUCCESS in 17m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f34d4664efc488d8bfb3c6b405c87d8 : SUCCESS in 16m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0158a3e915b54178826cec48c1fba87d : SUCCESS in 16m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a5f0e6e095e42089174e834a93c9f0d : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3e44205a81db39c49ad0a2525aedaf0f4c18a0d5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-20 20:01:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97aae2dd3cc19a0615ad3e72eda63758dfa44f15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 21:59:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac479858a9bd45a0be3d2b9babd543a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd287a8c6e7841eea98b35f706ccf809 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e8ff12c3fd148feab35595e12ea2bb3 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b2e8954365d4762b87643d5019b9de7 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888af0064b8a4d82ad4443d094eb5c42 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d36cfd32fd6148fcab3eb49db3066d9f : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9cdc16ee51de4bfeac610128e4dd92b6 : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6638f6ce018f4529b365037052305b28 : SUCCESS in 12m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b9c8aced0204abc9a75b7a984b7ae15 : SUCCESS in 6m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f1c156e520174cb8a8dc07b45038438b : SUCCESS in 1h 12m 01s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9c4f7044c6b14cdfaf16a1564f0ca326 : SUCCESS in 1h 12m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76f11da26f2c4d3b85437286073d06fd : SUCCESS in 1h 09m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6a507a08724c401daea19cc6d8f03fc3 : SUCCESS in 38m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8fbbabdf7d304538a1f22eb501a9287f : SUCCESS in 42m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1929a8ee03ea4879a4eb79422392a31b : SUCCESS in 19m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65f4b66b8b5d432592b347298cb8a128 : SUCCESS in 12m 24s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c21740c20296446e8ab7f649ecdcde58 : SUCCESS in 1h 42m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74797aa0cb4140c7acc6aeae7363601f : SUCCESS in 1h 37m 27s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/da186cce3abd436a93bbef06bdf49512 : SUCCESS in 1h 52m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ff61d7001c034e2e977bbe3ec3fb1d7c : SUCCESS in 20m 30s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/687effb1d4844d31abf9613a1d2fbf0c : SUCCESS in 20m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80b19619f72f417c8afba9ea89214d9a : SUCCESS in 1h 06m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9d5e33bd2f454530890f3f02cacfd9a4 : SUCCESS in 21m 29s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a5de3be40bb8442f95b94a0ba2b911e4 : SUCCESS in 1h 25m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/932a2c4a789d453780ab0d4b8d78985c : TIMED_OUT in 1h 31m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c24ac6cca5b493aaea061802f3dab09 : TIMED_OUT in 1h 31m 32s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b0da721406f6464c8e3e5e0c96a8110f : TIMED_OUT in 1h 31m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4b3dab0cddd84c6fb9b5a6150e4c83c3 : SUCCESS in 16m 33s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b6c89ccc4b03488882dfed939f47ad7b : SUCCESS in 21m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eab2304da20544849ffc37e30d93ca7b : SUCCESS in 17m 42s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/2a2e94e07c874d1ea8b4e88beafff8f1 : SUCCESS in 18m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9ec2e819e30247618b9ff338dd7964e0 : SUCCESS in 17m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/249fd5928f874c56943be69c786d9176 : SUCCESS in 26m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9eaccdbfff1715458812422e53d6bdce0d85fc8a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-21 08:16:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ecf2934c2134e3ae6a3bbf8afff46354524ffd8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:48:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c8b242d31de430aad665351950eff94\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/649bd31a37fa4eecb202b7f50ded130f : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ee8ef30e0a847c99c2b2522275da280 : SUCCESS in 9m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/153657f24077476c80b4d8fb0637387c : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2093d2ea4e244e6cbbd47a36b38c034b : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53c83f13e5b74e27bf687ce4f0e89b74 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4cbd392339304b65802a77bc71448bec : SUCCESS in 9m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0136e6aff442440ea1988a88af04fd2a : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c2d893786e44cd1af130c5cac16c526 : SUCCESS in 12m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2d50f3b6b9904a6584ce9c59ee16f1e6 : SUCCESS in 1h 03m 51s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0d08bf51cbff4d91963201c192e916ae : SUCCESS in 39m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd3612052252475f895ff009e9bde121 : SUCCESS in 1h 51m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59067df246b6470fb37b6aa69c55ffd4 : SUCCESS in 1h 09m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/35536e47a28a41ccaf90f37201adb819 : SUCCESS in 1h 18m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ac4d09617a9f4da092205ce054d6be9d : SUCCESS in 21m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f9c84d86fb34ec681952ceb7e1813e2 : SUCCESS in 13m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9f540e45139e46ad83665c0f0191ea55 : SUCCESS in 2h 06m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64af3e6610924888a4c461354f1444ab : SUCCESS in 1h 52m 02s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6f1ce94ce4f549e4ad827d6b7d25f9d4 : SUCCESS in 1h 58m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f41954e2cc154f55ac3ea2048525c7cd : SUCCESS in 20m 27s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/9eb6329f127a4404a8f60d9ea357acfe : SUCCESS in 21m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7043fb1a02f4601a1a58e48a6eac092 : SUCCESS in 22m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba4be374928a403a9180e0a56efdfe2b : SUCCESS in 14m 50s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/da2b17cb26fa47c2b667da7dace1e091 : SUCCESS in 39m 36s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5299d06296ca445293ee386a62d4a981 : SUCCESS in 32m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c1404ff1965496eb581def753bb4f73 : SUCCESS in 33m 37s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e7b030d15a364dfcbcf9737a3fce1874 : SUCCESS in 38m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ddab670e798743dbb785210f72a1b86a : SUCCESS in 19m 24s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/a7759dfa66f84bc581a4b2c7c3e8c691 : SUCCESS in 18m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47c69919453a4a7384a839629e468d5d : SUCCESS in 14m 53s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/dcf5a2d87a234ccc995ffc1d9c49ab2d : SUCCESS in 13m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0b88fdbc34dd4c6e905a6a637e71b188 : SUCCESS in 16m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ff159ed9d1041a99fe84f0b5d30fde6 : SUCCESS in 9m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e3dc1a11c32a5c3a7571a452a98416545597d972","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:48:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab5b045ff24b5944f91844bf864959e87fef7931","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 12:53:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04920eedeb8c44a5a890fddf30205dea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3c4c434a8d47d8a98e85b21d2ab1bf : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57ea0b5e564040b8b216b0b41a2d4da5 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/839ba5216612421c88621221b11b33e5 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/757fc570dc5d46d583f390058b69cf90 : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aad636d7b9de44a88a8559822b4f59dd : SUCCESS in 2m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f996115c9e5240b88fd1a02fd213bff7 : SUCCESS in 9m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dcadf75996384f25a6e3e6fd91875f45 : SUCCESS in 5m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7b3dc6f83be944dab509311dd3e887f5 : SUCCESS in 1h 00m 05s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/b0881c6fa91e4a0b8bced8bfb9379b48 : SUCCESS in 40m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1338a26ffde94ea2980788b9ad07d073 : SUCCESS in 48m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3102acfb88f4440297bb4d00f74da63e : SUCCESS in 1h 00m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/30eb312599724068b7c1c4e2bca28428 : SUCCESS in 1h 12m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/87e2cb5eeb0942c4b315e8ccce7a34ac : SUCCESS in 18m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1519ffc3a7c4ca2a0c9dcaa9040057f : SUCCESS in 20m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ec9f37471c204ff8bdf03f9771e87799 : SUCCESS in 1h 08m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f6bb46bf458406f8ab670e08a937a23 : SUCCESS in 1h 41m 39s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/43f34b8446ea4a55a4a131dcc4918f1d : SUCCESS in 2h 04m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c3f9441596dc434dbe9fe8661f846629 : SUCCESS in 20m 23s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/24a463969d5a4a5fb4f4aa054e28d548 : SUCCESS in 19m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac76f987d683472eb7316d462bfa95f2 : SUCCESS in 32m 32s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/15062f6fadc347eca0542a9c7ccd614d : SUCCESS in 28m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b406d98db81c479fa9b577bfa2b1935e : SUCCESS in 38m 02s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/10aa7cf5d0ae4d268cc4cd3e7a045fbf : SUCCESS in 37m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b74a84ecddc34991af1ab64c2e14fcf0 : SUCCESS in 10m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c8a63b3af0d4c36bb13710145264ed2 : SUCCESS in 16m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/78dc8e6aa55c4633bfee267302d1997d : SUCCESS in 10m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c86b17b17f9a46e1bfd3f16310a96692 : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c18bf176ed5a33fd4f3acac5c7459443b1468fb0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 12:53:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b4c084aed9f8aa47020cb762a5300c206621f4c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 12:54:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fe8a197082a4281b5710178bbc6a35a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ec052d29b5348b8bda70113c569f30d : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8a67e8ec92f47efb28036de29023fd7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05896059429c094735bd0bc96536ab968db62621","revisions":{"05896059429c094735bd0bc96536ab968db62621":{"kind":"REWORK","_number":1,"created":"2025-03-12 17:38:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/42/944142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/944142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/944142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/944142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/944142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/944142/1"}}},"commit":{"parents":[{"commit":"80494bf7da3a060194370b9782b54851cb89f135","subject":"Bump Michael.Rigart.Interfaces version to 1.15.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/80494bf7da3a060194370b9782b54851cb89f135"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-03-12 17:36:59.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-03-12 17:38:08.000000000","tz":0},"subject":"Bump version of stackhpc.linux collection","message":"Bump version of stackhpc.linux collection\n\nFixes a race condition in the vgpu startup logic by adding retries.\n\nCloses-Bug: #2102153\nChange-Id: I777e3caa782a3a9b77699a668b27b4334040efda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05896059429c094735bd0bc96536ab968db62621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05896059429c094735bd0bc96536ab968db62621"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
