)]}'
{"id":"openstack%2Fkayobe~945207","triplet_id":"openstack%2Fkayobe~master~If15bcfed7f212b9592049287514d8d601b70a41b","project":"openstack/kayobe","branch":"master","topic":"fix-netplan-not-removed","attention_set":{},"removed_from_attention_set":{"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2025-03-21 19:06:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If15bcfed7f212b9592049287514d8d601b70a41b","subject":"Fix netplan packages do not getting fully removed","status":"MERGED","created":"2025-03-21 12:31:43.000000000","updated":"2025-03-21 19:07:08.000000000","submitted":"2025-03-21 19:06:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945207-fix-netplan-not-removed","meta_rev_id":"7ad15a71584674266ce8fd1dc5bd5f3b06c0d08f","_number":945207,"virtual_id_number":945207,"owner":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-21 19:06:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2025-03-21 14:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-21 16:42:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-21 16:42:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-21 16:42:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-21 12:34:27.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-03-21 14:39:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-21 14:40:31.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-03-21 16:42:06.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"4b63219d0af1b9ce58ab1382803444d665af45a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-03-21 12:31:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0063b703e2d37ba869bef57d9aaf7ccef8ad5677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 14:39:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fcb69e84ecf4affb1d9ba51dcfc45ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a56291db42074e25ac97e8f37dcd4481 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3446ed71883a40ae96c98f99be9f2923 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21c5795e4fb645df9748b2d316c8bc35 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3b0fe726404186a347f7bffa9ce778 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b545e1e3baa040c99e9fa6668912972e : SUCCESS in 1m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7cc22c58ca564b9b826bed285b79adef : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4fff845c1caa459a9fecdabce1db7fe4 : SUCCESS in 8m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c95bb84b6164d4793169d7ab1d94c95 : SUCCESS in 5m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/88a3e22c4b014eda91a23fb02675c344 : SUCCESS in 1h 05m 13s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/b13d561ccc2f4c4eb6b89e76902612a8 : SUCCESS in 52m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68ecb76a8f3547bf98ea10484466c78a : SUCCESS in 57m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/70dc276ce22048fa886e24912bf303cf : SUCCESS in 34m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0b13edee7a2d4f249e070c6bb86e6368 : SUCCESS in 53m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9283037954f743fb97e85f5ed5c2f50e : SUCCESS in 19m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a6e6d49ad6e486f9332ad1d73e76bbb : SUCCESS in 13m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1dd264399ba048c682d5aee466f06614 : SUCCESS in 1h 44m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbc8148677ed47faa9eb86cce4f7ce7b : SUCCESS in 1h 10m 39s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/18aa5fdf9b6e43348ba2b146185f6d59 : SUCCESS in 2h 02m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8a36005117a74af480e0a4bc08ae4a9e : SUCCESS in 17m 14s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ae2435127e87410dbb3e96b72b4fdee5 : SUCCESS in 19m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1aaf762b5764c718aef0b5316c350c6 : SUCCESS in 21m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/46742d854b91464caeb933f2f92d498c : SUCCESS in 21m 21s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1da4fc384abb436da9eae38f6be518ea : SUCCESS in 34m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1c00faa322124bada1a3518afecd756f : SUCCESS in 34m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a429301c82cc4b529a8bceb21587bc14 : SUCCESS in 31m 05s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/785f951d88b243478aab2acba8cca1ae : SUCCESS in 34m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4904b0176aa246b1b8addbe7a53b97d7 : SUCCESS in 16m 20s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/dfd8d71ddbb740608fc4aee5268f7c96 : SUCCESS in 17m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0b3f1ed5b4c48d7a724465ed8ddc5e8 : SUCCESS in 14m 57s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a9f8afd55b584878884f618fcf24bdc2 : SUCCESS in 13m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/685f0af220ce46c09d95ee5737633549 : SUCCESS in 12m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a927ab1eb5e142ecb559e76601a00e71 : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8f3b1317111bb0609f1340b7441f17c13bcd2831","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-21 14:40:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"059cca73ce66c9c5d810afecfe74fd8ef77eb8a5","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-21 16:42:06.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"39f3e228603e208c5ff2c9f00ef55155ddb08cd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 16:42:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2f91491545b0f8ada270e5e17bf40256f492c2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 19:06:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1543a1f80e774961a10c08788e1c6505\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f531a5c06086431bb92e2b708f765845 : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81a37c989011441293a2fdf6e55b99b5 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b002d420f3ba494980bcc5884c1f417e : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3e77d1628704ac387eb8c60c53bf1c7 : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b458b2c804cb4b83bb12f4ac72ff758b : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0f1293dcb3714847a03739dd4c8f5db3 : SUCCESS in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b8cf707ae8d24fd4b0ddccccf0d5dc3a : SUCCESS in 9m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/13003c57c2e24ec5b88fa6e1a98bc363 : SUCCESS in 58m 37s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a76da185d01f4eb5bfdb0fb5ddb2ecee : SUCCESS in 1h 14m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc766fbd5eb94c4e87cfa7e129251951 : SUCCESS in 59m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ab96914d14494dd29ffb758662fc0f55 : SUCCESS in 1h 12m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9388540343a14a0eb6b8cc58de35652c : SUCCESS in 1h 14m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8d65ae35f1734c1eb067e0bebb37c4b2 : SUCCESS in 11m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1c91654b9cf4531b01a6c2b77bc6208 : SUCCESS in 13m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1a5e3428eb5c4ef2b1142238d4cc6456 : SUCCESS in 2h 14m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4a8627bf91640fbb965a57edd3c4fc0 : SUCCESS in 2h 10m 31s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/252c35677e124dc3a528a57109264872 : SUCCESS in 1h 58m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5c5bbc5fd15e4a1dab29be472b8ced25 : SUCCESS in 23m 54s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/eb60beb26f044e29b1557d5792611468 : SUCCESS in 22m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc4cb996927e4437a36b69b1ca8b007d : SUCCESS in 13m 37s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3e9c61fbcc8545d48ea231df695c96e4 : SUCCESS in 43m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93afb401f5de4273979a09ba6e57b4eb : SUCCESS in 38m 59s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8854d77623f34115922d6a34dd506fb1 : SUCCESS in 44m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/499b99ecc8924eb683c77358d05824a0 : SUCCESS in 17m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86c27c15b8d34dac924bdf13db87a0af : SUCCESS in 15m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0e468ab3b985448f8eb941eb413e0832 : SUCCESS in 17m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0eb0806a7264586ae040ba1093cd86a : SUCCESS in 15m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9e994553de551bfd69630f3c76e4ab7783e21891","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 19:06:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ad15a71584674266ce8fd1dc5bd5f3b06c0d08f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 19:07:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b58013367b0d4bb6aead60390d74aadb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f90b0ab86ad948999756df71098bb49c : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8afc0aa6aafd4d54a3fa27fdc719ce44 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3490905286953484f96f6ed07698fb8e9c6d2eb5","revisions":{"3490905286953484f96f6ed07698fb8e9c6d2eb5":{"kind":"REWORK","_number":1,"created":"2025-03-21 12:31:43.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/07/945207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/945207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/945207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/945207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/945207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/945207/1"}}},"commit":{"parents":[{"commit":"62b307c48f37a2d998a68888effd071fde881fd4","subject":"Bump MichaelRigart.interfaces to v1.15.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62b307c48f37a2d998a68888effd071fde881fd4"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-03-21 12:18:48.000000000","tz":0},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-03-21 12:31:42.000000000","tz":0},"subject":"Fix netplan packages do not getting fully removed","message":"Fix netplan packages do not getting fully removed\n\nRemaining netplan packages ``libnetplan1`` and ``netplan-generator``\ntemplates default network configs which prevents systemd-networkd\nto apply Kayobe generated network configs.\n\nCloses-Bug: #2103794\nChange-Id: If15bcfed7f212b9592049287514d8d601b70a41b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3490905286953484f96f6ed07698fb8e9c6d2eb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3490905286953484f96f6ed07698fb8e9c6d2eb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
