)]}'
{"id":"openstack%2Fkayobe~945209","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I777e3caa782a3a9b77699a668b27b4334040efda","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-03-24 11:11:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I777e3caa782a3a9b77699a668b27b4334040efda","subject":"Bump version of stackhpc.linux collection","status":"MERGED","created":"2025-03-21 13:24:26.000000000","updated":"2025-03-24 11:12:18.000000000","submitted":"2025-03-24 11:11:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945209","cherry_pick_of_change":944142,"cherry_pick_of_patch_set":1,"meta_rev_id":"5a1b82c0f184d61f09c12b4f1be43325b01af625","_number":945209,"virtual_id_number":945209,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-24 11:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-03-21 20:00:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-21 18:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-03-24 09:13:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-03-24 09:13:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-21 13:24:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-03-21 15:30:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-21 20:00:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-24 09:13:24.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"770e5bc6ddbe0ef27b8101e4d64f5ae8626bc0d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-21 13:24:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"66d90e7f2601ed9c0033d98983a924a16e3c9dc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 15:30:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bada470cf2484eca9f9480820c13f244\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70c258eb53804c7c969d0641866004cf : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d54d87b3b902414d9a0126d5b7733d25 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8847f0c4c4bd4d6081b655590a57e6f7 : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1be32c4a21164c4085d0acb7597fd77a : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86dc6655096d45e482aec9cf47e95670 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f58f2c0497d84684b991c7fa34e56f7a : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1825c1336bcd48fa9fe4f2cf1c7ba05c : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4fad8846c19047658b5e74408cc41339 : SUCCESS in 5m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/483523606fa0422499f360c0b9b55051 : SUCCESS in 8m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5ae45636bd724544994814aacad9351d : SUCCESS in 1h 01m 47s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/19eac132fc994031b956e350048614e2 : FAILURE in 4m 16s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b15792f38c674e2581da990152664ae1 : SUCCESS in 1h 12m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f1a1e76073a5457685819b4441039c08 : SUCCESS in 1h 01m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/87c0118f60014424b8df4ace0b5c7878 : SUCCESS in 14m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/aff33fe452344825b0a1718bd59e3a96 : FAILURE in 9m 21s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d18ecdf6ec047a7bf7970591c92f6bf : SUCCESS in 10m 33s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2ddbf2e4088c43f1ab0d169963ac4051 : SUCCESS in 2h 01m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db4fc2dff58948c5add43a5cb0d58dc6 : SUCCESS in 58m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/159cee05c5dc4c7c8a5718ae16828ea6 : SUCCESS in 18m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c727f95750434372adfef6740ff9ea08 : SUCCESS in 19m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b22fd9267daf4172aa87460463c71806 : SUCCESS in 30m 37s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5a7a73b7c44d4fdfa54c3fc027339378 : SUCCESS in 28m 53s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a435674c7af14daea652302eb8eb0f54 : SUCCESS in 22m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/09418a61c3a24b8288a64e1e26b4ba30 : SUCCESS in 12m 24s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/c0386e4150b0446aa4afcc409dd2955b : SUCCESS in 17m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8692024fd8544ff5829e579b624d2fc7 : SUCCESS in 14m 53s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/1786bb0dc15e41f4890ab46c86ea6d5c : SUCCESS in 14m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/87a671ce0e134ca08492a619f2f8cfc6 : SUCCESS in 23m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c951426ad4c1483db701b198c0f04fae : SUCCESS in 21m 34s","accounts_in_message":[],"_revision_number":1},{"id":"4bb1eb4ac194ce2c8922ff0b698c870fe50a6793","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-21 18:37:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5836d190189d7beccd559025dad3f2ba71da99ea","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-03-21 20:00:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"acd842d32a747c7add763e23c0e4a7ce7a883447","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-24 09:13:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a22a89d9e4f428ca130deb8a28196c6cdf459149","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 09:13:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7af164913d22a1bc3c2db499cfa89e43ee9a6e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 11:11:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42da36f560964660b521023f0a66afcd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c02bdd8404d473e8285a2c3792d899b : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b79b9d3d007a4cb9a724eb115f6803a6 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f49cdffc1b8d4a5ea0a62f0ac6bf18ab : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61baf82865df421f89d6b0c7a0e80edd : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/37f0837cb912486c8e08232ef98078d9 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c21308ad17443548e20bfdbc9771c8d : SUCCESS in 6m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4cd98b256eb34cf08ce8cf9f152e9a0f : SUCCESS in 5m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/40d979dd50ab4a73b071f79aa0a2b7bb : SUCCESS in 1h 06m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4488f4ed07d644549bd6a91e291538ce : SUCCESS in 55m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e24fd475e8a948d58c908db20ad96158 : SUCCESS in 1h 15m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f1522a2fe8d9489da3bbecf45d1a0dae : SUCCESS in 17m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ead3e65b15d84ce4842e26f6194fca4b : SUCCESS in 12m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/98f23c0522374a5c98e314e81e45300a : SUCCESS in 1h 52m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/123630b0b4d04c03b5aed594984f083f : SUCCESS in 1h 40m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/eb24ae0ce9234f89bc663000a45adb3c : SUCCESS in 24m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5405a089b1947d4bbeaf8009bdb7d27 : SUCCESS in 19m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2e04b45bec714e17b831a4b30fb6b45c : SUCCESS in 33m 38s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db11946336d64fb0a77e73e3b9647a0e : SUCCESS in 32m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/be3c9d1e8cc747a897617557c46fee2d : SUCCESS in 16m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5aa9025108d0486bbd4db5609708d672 : SUCCESS in 15m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/22cce6ab672040c886e816d5a47c4d33 : SUCCESS in 16m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89e0af7681634589ba5d4177b9435172 : SUCCESS in 14m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2871c0d26d4a85ebe5c64ac95ecdc47ceb1dfb95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 11:11:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a1b82c0f184d61f09c12b4f1be43325b01af625","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 11:12:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdb4a0dca49c4c4eb8bd54f5985a392b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/174f5e35bac74c74800d1f4b80573f27 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/436d7903216d44f7a7f9151c2f0b42cd : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7154603abe0b8957e9803706f23c0d13f4b5b08f","revisions":{"7154603abe0b8957e9803706f23c0d13f4b5b08f":{"kind":"REWORK","_number":1,"created":"2025-03-21 13:24:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/945209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/945209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/945209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/945209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/945209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/945209/1"}}},"commit":{"parents":[{"commit":"75e6ce36502886d0e1526371288aab177d3556a8","subject":"Merge \"Remove graphlib-backport requirement\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75e6ce36502886d0e1526371288aab177d3556a8"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-03-12 17:36:59.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-03-21 13:24:26.000000000","tz":0},"subject":"Bump version of stackhpc.linux collection","message":"Bump version of stackhpc.linux collection\n\nFixes a race condition in the vgpu startup logic by adding retries.\n\nCloses-Bug: #2102153\nChange-Id: I777e3caa782a3a9b77699a668b27b4334040efda\n(cherry picked from commit 05896059429c094735bd0bc96536ab968db62621)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7154603abe0b8957e9803706f23c0d13f4b5b08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7154603abe0b8957e9803706f23c0d13f4b5b08f"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
