)]}'
{"id":"openstack%2Fkayobe~945283","triplet_id":"openstack%2Fkayobe~master~Ia5899e06199f38be8c81b3ba9474e52d2ccd3690","project":"openstack/kayobe","branch":"master","topic":"bump-ansible-chrony","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-03-24 12:37:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia5899e06199f38be8c81b3ba9474e52d2ccd3690","subject":"Bump the mrlesmithjr.chrony role to v0.1.6","status":"MERGED","created":"2025-03-23 20:50:45.000000000","updated":"2025-03-24 12:38:12.000000000","submitted":"2025-03-24 12:37:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945283-bump-ansible-chrony","meta_rev_id":"3276e636fadcfb80a851d7a983db7bcc156d885a","_number":945283,"virtual_id_number":945283,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-24 12:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-03-24 10:50:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-24 09:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-03-24 10:50:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-23 22:30:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-24 09:14:48.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-03-24 10:50:01.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"36e04552d203f681235de40cc6cd0e1914cc448c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-23 20:50:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0388062c206e4e01bf942b8e94b9f23bed5d046c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-23 22:30:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4be6989e3994317b124d3ccbed2edb6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1680699698ed47a3abb5eff2eb608e71 : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13d1e314b3854e72b9c8bc6c5eaf7696 : SUCCESS in 2m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aeeeed6838cc408db59ac30c7444decb : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cfd42c8e6ef483682e5b3eae58f2239 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/995ce1aa83ea4006abd2f594bd8b80eb : SUCCESS in 1m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d33a4d7620314a28ae5f4be2772c035d : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd2047036994403b8df552df0584e0da : SUCCESS in 5m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1543a71ca6a5478483c0d736e1c3e3b5 : SUCCESS in 8m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8040a1aa82754eaaa6635321c4620ae5 : SUCCESS in 29m 28s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/8763024030e84c41ba935c0d132b8af6 : SUCCESS in 49m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2ae62a8af3142a998434369054ddfd5 : SUCCESS in 27m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5c95819c21d249d9b7c4146d59bfbc41 : SUCCESS in 53m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/40e4950ad9e848b99bf357960c5a514e : SUCCESS in 51m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ad5ab6748a984cd592bd7e627f711a29 : SUCCESS in 16m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e93383392fba45948d6e4950e6a014ba : SUCCESS in 7m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/03a0b46001674c07b251b0f3e3894f2c : SUCCESS in 1h 35m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/830bde7e3fd147da91e4e3c1fd4e5275 : SUCCESS in 1h 39m 26s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ee6abe2684eb4924a9d34b5eefc7b12d : SUCCESS in 1h 37m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/df4720eaba6d4ef0be647d2e499c3cbd : SUCCESS in 11m 07s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/adcce86f097b481c8add3f41a8e222c9 : SUCCESS in 16m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83fd3c31541c4f118035bd93816a4eb3 : SUCCESS in 8m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/82ae610eb1574810b3c103a143e4524c : SUCCESS in 13m 41s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a834270457274de1b0102184bb2fd242 : SUCCESS in 33m 15s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2e761c3195cc42ecb81717a226ebd5b7 : SUCCESS in 30m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8480eb3f99fc4dbd88c5b13fe19a579c : SUCCESS in 32m 14s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7790f0e60a0c4b6aaacdd3ad6177981c : SUCCESS in 32m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b9672f16012744b0b2052f17364d3406 : SUCCESS in 12m 55s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b44e85b4293f4cac87395fef2a755612 : SUCCESS in 10m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59c1ff1e2f8447ebb4b9e051efd25272 : SUCCESS in 5m 59s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/eac73825b10948b08035e53cfb2faa6b : SUCCESS in 11m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5e428a29d6974d2b9751406477dac4b6 : SUCCESS in 12m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f51427dfb2e84328bff5cca2120ad78c : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"f2a5cc5889ff1d100fba3ef7028de62903f55eb1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-24 09:14:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"81f8240b08e41e1cb69ad89fcdacea2a7193da3a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-03-24 10:50:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3352b8486e551cd71d693deebdb09e3db617ea38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 10:50:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"296421dab31a87b2e89232516cc65fde4d08c21b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 12:37:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/975620810e124dbdb957dea872f5bf89\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84544389dc774fca9a576a37254e40f3 : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34f4f3317ad04a61be9a42cad24fcc93 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab47b3d0f784aa3a38368a29dc01c02 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c63f519c12794001b6bd36dad93bd102 : SUCCESS in 2m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/07a169b9216c4080a99a928531e06043 : SUCCESS in 3m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c8abdbc1a48747048c5aa6cbe78d664d : SUCCESS in 12m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f61c19921ed648a6986bb21500449ff8 : SUCCESS in 5m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4abbc2752e624267ace0c5c668371755 : SUCCESS in 34m 36s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/ac6723c0331d446fb745ee5a8f571ed7 : SUCCESS in 35m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28248c8b4fc242b483c4a488ee15b693 : SUCCESS in 32m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3a4c9eee5d994e43881bc85218f73543 : SUCCESS in 34m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5415f17e4ce4455aa95767680d6bbd64 : SUCCESS in 36m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/49acda485efa4a7fae0d0dbeb213eb92 : SUCCESS in 16m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/522772daef5640d68319e2b550283423 : SUCCESS in 13m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/308feefb253541c5944d7d40cf179712 : SUCCESS in 1h 40m 20s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbf8da641d44440199bb967c3609dfbd : SUCCESS in 1h 38m 22s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0b28d569baf54b7aab939f988c5d09db : SUCCESS in 1h 44m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5240f8061d8c42fc931e49bba45e2f5d : SUCCESS in 14m 29s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/f2c5febe277f47bc9fc27c9f4d5e0cee : SUCCESS in 15m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41cbd3ea383549dc9ec7f60441dacca9 : SUCCESS in 13m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/14284df7f7da4758b019f564f339d919 : SUCCESS in 34m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/695cbc31bad54891bbb4693868296fbe : SUCCESS in 29m 42s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ce77e95b6cf7448b8978bc87208f3683 : SUCCESS in 29m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f997b949610e40a2924ecacf7698c0e1 : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2c65276e0324763971754158e316136 : SUCCESS in 10m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/051c7097e44c47bfaddf824bdc17dd87 : SUCCESS in 11m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c7e1ce9973647f88ab06bc79b155a5f : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"51a1674567ec08b1ffa8cb14ddc63af37929bb94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 12:37:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3276e636fadcfb80a851d7a983db7bcc156d885a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 12:38:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600c8bc16e8d49f289ee6e1083f844c0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be7222e973614eecba41bc69c944502e : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1546361319144075942df0e7cdebcee9 : SUCCESS in 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f615af72bb51c7bc5424961b6b1182e938daeb3","revisions":{"9f615af72bb51c7bc5424961b6b1182e938daeb3":{"kind":"REWORK","_number":1,"created":"2025-03-23 20:50:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/945283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/945283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/945283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/945283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/945283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/945283/1"}}},"commit":{"parents":[{"commit":"097bddb3b7329507fe1342b257fb92ac0532ee03","subject":"Merge \"Fix netplan packages do not getting fully removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/097bddb3b7329507fe1342b257fb92ac0532ee03"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-03-23 20:50:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-03-23 20:50:15.000000000","tz":60},"subject":"Bump the mrlesmithjr.chrony role to v0.1.6","message":"Bump the mrlesmithjr.chrony role to v0.1.6\n\nChange-Id: Ia5899e06199f38be8c81b3ba9474e52d2ccd3690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f615af72bb51c7bc5424961b6b1182e938daeb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f615af72bb51c7bc5424961b6b1182e938daeb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
