)]}'
{"id":"openstack%2Fkayobe~945341","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I577bd145173d6a474f20aac9fe25a46defc9e47d","project":"openstack/kayobe","branch":"stable/2024.2","topic":"centos-iproute-stable/2024.2","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-03-27 10:58:19.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-03-27 12:48:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I577bd145173d6a474f20aac9fe25a46defc9e47d","subject":"Bump MichaelRigart.interfaces to v1.15.4","status":"MERGED","created":"2025-03-24 15:40:15.000000000","updated":"2025-03-27 12:49:54.000000000","submitted":"2025-03-27 12:48:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945341-centos-iproute-stable/2024.2","cherry_pick_of_change":943350,"cherry_pick_of_patch_set":2,"meta_rev_id":"a35a7af29f9477a313540c22528ba7e023505de7","_number":945341,"virtual_id_number":945341,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-27 12:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-03-27 10:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-03-25 16:33:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-24 16:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-03-27 10:58:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-24 16:14:52.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-03-24 16:41:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-25 16:33:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-27 10:57:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"d6e3119e27fe9fac0150be0796ab1edd6ce48bf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-24 15:40:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"77e20021415f3d6b3e007d018485f385f08e3f6b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-24 16:14:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1607b95bc9b0b9f2cc4f54531ed0f7a54c08963c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 16:41:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5274a1b2740c4594a4873bdd3b215457\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c3c5db55afd41f88a77d9c9dceabe56 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/407c1e42486d4fbfa8273a32b7ddd181 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e74a403fe824ef6b35d1572a392d6f9 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2914f018a2c4be6ad5197d20396d875 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2929e33ecb4651b681815644812597 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/235c3ebdc00745c2935cfd6d154a7367 : SUCCESS in 2m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2a44b2a827854872a444bb1642970ff5 : SUCCESS in 2m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c17a5a83884e4373b3a9b85664f2eeb4 : FAILURE in 6m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d29ffdf104a4436c899502aaf6dac721 : SUCCESS in 8m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d3e950c968154f2a8c79cd3f58162e7d : FAILURE in 10m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4481afed95924d83a30b6eb54180f7ec : FAILURE in 10m 32s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ed3dd6b465343a3a23559455b74b465 : FAILURE in 10m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9c07d6a998db4b918fb690fc37d15e80 : FAILURE in 8m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8985e49038d34d36b5b898f68e8293bf : FAILURE in 6m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2adab0f0ff9d4f7eb9ac328fe7100a3a : FAILURE in 11m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e9f02c0636c48fab8c9385ca0ae91a1 : FAILURE in 5m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/292ea63ce3034aa5aea154806815e9b3 : FAILURE in 49m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78540c9a3cd9483eb6b70d04db0fce69 : FAILURE in 59m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/eabe98423c1a4168a82eb2ac45c9cdcf : FAILURE in 11m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acaeebda3d964001b4761806f0df60aa : FAILURE in 11m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0a3a607c30ca4ecc900572012d517a3d : FAILURE in 11m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4b8a89de028f487b9bc9eea136c6b4e2 : FAILURE in 39m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07c651ed691f416bbd21f3b9edeab66f : FAILURE in 32m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a26cafd59618455db6fa9c100b4eca65 : FAILURE in 11m 32s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/642f907181a2424ba323196cc94f0cf2 : FAILURE in 11m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30e38217488241dc912687d9761b241f : FAILURE in 9m 14s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/7344a04ffc8d4b3da464b18e7a885f2f : FAILURE in 9m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/23acb616a1aa494e9d85c86cc1f8e965 : FAILURE in 9m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a008119681d74870a39245d0e623dfa0 : FAILURE in 9m 22s","accounts_in_message":[],"_revision_number":1},{"id":"12e6068f664bfcb5cac8360a69ec5ef5ebc59086","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-25 08:06:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10b92ef0c82ba5459a4e26977d33d96481c03cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 09:20:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de9df00081fc4bb49776a9c1b8eca03a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ccc02a9c4da404797db4871768ab056 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cec0e481857441dd8b45f7abc2ba85a7 : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0b7993770f747e9aea33bdaedcf650d : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b07e5e85ac44ce99eb297ade1acf6df : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569d2554cdff430185e1ff0a6f21dbbe : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5f3b3cd2b4f4b15a3a50c546e6ce402 : SUCCESS in 2m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/49d34339989142ceb6f192ca31872a1f : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7f66bd5e41524a6d84b72712554fca24 : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2404314fff6649139124895c6258fef5 : SUCCESS in 9m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a46005b95a48493594386640370a8e57 : SUCCESS in 58m 22s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1389076bc7c84dac88c41276dad253d3 : SUCCESS in 1h 10m 06s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da03a166265c41e99a2dc9ecbb068ad6 : SUCCESS in 59m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cc16132e4a4843e4a34e2dc29fd12a17 : SUCCESS in 38m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f6bc89eebbe248f9b2203e0817a66930 : SUCCESS in 16m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6bc6cc031f3c421492e44540ed9cf0be : SUCCESS in 12m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8cd94a7653d4f2ea5b4bffca51ae74a : SUCCESS in 12m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/80c58491035d4641b4e1396c1665da9c : SUCCESS in 1h 05m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a0bd30df6244c57801267f46219b66e : SUCCESS in 1h 06m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8abcdeb856444fe189969710d43d7677 : SUCCESS in 16m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efd5289fa225402988b29107ac4afddc : SUCCESS in 18m 49s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d69a93e151cb4ad6a81bb9760055de99 : SUCCESS in 25m 25s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d143d173c76d49bc95bd8d33dcff2e92 : SUCCESS in 22m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edb393d4040c4a0bad65b9205648c4b6 : SUCCESS in 30m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a60aeece89f64ce7ae82dae1b60ae8ad : SUCCESS in 16m 13s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/095c3dec70174cb99bd5565dbb412789 : SUCCESS in 15m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d06e93e1a0ac4ffa8ce39a0a0c6345dd : SUCCESS in 14m 38s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/70cf16018f3c41718a6e136deb1ef3f6 : SUCCESS in 14m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bb44f838d5444b4095bb61aa9562e561 : SUCCESS in 15m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/986038f1fa7b4481bf086a196da52983 : SUCCESS in 18m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e7a321edf888eccccc52fc5a4123fabfa183bc1c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-03-25 16:33:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cc92ab2e3d017ade6b647369c73b6b7aeaebff00","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-03-27 10:58:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"830079330420f5de09e27be29055f193e55c7958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 10:58:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5076b1640f5bddfe9760ca5a793dd804cafd15a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 12:48:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8389ce56d6414c288f3fb465cec8e313\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28cc77a0d4cc44358ea5b78bf7127b6a : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc471e789b5642ad9b55cdb0d754006b : SUCCESS in 2m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5d63e9b750468cb09b2a62853f0dd5 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df978fc892c149a4b86625eb33bff745 : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dd1172b67ab8473dbd26f139618cbdc3 : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb12fda28a3a462791e63e394fd9164a : SUCCESS in 7m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/982f7fe5a3c644fd97fa9b3c4104f661 : SUCCESS in 9m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cd7133f7cc28460a9617c1ded5e773ac : SUCCESS in 54m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5708ba2e3d94380b74a9027e707ccf8 : SUCCESS in 55m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ccdfa22f1e264ff8a2dc6787f9dfa9be : SUCCESS in 52m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/33a05b4c173046ac83ec9abb1b4a4644 : SUCCESS in 19m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9594569b94e44b4ba1eeb4d8e7d62587 : SUCCESS in 10m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fc0d04092b0d40b4a104301b4607882f : SUCCESS in 1h 49m 33s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/392d7f2622a8474f892802a1c142b8a6 : SUCCESS in 1h 42m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/55c3a87037f047cd9f1ac6aa01c1557f : SUCCESS in 19m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb2fde38390e466aa1ad212aa4ae5104 : SUCCESS in 21m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/feb1b1dee73848be84a32d8d11fcb42a : SUCCESS in 33m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/359d124f919c43518d38da53cba224f4 : SUCCESS in 31m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1937d4fe47e440aea4e2c03cb2d79121 : SUCCESS in 15m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a05a8d73496c495f92b0ed6063ed1305 : SUCCESS in 15m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bf73398fa9d844ceb95c970313898ced : SUCCESS in 16m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db00c1d4dfbe417097eca3f3b11df169 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":1},{"id":"548a3419ac316cc3c65823a8ef96999c17d375a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 12:48:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a35a7af29f9477a313540c22528ba7e023505de7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 12:49:54.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/832ec7f1ef724972bfa16ca2f82e87a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a966ce6de8004ffb97962769fab46bc5 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd6f142f151c4b4c98c8e690736728ba : FAILURE in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3cef020b3743655e1cba44bba1ba23485d6dbe1f","revisions":{"3cef020b3743655e1cba44bba1ba23485d6dbe1f":{"kind":"REWORK","_number":1,"created":"2025-03-24 15:40:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/945341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/945341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/945341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/945341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/945341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/945341/1"}}},"commit":{"parents":[{"commit":"0f121c754bf2db4c62f444589b1f522f0150bb8b","subject":"Merge \"Bump version of stackhpc.linux collection\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f121c754bf2db4c62f444589b1f522f0150bb8b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-03-18 13:00:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-03-24 15:40:15.000000000","tz":0},"subject":"Bump MichaelRigart.interfaces to v1.15.4","message":"Bump MichaelRigart.interfaces to v1.15.4\n\nThis fixes compatibility with CentOS Stream 9 due to changes in the\niproute package.\n\nChange-Id: I577bd145173d6a474f20aac9fe25a46defc9e47d\n(cherry picked from commit 62b307c48f37a2d998a68888effd071fde881fd4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3cef020b3743655e1cba44bba1ba23485d6dbe1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3cef020b3743655e1cba44bba1ba23485d6dbe1f"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
