)]}'
{"id":"openstack%2Fkayobe~945472","triplet_id":"openstack%2Fkayobe~master~I1c4aa7f22a9cfa6d2c1abad8efaca880572a0bdf","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2025-04-03 11:38:58.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-04-03 11:38:58.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-03-28 08:43:37.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I1c4aa7f22a9cfa6d2c1abad8efaca880572a0bdf","subject":"Bump up Ansible supported versions to 11.x/12.x","status":"MERGED","created":"2025-03-25 15:28:44.000000000","updated":"2025-04-03 11:40:02.000000000","submitted":"2025-04-03 11:38:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945472","meta_rev_id":"ab06feaae7a9452162335a15a61bc56f847fe916","_number":945472,"virtual_id_number":945472,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-03 11:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2025-04-03 09:28:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-03 09:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-03 09:55:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2025-04-03 09:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-25 15:39:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-03-25 16:45:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-25 17:34:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-01 12:46:59.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-04-03 09:28:04.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-04-03 09:55:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"78e7ecb55e3d4c61a708f16606fdb430dd562d0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-25 15:28:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9e27f6ef8bdce28d0050ae141b392f0b4d5dad2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-25 15:39:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4e13129ec933b742ff56e2f4657cfb0e852cde3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 17:34:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dbc26f716394d5a9d57f81bb5be0f8c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df21f79241ff42c6adc1ef6a78ac3af6 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d566cc549524c7592b59de663ee71f4 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fbf05f232288432f93566f980f104dad : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84c54a9249943f9b176e6e9cb61ef01 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e81ece537e9644d9b44e607a6504e02f : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4db1bdb2f9054423aaa03431074f8de5 : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bdaa2e526afc494eaca132ae73be7db9 : SUCCESS in 12m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b759f699a9cf476488379497e89455db : SUCCESS in 11m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/52e7c5670d8b4ed680e425149b7a9450 : FAILURE in 40m 46s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/2e3aff71760940fe8e465e7e8e3aab4b : FAILURE in 41m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/507632e8bfb542a585c0e8d4a06ebc28 : FAILURE in 35m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/df80cf83472d46d487ee29633f976b94 : FAILURE in 45m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0139810ea8e840fb939f77f2cc2d6a67 : FAILURE in 47m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/19fe43bc070741768602efe4aa140b8d : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c043f8bd2d74db8a245b58a293857ee : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/96f364535cc34de799acd3a29ef7e9d8 : SUCCESS in 1h 52m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcdd104860e944dfbaef71cc3e735c23 : SUCCESS in 1h 48m 27s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/61187eeb2aad4a2a919227d4701b9619 : SUCCESS in 1h 54m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e78f720e33f34c43bc0ba81bf3a99761 : SUCCESS in 16m 24s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8f790276e1a841bda053740b84c1a501 : SUCCESS in 10m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/217c23c444024397894652a624c373e0 : SUCCESS in 14m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a0a2c0e31008404a8ea3c9e0ad1d36c1 : SUCCESS in 13m 55s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/39dc6695052244efa92d2b5ccc5e3f17 : SUCCESS in 38m 07s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4d24f16431034759a309af2afa84943a : SUCCESS in 29m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d839f7412cfe4b02891f5f0f1f63d9e4 : SUCCESS in 31m 23s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/363b8c971d4c4286be971a5e8bb612a3 : SUCCESS in 35m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2af55337d42e4cadb74c91fb96debfaa : SUCCESS in 12m 31s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b135c3e34c074fd3a05d5b8a1e985093 : SUCCESS in 12m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fc9b6a798b5423cbb001ec1a5e971ec : SUCCESS in 11m 16s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/cf81b39c6b454149b0171620bb32cda2 : SUCCESS in 12m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e7773720c7ab4097bcd8c66d736d359f : SUCCESS in 11m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/424bea989ec04eb48d152a5f4b17e464 : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"01e5cc13fa1645cdeb8f185d5d7b3f2492a5e45b","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-27 13:11:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"697c88296dc1dede41a2745bf0f490c2ff54062d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-28 08:43:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"795b0e601e7958898f2eb9043cb082d16d97723b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 10:17:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59a31fa832234840b94a362e59d6ac1b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0aeb6486afdf402990dbd3b584881291 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b8a6cf72fb1466cac666969f8a587f8 : SUCCESS in 5m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a6854b866434e53bc6bc205b90194f7 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7df52476a234dda906133a09aa7cfb8 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86767bf9f2c948ed992ff478b553c1f7 : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ce8d60a9a98342ae89f8c360a05817df : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/93d71d9b0e584567af35536a4029a16f : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ce6136aeebb4e5789280988f30f0697 : SUCCESS in 5m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7a5b227792da4cccad3a9e96eaafbbf7 : SUCCESS in 1h 04m 34s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/ba1c53e19b7f4d389c803c59259ccf48 : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3790f88e4ee4232b6abcf560b879c18 : SUCCESS in 58m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/07a5832260aa432eaad4680f07e51d0e : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/11e78fb2d3b24f6dae40202257465d0e : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6e0c96cdf8424b55952bb84f9994f3fa : SUCCESS in 20m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ce3b8af8d2b442cae576524ab145920 : SUCCESS in 14m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e955ec4bac954474a32fc901e82f9993 : FAILURE in 51m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9cb3826b85a4ccfa57e8c850c57b83e : FAILURE in 1h 31m 40s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c6371359fcbf4235ad0f6b61219974fa : FAILURE in 1h 17m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bcc932b969964478b27d1bf0fc5c8ab4 : SUCCESS in 15m 09s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/d0d990e9fe9640bcbcaf3c4696b096bc : SUCCESS in 17m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1069b2704514e7fbc16499771e9b85e : SUCCESS in 11m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7b911c12331a4d93941a0fd16ee1c820 : SUCCESS in 7m 05s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/62e5e97d872846f689f191ff3f254e5b : SUCCESS in 36m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9c61127b1ab646a9908ca773e5c6b26d : SUCCESS in 32m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c3143a948f64253a3bec8c89798095a : SUCCESS in 28m 57s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a6f7d92afd8c4132b4cead09ca8c263b : SUCCESS in 28m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b571fbd510e84d58961ac8f6ae5d0f8d : SUCCESS in 12m 49s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8c76d664f57442698813f3ade52f1240 : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecdc170462114b6c85aad2f1e23614cb : SUCCESS in 6m 00s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/ea84c7817bee495db4d90293126b2528 : SUCCESS in 10m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9b654259c7544768a5af74d1c799e2fd : SUCCESS in 12m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83feb22f7c324ee4b6adb3e50284470e : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d554f0d04cd0ea61aca893cd1f7cfe12766d1a1f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-28 21:08:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c0d4a69104ecf8515e03ab242a2413f7e880346","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 23:02:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f48c0b326ec34d3485e5292076988687\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fb9dc8063164d0997098ec53b3f72c3 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35b8444e95340e1918d553e9df9849f : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab8b9de693994a4c92d18b2aa1dcf928 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8680fb62ed5c455da6b6c62cdfb5d59e : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88d039f1354a420c927e50f902dd31dc : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b8b842669c84b2f99781f70cd036991 : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/de2d4ce30bae4518a3db2b2fe0d9676f : SUCCESS in 13m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6ed28e180dec4abe90bbbe385dd9c073 : SUCCESS in 12m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/96d6038dd69e426fb3e04615f8a2d0a2 : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/4f373ccd44f44264a2572a5591a14416 : SUCCESS in 1h 04m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9d0bee03c6643c2a959c44b371f9096 : SUCCESS in 44m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ffa24abe15b4ce2b4a2aef5028d1631 : SUCCESS in 55m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/16a278e1418549948f33d2e26dcfea71 : SUCCESS in 54m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4175f1707ed746d8b60fc870a4d0bc86 : SUCCESS in 16m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31b6690068f54655a8b389faed5549ab : SUCCESS in 8m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5913b104ed5c4b07b64c8082b9f2e865 : SUCCESS in 1h 41m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/198a4f4028a546eaab5c1c7cb2ff43b8 : SUCCESS in 1h 36m 19s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/49366754d31845d6928d9f67128c568b : SUCCESS in 1h 49m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4088db15cba64cf581602655b38eae84 : SUCCESS in 19m 15s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/d4e0eed4070d445793b58ca0b92d855a : SUCCESS in 8m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ffe4fb98c8b4e47a05c444a3f822fd8 : SUCCESS in 13m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2e53532e7ba34aa8a15a9ede5d71c9c0 : SUCCESS in 19m 12s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f3dca2722df84b2cab50aa43bae2f6f5 : SUCCESS in 37m 23s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6c54ae8d923a4866a03102348466bbea : SUCCESS in 29m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b821a4972b9403391461f56fdc31f4f : SUCCESS in 34m 19s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/569963bd0e5f49dbb288fd1b9dabc131 : SUCCESS in 22m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/46c09535d46a4315b9a5f6808f0016da : SUCCESS in 11m 35s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/ec7b419973cc41338f3dcc4087fabb61 : SUCCESS in 11m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e0698b55d5c4c158ae56aa8b087ead6 : SUCCESS in 10m 27s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d008925b97c942f5819294f91bdb70a4 : SUCCESS in 14m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/45dc4fed1ee549adbd135ca09fc89c63 : SUCCESS in 10m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/491738f87f3943e38ce4896c266173c7 : SUCCESS in 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9ccae907c911c01df360e74ad35b4d255cdf52a6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-01 12:46:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b13af479d1c0b1c743435740eb11aba1a7af3e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 14:46:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a324c0ab82f417aa2e0488bb571b85e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/372b63940b7947f390dea8f6d297639a : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4132dffb6749aba2c0e88e058a20b7 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4ef740e3aab4bc0b79513363c649dd3 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1165dfd372a4b629e24d15e8de4cab0 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fb9670b1a9d4b8d81f6abf7a4ae895e : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/55b571d25e0b4627b51882d1c0d6cc4d : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b2ed0007e557411b8b953d933edd05f4 : SUCCESS in 10m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/34af2ae3e15d4b159432a38ba5e89237 : SUCCESS in 9m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4398304967be4220abf6b9f84f63cae8 : SUCCESS in 30m 35s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/c21e8921c0ba41f2b2907d1d89cc57b8 : SUCCESS in 31m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/654ebaf701134595964d09adddf67b4c : SUCCESS in 50m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f5960118c2884e43903150c984134540 : SUCCESS in 53m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/895de7e17c59417ab8d8a64473aa37e5 : SUCCESS in 52m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2c04f000fce24488b17fe29cd1e6ef58 : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc5d16a2661942a0960e1a4822513f2f : SUCCESS in 10m 57s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/03a0411bd03e4cc6b8f988f3685f2db8 : SUCCESS in 1h 29m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc2f9f1bbf1f4f4a812c6644dc97fc32 : SUCCESS in 1h 33m 01s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bf0bd3469f09439d8bad668a501e8cfb : SUCCESS in 1h 47m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c80d7f3ba8534929a4fe7e80a2623ef7 : SUCCESS in 15m 42s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fc9d9bd4d5cb49e0b0ace2d2a81b07dd : SUCCESS in 17m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/036cc39b4c5049f88290ad12a1b08391 : SUCCESS in 14m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/48c884e728824b6095e719435cdad552 : SUCCESS in 13m 38s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/79cdc704e8bf449b95d48510df53eea8 : SUCCESS in 39m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/58f5a3f621b942ab891cdef44afbcbb7 : SUCCESS in 29m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afc69236af2b4d8c9b27f86525e7a03d : SUCCESS in 28m 16s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7b08c1d2789f4180a69f88bd8658d072 : SUCCESS in 32m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f8ef5f624522484ba4adbf0279f756dc : SUCCESS in 13m 24s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/bbb7f6b4bf69460baed132912d624678 : SUCCESS in 11m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a818cef2f5e4d5ba90c461757edcf63 : SUCCESS in 13m 26s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/4801242fd67d42199a5f4d87d4eee969 : SUCCESS in 10m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4f17e96efd304436b2b5ea08e6c42892 : SUCCESS in 12m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/840084aa6b154b7f8da8f8d84e10c9ba : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":1},{"id":"996f2ad97b9193e7317f7be50145c2e61f2fa925","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-03 09:28:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ec1268c709c1917701b26fe0fc3b9141cc04b7dd","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-03 09:28:26.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"5f80e2bf49752cfdd3552a16d93c240202144411","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-04-03 09:55:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4fa93f2edaffdbcd69d1209beab52b63da42875b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 09:56:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d8bd3815bfa5f518123d52438800cd836aa9afdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 11:38:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d14dc5b18de84f2ca9626819b9d30dbc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e27f014bf284bc686564b2716d9f150 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a32e537d9f04908af340e16e658fad0 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a246a20be46947f194115566ceede4ae : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/828db55c294d4fd196f14fac9b56a9fb : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9c487bb7ab14dfb99af8042260d5b64 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c4cb838201743f9b79e6278f86ac96c : SUCCESS in 3m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b971c1fe75ab460b99b33a7f4c0f6639 : SUCCESS in 9m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1f6fe45029a4170a127ea8a717fccb4 : SUCCESS in 8m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f5cd9ebb0d3f4ad8adc640b2f8201746 : SUCCESS in 49m 11s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/dcf58fdb50184fb4b2e55c54b295f2ff : SUCCESS in 52m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc81442b6a694ddfb07132f935a98c9a : SUCCESS in 1h 01m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/885da8a8d5c74b6d8d005f3c1e83af99 : SUCCESS in 50m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c1534007a5504e219fe087a62c9ab1ad : SUCCESS in 33m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ba57c384448a44b8b8191df2b4dc661d : SUCCESS in 10m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/838962c7e4584da7b7ef41328cb1900a : SUCCESS in 11m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8832995580fb413a83cc628cab01570a : SUCCESS in 1h 04m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39937f8c4ace455fb6a41f46ae6f60ed : SUCCESS in 1h 42m 27s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f759c2f2b5fe4dfd8135992198616bb7 : SUCCESS in 1h 02m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4f483ba0913a4defaa77dc140a6258be : SUCCESS in 13m 03s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/49f7ef6773b84ae29cd3f3598de85ecd : SUCCESS in 8m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57491af78c27450b891f75ef5a3bcdfd : SUCCESS in 9m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/36b96be63d584baf8fa2451fff8b93cd : SUCCESS in 21m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/034a8c35102f4f1a96768936721d1650 : SUCCESS in 19m 23s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/513729dcbb8b48f88e18f4ab2a561850 : SUCCESS in 32m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/573b0f22e4e54cab8107e8d88ab0a993 : SUCCESS in 7m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07a183aa4d664189a0632175d2a286c1 : SUCCESS in 13m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a33873a12a9e4865a779378cd0d614a3 : SUCCESS in 7m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df00d73122e54d3db7e1c4f9244a01e3 : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ecca2a10c3471f96bd7b69d75778810c963b6f24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 11:38:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab06feaae7a9452162335a15a61bc56f847fe916","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 11:40:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/315eae44431b41d8a5e0e1b71a524e59\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdc06de0d75a491eaafed227fea96505 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97b73a26da2b4169afad04e58c5ea833 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d1061d036f43c615171ce11e0e509e6269cb53c","revisions":{"7d1061d036f43c615171ce11e0e509e6269cb53c":{"kind":"REWORK","_number":1,"created":"2025-03-25 15:28:44.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/72/945472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/945472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/945472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/945472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/945472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/945472/1"}}},"commit":{"parents":[{"commit":"6098c8fe0141a6fbf1ff56dd4086cadeba1c765e","subject":"Merge \"Bump the mrlesmithjr.chrony role to v0.1.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6098c8fe0141a6fbf1ff56dd4086cadeba1c765e"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-25 15:27:28.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-25 15:28:42.000000000","tz":60},"subject":"Bump up Ansible supported versions to 11.x/12.x","message":"Bump up Ansible supported versions to 11.x/12.x\n\nThis change bumps up the maximum supported version of Ansible to 12.x\n(ansible-core 2.18.x) and minimum to 11.x (ansible-core 2.17.x).\n\nChange-Id: I1c4aa7f22a9cfa6d2c1abad8efaca880572a0bdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d1061d036f43c615171ce11e0e509e6269cb53c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d1061d036f43c615171ce11e0e509e6269cb53c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
