)]}'
{"id":"openstack%2Fkayobe~945705","triplet_id":"openstack%2Fkayobe~master~I83bfe555bf4cd7038fab8a95dd5fc13f07ade8e8","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-03-28 08:42:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I83bfe555bf4cd7038fab8a95dd5fc13f07ade8e8","subject":"DNM: Test kolla-ansible ironic-inspector fix","status":"ABANDONED","created":"2025-03-27 12:48:15.000000000","updated":"2025-03-28 08:42:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"018c93f5befb47ad69f93e6ea04a9073806b2dbd","_number":945705,"virtual_id_number":945705,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-03-27 15:29:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-27 15:29:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69be828676fc264a2271dbf8ba4fcc8468c2d324","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-03-27 12:48:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b5767172846bd97930a832f0dfc6d77e470164c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 15:29:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95d80b77ee824f8dae3729389cd3901b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a624b50da0040ee8e632640ca138fab : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d695440fa5f64e68a8d9729a94fccd43 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2fd55fcaa3c48cf816a8b05e8213ac0 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8e2bbc88f314b49adbd896b51a8d996 : FAILURE in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49332e1c71034ac29a938926ed7a970c : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d21ff4fb527a4518b34edf092dcb7242 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d98c7c8ac144db5b19cfd57b9626803 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c25e16d0f8484a748a030c0d9583fd4a : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dc46dfe8df374868886a7c102e1ad272 : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/277b3785d1c34fbcae7906f414a30647 : SUCCESS in 10m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bd7e638037c641828f61cb30f8688309 : SUCCESS in 1h 02m 56s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e317062b15784cd0a7e558aea8520bfb : SUCCESS in 1h 00m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecb16e496953447c9b76f6b952a5cce2 : SUCCESS in 56m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a03857eb1aec438aa4cf0c38116e119c : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2bcbe3a4b5c64b47921a5ca9bae2c59d : SUCCESS in 1h 26m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/36446155faed4e2cacf798605a912774 : SUCCESS in 17m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b15d72eabb334a53ad11b28d828659fe : SUCCESS in 20m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/500a44a8a28749f18d02271d166e64c1 : SUCCESS in 1h 55m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9fe68aab74b4431a9bd12c2b2e53625 : SUCCESS in 1h 45m 04s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7d62bf6095b14478a1f1b695d26e4cab : SUCCESS in 2h 00m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9d5ad3fb33154adcb803778b12272aa3 : SUCCESS in 16m 58s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ae9b785c17b545a78340b39aa9f43308 : SUCCESS in 15m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b34bad292394151bbe2b78179c0d13b : SUCCESS in 16m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fab1b39e9fe441c0aa5608f2523d50fc : SUCCESS in 14m 45s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/87033fcee5a04622aea96ed65d0946a3 : SUCCESS in 37m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a45c631623c74eb88b60ad6b7f53b07e : SUCCESS in 32m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/715d8c3ef3db47469e09e081ef8e80ce : SUCCESS in 28m 26s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b9fb47bf81ab483f93a457fe3206117d : SUCCESS in 34m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/14c3e25d059f4b058340de2346ed9445 : SUCCESS in 14m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/42018c7ff173430fbe014107eb54c9c5 : SUCCESS in 13m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1db2216e3df448e587e25d737f5f70ef : SUCCESS in 11m 56s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/8b4987456ca347d9a84ea8feb2400be8 : SUCCESS in 11m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5e73a8cd2836499bb87c534d97b6871f : SUCCESS in 12m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3d8e60f76594b21bdf349dec0da0b5e : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"018c93f5befb47ad69f93e6ea04a9073806b2dbd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-28 08:42:33.000000000","message":"Abandoned\n\nkolla-ansible fix was merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9420925d7e3b8288ce4d1ea48e4faf3ec93affff","revisions":{"9420925d7e3b8288ce4d1ea48e4faf3ec93affff":{"kind":"REWORK","_number":1,"created":"2025-03-27 12:48:15.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/05/945705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/945705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/945705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/945705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/945705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/945705/1"}}},"commit":{"parents":[{"commit":"6098c8fe0141a6fbf1ff56dd4086cadeba1c765e","subject":"Merge \"Bump the mrlesmithjr.chrony role to v0.1.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6098c8fe0141a6fbf1ff56dd4086cadeba1c765e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-03-27 12:47:49.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-03-27 12:47:49.000000000","tz":0},"subject":"DNM: Test kolla-ansible ironic-inspector fix","message":"DNM: Test kolla-ansible ironic-inspector fix\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/945704\nChange-Id: I83bfe555bf4cd7038fab8a95dd5fc13f07ade8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9420925d7e3b8288ce4d1ea48e4faf3ec93affff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9420925d7e3b8288ce4d1ea48e4faf3ec93affff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
