)]}'
{"id":"openstack%2Fkayobe~945804","triplet_id":"openstack%2Fkayobe~master~I50d04bd4d3cc02bbbc667adc35915cbfdd1b094e","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-03-28 15:33:11.000000000","reason":"Change was abandoned"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-03-28 15:33:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I50d04bd4d3cc02bbbc667adc35915cbfdd1b094e","subject":"CI: migrate RabbitMQ queues before upgrade","status":"ABANDONED","created":"2025-03-28 10:41:22.000000000","updated":"2025-03-28 15:33:11.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"099072f8e9965982b96c3d09206ad861ab8a9cbc","_number":945804,"virtual_id_number":945804,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-03-28 15:17:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-28 11:01:57.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-03-28 12:29:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf78a8d32fb214c8cfb4fd0ef3103aa8456a5fa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-28 10:41:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e8f049498250ba35b4ab5e4a91584b8be72ba7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-28 10:43:04.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f25da9f77306cda5e1bea6abb5e37f8de1d263f9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-03-28 11:01:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"431c470e68606f3be0b86b3f7077730f97bbd22f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-28 11:07:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8e4533171b0bdd3b5b69b1374fae33a68673de12","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-28 11:07:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0eb024cce9c9618a96b47565968629e97392a905","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 12:29:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ff8d3c0137b4db1b2f87d2954ab3c44\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0760b771af47400e86750860e0ef92ce : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3448c35e83442a5a0b9fcbacb4d76b7 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2362759136034a308579863776bdb06f : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65472f733bb94902b4e98428e38ed21d : SUCCESS in 9m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5acb253d400944c08ebc0edf91c3a653 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c0a141da2f214abda2737e7c8b09a805 : SUCCESS in 15m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81bcfe3d9cdd44edb3bcea83b5f79d20 : SUCCESS in 9m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a935da7cc22b4e61aeffdb9371c2bba9 : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/fce9a1639aac4fa08ae2041eb17494e5 : SUCCESS in 1h 08m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/607a097fac6e4c0b89b75d334385f2f4 : SUCCESS in 27m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6664e2aa331d4c8a84bbcc3ecb1fddfc : SUCCESS in 30m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/947b232d44dc48ee9d38f841aa06b5c4 : SUCCESS in 1h 13m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4cbc477c0e474566a7ab660a034a1d3f : SUCCESS in 16m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af4d0a25f9fc448a952f2b5fe8465bc6 : SUCCESS in 14m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f2425bbe7f9f41b1b25b1f8ec12a8a59 : FAILURE in 59m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/876bfc84cb9342e383bb6122c4d884b2 : FAILURE in 33m 24s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8c6b9b6220774a5895b64fc04aced3d2 : FAILURE in 59m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1105ff1656954c9490e98586cda20ec2 : SUCCESS in 15m 52s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/903bfb5e73dc46febca8cedc5d581465 : SUCCESS in 17m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81e95b09b4b34f868c02c6674cf0bb75 : SUCCESS in 16m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7544092ab86242e9aab8609da11fff6d : SUCCESS in 17m 36s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f7925a07d3fb4848aa55ed54ceb9b1ef : SUCCESS in 42m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/84fad812875d4398909fcf5b84da36a9 : SUCCESS in 32m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b004d2abc6fd416bb28e1e72b92411de : SUCCESS in 29m 28s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/19cd7f1ab4364e5aad8552442a31ed84 : SUCCESS in 21m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b211d42dcf264c23985ff260857c5841 : SUCCESS in 11m 34s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f02efdd62c28413bbaefee03cebb790b : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66b50f5222a44abeb6f0a6aa62c1e767 : SUCCESS in 10m 52s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/76d0966b2d4e4e15be235f4f02cc3d70 : SUCCESS in 5m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/32442caccf7b4d019dec0499077e8643 : SUCCESS in 11m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/264c27db451b4f45b66ccd046d2992e2 : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":3},{"id":"dafcf98b9afd7bf3c96a48a9d2b0de29d86e94ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-28 13:31:17.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"13e7dcc88392512b1d688250196cf420828a10a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 15:17:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b1228355af2494194f31d4605dc5591\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/684d4bf115604ea9b2f6c4e7b81ad58c : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d3ba18966da47c79911a312ef988a66 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d242ca64bde41769a127185a8907af7 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7bfeac89756435689118a7b834c4b84 : SUCCESS in 8m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c07e22fb4ab4aec9c549a1c5391d7ed : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f650a0af49ab4f3eb22e75d6d2fde24e : SUCCESS in 12m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7fc0dd3f33e34d92b1fd2c9d9e02b77e : SUCCESS in 8m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bab80d6c6cfa4ccc8d27352494262bfa : SUCCESS in 1h 02m 10s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a15cd28f43a14528a060a5d54c95a928 : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37f9dd6e8d894465b1b34d9f711a23f2 : SUCCESS in 59m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9eb84ac946e2498488609cf13cc888a8 : SUCCESS in 1h 04m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/38f136344a784bfd97379b34e0c627f8 : SUCCESS in 32m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/92850518225c4a1a988a3ef3b07b1220 : SUCCESS in 11m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c38aac790ed4dbdb25de131eb93923a : SUCCESS in 12m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/35dc58bc5aa94683b6dd2a9f8861dad6 : FAILURE in 1h 06m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d34ed05a02547748b23b28544c92bcd : FAILURE in 58m 52s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7029f31480f44413b67fbeda31e1e31b : FAILURE in 42m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f75bfa70a636485bb0b2c7551d354799 : SUCCESS in 15m 27s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1849d3f8fb62452bb8d223e2f26072ce : SUCCESS in 15m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4705feab6679450eb203b4d1b715bbae : SUCCESS in 7m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/747885a91e5748c3a532584f873636fe : SUCCESS in 15m 23s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6a78b71bfec3459487e61e952efbe6e8 : SUCCESS in 38m 05s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/56308950927843ed9578fb0a0cec3081 : SUCCESS in 32m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bae7526aa7341329ceb9d06e81f0bea : SUCCESS in 18m 41s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/47d78ffd81d0412c8d3c954da9405df6 : SUCCESS in 34m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c41df6ee3f0042b1b811d9921a8000e4 : SUCCESS in 13m 11s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/9b8b845afc294d89b6726e5a09d4dec5 : SUCCESS in 14m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/115900e300774391b589c120d60ad038 : SUCCESS in 10m 52s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d53257833c7545af9c6b3b4b3995a087 : SUCCESS in 11m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f5614f375d5946ee8dc3f79d2da74e91 : SUCCESS in 14m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b556802ea6e845d38f781bfc1487fe78 : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":4},{"id":"099072f8e9965982b96c3d09206ad861ab8a9cbc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-03-28 15:33:11.000000000","message":"Abandoned\n\nWe can\u0027t test these migrations properly yet, let\u0027s disable the features in upgrade jobs for now: https://review.opendev.org/c/openstack/kayobe/+/945847","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"93dd02179a60f57c70986c02b16844feb23ed743","revisions":{"b53a37f8b5ffeca2aa339b3d125661efa94fbc63":{"kind":"REWORK","_number":1,"created":"2025-03-28 10:41:22.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/04/945804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/945804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/945804/1"}}},"commit":{"parents":[{"commit":"6098c8fe0141a6fbf1ff56dd4086cadeba1c765e","subject":"Merge \"Bump the mrlesmithjr.chrony role to v0.1.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6098c8fe0141a6fbf1ff56dd4086cadeba1c765e"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-03-28 10:37:03.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-03-28 10:37:03.000000000","tz":0},"subject":"CI: migrate RabbitMQ queues before upgrade","message":"CI: migrate RabbitMQ queues before upgrade\n\nThis will fix the Kayobe CI for now, but it is not the proper time to do\nthese migrations. Instead, they should be performed with the previous\nrelease. We should change this behaviour once the following patches are\nbackported to 2024.2 and 2024.1 (for SLURP).\n\nhttps: //review.opendev.org/c/openstack/kolla-ansible/+/928665\nhttps: //review.opendev.org/c/openstack/kolla-ansible/+/928669\nhttps: //review.opendev.org/c/openstack/kolla-ansible/+/924623\nhttps: //review.opendev.org/c/openstack/kolla-ansible/+/927497\nhttps: //review.opendev.org/c/openstack/kolla-ansible/+/916911\nChange-Id: I50d04bd4d3cc02bbbc667adc35915cbfdd1b094e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b53a37f8b5ffeca2aa339b3d125661efa94fbc63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b53a37f8b5ffeca2aa339b3d125661efa94fbc63"}]},"branch":"refs/heads/master"},"cc1c5792a4f90f70a1b85cf886d1832f687818ee":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-03-28 10:43:04.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/04/945804/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/945804/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/945804/2"}}},"commit":{"parents":[{"commit":"6098c8fe0141a6fbf1ff56dd4086cadeba1c765e","subject":"Merge \"Bump the mrlesmithjr.chrony role to v0.1.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6098c8fe0141a6fbf1ff56dd4086cadeba1c765e"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-03-28 10:37:03.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-03-28 10:42:48.000000000","tz":0},"subject":"CI: migrate RabbitMQ queues before upgrade","message":"CI: migrate RabbitMQ queues before upgrade\n\nThis will fix the Kayobe CI for now, but it is not the proper time to do\nthese migrations. Instead, they should be performed with the previous\nrelease. We should change this behaviour once the following patches are\nbackported to 2024.2 and 2024.1 (for SLURP).\n\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/928665\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/928669\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/924623\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/927497\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/916911\n\nChange-Id: I50d04bd4d3cc02bbbc667adc35915cbfdd1b094e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc1c5792a4f90f70a1b85cf886d1832f687818ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc1c5792a4f90f70a1b85cf886d1832f687818ee"}]},"branch":"refs/heads/master"},"8b0ce01ebd32ea46b211365605a0bb59a56796c1":{"kind":"REWORK","_number":3,"created":"2025-03-28 11:07:20.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/04/945804/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/945804/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/945804/3"}}},"commit":{"parents":[{"commit":"6098c8fe0141a6fbf1ff56dd4086cadeba1c765e","subject":"Merge \"Bump the mrlesmithjr.chrony role to v0.1.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6098c8fe0141a6fbf1ff56dd4086cadeba1c765e"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-03-28 10:37:03.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-03-28 11:07:13.000000000","tz":0},"subject":"CI: migrate RabbitMQ queues before upgrade","message":"CI: migrate RabbitMQ queues before upgrade\n\nThis will fix the Kayobe CI for now, but it is not the proper time to do\nthese migrations. Instead, they should be performed with the previous\nrelease. We should change this behaviour once the following patches are\nbackported to 2024.2 and 2024.1 (for SLURP).\n\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/928665\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/928669\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/924623\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/927497\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/916911\n\nChange-Id: I50d04bd4d3cc02bbbc667adc35915cbfdd1b094e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b0ce01ebd32ea46b211365605a0bb59a56796c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b0ce01ebd32ea46b211365605a0bb59a56796c1"}]},"branch":"refs/heads/master"},"93dd02179a60f57c70986c02b16844feb23ed743":{"kind":"REWORK","_number":4,"created":"2025-03-28 13:31:17.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/04/945804/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/945804/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/945804/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/945804/4"}}},"commit":{"parents":[{"commit":"6098c8fe0141a6fbf1ff56dd4086cadeba1c765e","subject":"Merge \"Bump the mrlesmithjr.chrony role to v0.1.6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6098c8fe0141a6fbf1ff56dd4086cadeba1c765e"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-03-28 10:37:03.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-03-28 13:31:08.000000000","tz":0},"subject":"CI: migrate RabbitMQ queues before upgrade","message":"CI: migrate RabbitMQ queues before upgrade\n\nThis will fix the Kayobe CI for now, but it is not the proper time to do\nthese migrations. Instead, they should be performed with the previous\nrelease. We should change this behaviour once the following patches are\nbackported to 2024.2 and 2024.1 (for SLURP).\n\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/928665\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/928669\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/924623\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/927497\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/916911\n\nChange-Id: I50d04bd4d3cc02bbbc667adc35915cbfdd1b094e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93dd02179a60f57c70986c02b16844feb23ed743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93dd02179a60f57c70986c02b16844feb23ed743"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
