)]}'
{"id":"openstack%2Fkayobe~947088","triplet_id":"openstack%2Fkayobe~master~I6af9c6fe5b1ec16e55668dfd83f396386ae326bc","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-04-22 17:29:08.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-04-22 11:52:04.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-04-22 13:06:54.000000000","reason":"removed on reply"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-04-22 18:43:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6af9c6fe5b1ec16e55668dfd83f396386ae326bc","subject":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"","status":"MERGED","created":"2025-04-14 10:49:07.000000000","updated":"2025-04-22 18:44:47.000000000","submitted":"2025-04-22 18:43:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"revert_of":945847,"submission_id":"947088","meta_rev_id":"34cd737a97f77ef11d9bfe8b0009ffb7b130029d","_number":947088,"virtual_id_number":947088,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-22 18:43:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-04-22 17:29:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-04-22 11:52:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-22 13:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-04-22 17:29:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-14 10:49:07.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-04-14 10:49:07.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-04-14 10:49:07.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-15 07:56:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"20d1c1a59032f676b246cf04ba73b7f39ce2ad87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-14 10:49:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9872b46c36f5d7bf48624732ab909e643405863d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-14 10:50:14.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"61e3c2b686b57d185c9b6d6444e6066bbe08b532","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-14 10:50:40.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"49266545be50af27a3a6306991e9176a4539ca50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 12:27:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a314f2cb9da4d1fb60bf3d89af3a9b7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1564289f841b4ebfa2dcf7e980758f18 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c669f862ffce431681a8996f7ce8b9fe : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a09de944549e457eb365386ad9812684 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3efc2bbec35747f4a07ab2fdfb084515 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61d11677daac40d9bec646f2dcfed1e9 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b30499fab214ac3924d1c27cbc39bdd : SUCCESS in 10m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/923f18b336024b5a95a85742b3e02f31 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3e32e47700e34e0587a7fa90ad3660fc : SUCCESS in 6m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac6989d22f2f4ce99363a80a50508d20 : SUCCESS in 10m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ecb74d5eddd4457485f959705d10230e : FAILURE in 7m 00s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e6a52d5995d141b0bb9509934bd5c867 : FAILURE in 12m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47aa9e9c75e0449eb113a2a9fd0245f8 : SUCCESS in 34m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b3b456eb17c149a4ac854fa967fb25e1 : SUCCESS in 31m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3db4014a70574574973a9457d9f345bb : FAILURE in 9m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/21703b9072d240fb84f899642813b47a : FAILURE in 11m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc3ad8bcc25d4062b1c6a2f8b866c991 : SUCCESS in 13m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c6373031021147d88369a80d9296d6f5 : FAILURE in 10m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0889f7e04d154dd189e6d806cf3da460 : FAILURE in 1h 34m 38s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7d5c15a0c4274463abbb3dffccd9b2f9 : FAILURE in 1h 28m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/eba3d3c1d51a47d6acdf8fee60cea48d : FAILURE in 8m 29s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5e85d8c2a1ec48a4a727c120d81160d7 : FAILURE in 5m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9eea9485f45e49aca264dcbbba4c4dd4 : SUCCESS in 7m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cceb00a26be34bc2bcc10fc2def0bd58 : SUCCESS in 14m 33s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/90d4b3616a1b4df8b4465b93ff6c4993 : FAILURE in 9m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cea66452551c4095a18b9ed15e6f486e : FAILURE in 10m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79128f0eefd2416a9b5d27a04999c81d : SUCCESS in 40m 03s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/34eff10cff854c8993201aa5d1351a0b : SUCCESS in 30m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4576daaae654428597975961882baf9b : FAILURE in 7m 32s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/c42df71b5e6848dc80902396d683ae45 : FAILURE in 7m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba88fdb650cb4f96a79e75c34b67494f : SUCCESS in 10m 12s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/47e46fae0819464dbcadc5c2de73b699 : SUCCESS in 11m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7d6aa7f2b6774abe8e9985c6834d4c1a : FAILURE in 8m 24s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d18d20a1587a483fb3cfb236d58cb9a0 : SUCCESS in 12m 33s","accounts_in_message":[],"_revision_number":3},{"id":"22749eb364419060ed40a47daaaf731256ce4a0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-14 13:56:56.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f2ddc6018e2a78ca85b77c3ea26fbc95d9a2c4cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-14 14:43:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c9336dfcd606913f2c769e572403d8455d4c9c7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-14 14:45:53.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"94e5e2e983b75a83b53bfdb318b64ed194bef615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 16:30:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35f246a7edf84680a205e8461451b2fb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2987c8f01eda41b79268af6a2ec594c6 : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e33df22207d4dca8b04f3cef55502d4 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53c55e87e0c343ea8eb778fadff609bc : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b2629c2fd054bf0a4fb6ad509192c5f : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee3bea664a1c40f2aab51eaf20211f3a : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f894b48f5df4f48b673bd4b0b470338 : SUCCESS in 9m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/015d424f8f3e41f6b8118d15b8a6ff99 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c3ece3724b994de68d4ea7694b550f3c : SUCCESS in 10m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bf18d315c91c4d428108e4185a78484b : SUCCESS in 10m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1f0794cc6f154e658411c8fc8be6c611 : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/cb3ce60acb38483d8377be01d1ea8c6d : SUCCESS in 33m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f8b03a1438f44fda38a253a4e619054 : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/da258d7d65c745f2a0c3695756af27a1 : FAILURE in 16m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1bb0184c3d434317a86c7a63740dd555 : SUCCESS in 1h 01m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/520ffe812d6b4524a5b888f9cda599f8 : SUCCESS in 16m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9e5205f22eb4b68b2cc678834c061dd : SUCCESS in 12m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/46450b0f8eb946b783b7bd19558a8128 : FAILURE in 8m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b38b9f3ea8d401da09cf4c71500d19b : FAILURE in 1h 37m 29s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6723d92ecd294debb0ebb86f4536b890 : FAILURE in 1h 35m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b00ca78e6c1a4d8c93185e6a025dad73 : SUCCESS in 17m 19s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/31b2a68c4d124910a2bb7db1a847dc92 : SUCCESS in 16m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ef03d2c386c496697391dd23a250538 : SUCCESS in 17m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9b1b192c35934b89b1623076f6f6f29d : SUCCESS in 14m 57s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1fc9bde6068b4f6491ee44ac5ab52e74 : SUCCESS in 41m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/01428e587c314f9f9c30896db1754516 : FAILURE in 10m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0041100d991946a485acda817e4b47e1 : SUCCESS in 32m 12s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a070b90509cb4c0bb7c3670413edf711 : FAILURE in 16m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5b6169d7ba764cf89c94d060e638b2c1 : SUCCESS in 7m 23s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/13e37d7df3814b3795af86ec24c1d306 : SUCCESS in 11m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad3b14338f794abcb80e61a881f00942 : SUCCESS in 11m 08s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d231577bac4244d09b943651bdfb8d5a : SUCCESS in 9m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f63a6cc8c6cb4fcdbb427ae334735bfa : SUCCESS in 13m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adda5bac70f54da5911a8f77db071b47 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":6},{"id":"79a26a882dc02e6263a38477c56eedafad922689","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-15 07:56:26.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"4ac7d1ba9f6f391594271d856856d1f6310e6880","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-18 13:29:33.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_35263\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"_revision_number":7},{"id":"e2d6b02d231df881c308c5e025f38b8483251865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 15:45:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/576a779d60984ebcb614d992dca4cf81\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33d2fac91df8486eb00a2015d3fed03c : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/376eb065571c48f89d26e68f0d9c4d3b : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/374581e7c884476480f07ca1fc7f9045 : SUCCESS in 5m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ef4546323bf484788ccfb6294adcae4 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2364a758561d43398b5c25fca2e8f9e1 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3bed513d534a459fdadc81dade0836 : SUCCESS in 9m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8e823e724be741a49592877123c0cab3 : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/240b7098db814a38b58c607cacdb3ecb : SUCCESS in 9m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ea912f78ab9b4274bbc8da63d96eeaf2 : SUCCESS in 11m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/224b310d59dc48dea79f207f9c46219a : SUCCESS in 1h 01m 30s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/d370ad8623684cb4a09fd0341bb9149a : SUCCESS in 31m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4a69b638bf246b1828a77703d5efab9 : SUCCESS in 47m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/56b9c794160b43af9d3c245d1826f231 : SUCCESS in 53m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/328301063e414abc91463bdc1a1b79c6 : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/88ffe6d71baf4cf0be2d850691becfec : SUCCESS in 14m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce02f83fec304c3b8bd651a743c73f95 : SUCCESS in 7m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1350f073e7144969806652d8149cc68d : SUCCESS in 1h 55m 41s (non-voting)\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab913551090849a9a8fdb59a1c805a27 : SUCCESS in 2h 10m 48s (non-voting)\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/62f13cefcba34eccb3b2db332e85f24d : SUCCESS in 1h 50m 23s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a10e458c728d4810b4c8ce140f7fc581 : SUCCESS in 16m 08s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/a979f6685fca4243a8cf987b62635afa : SUCCESS in 8m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a44e601997ca40639140ee63a12b0a29 : SUCCESS in 14m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/62e830f9c9f94a658d96f1f3d351a6dd : SUCCESS in 13m 40s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/332abe8aef574f449085ad4b6fa47a2e : SUCCESS in 29m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5bf94ca85b034112b14bf1923a8ba0a4 : SUCCESS in 38m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/faf8c0f2f11a477791a1149c2cfbbb2d : SUCCESS in 28m 55s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/72535dfe35c04b1d821f4e0babc5feb7 : SUCCESS in 34m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1e43c142e6c941c5ab4a97287ebbda91 : SUCCESS in 14m 35s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/16dd65ae0cc64c4d98df738a4c55f0a8 : SUCCESS in 14m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cf6cfbd7cf349a38dc89316d2861dca : SUCCESS in 11m 05s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/1e5a0b8dbd9c468a9d3f7528076a8762 : SUCCESS in 13m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/89cfd069e0cc47ba97e8112ea69acf62 : SUCCESS in 13m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2559b091a5543fc9f04ab532923c3fb : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":7},{"id":"1df4a1a9eb846dcc10a2812bbdd2ad8ff16cdd81","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-04-22 08:48:40.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"bc2c863b02e6693a949b27a6810763eb2b4a4c6b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-04-22 11:31:59.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b7d2d6d4d5045b10a23e8ecbd9521cac285de442","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-22 11:49:17.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"dc86ef9e638027da21fdac51a61ff625cba21663","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-22 11:49:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6e1a3621c02e72ed89201e35fe417f98aa64daa6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-22 11:52:04.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"bbf9246c92ff2ac3dbe0aee715edd6f1f3143d69","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-04-22 13:06:54.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"c9d15931bad14fa10fd658349bbb663c6a0c74ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 13:58:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecdcd6c236694e1bb83134434fd95d20\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d5f073a70b44975bd15fe02f2ab882a : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5a0604b3fc0474180dee19c478a6cdb : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5900aa8ec8e341d98dd30a7c26b29ca4 : SUCCESS in 2m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/165364753d234fc2a3ef5ba0df59b399 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a877352d9d94eb5aac11aba113a081d : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4973a53389eb4ebc8dd1a28a38543a77 : SUCCESS in 8m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b5fe38352c52488893baf880c902e8f6 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/af068d24a4764a678fa87ca3b627dc23 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3da846aac9724bf3a1e25a396301bd58 : SUCCESS in 5m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d255dce058404cf7b789db6f893dcef1 : SUCCESS in 52m 59s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1e1d8d7fd69e4e2e960758b9200d8ab8 : SUCCESS in 51m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6118fcf54890490ab21ef403a835d222 : SUCCESS in 54m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6c8e89589aab4b978e1a7697be0ebb55 : SUCCESS in 59m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ad53bd7288d147bca1fb438b4d3c5e1d : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ff2cdf3667a64ae99c93bd5529a7be56 : SUCCESS in 11m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17cbe59f4e154ec58fb73f33d209d874 : SUCCESS in 12m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4a57b45f6bac44349eae71c9d351d9a5 : SUCCESS in 1h 59m 47s (non-voting)\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1364f0fa94f4c7799b71303235814e3 : SUCCESS in 2h 04m 32s (non-voting)\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a362a242c2434196919bc10205944617 : SUCCESS in 1h 26m 54s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/873e840f1436447db24eb864c57aa848 : SUCCESS in 14m 38s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/bb018342f28b48db9ee63f99a5f9dd6e : SUCCESS in 8m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0a72007309949dd9a64c5f6b5a76271 : SUCCESS in 18m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/87ae6e11b09149368a3e8f8911b0d19f : SUCCESS in 13m 21s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b17f726472fd4fa8911323c9e99a4372 : SUCCESS in 42m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e66c03b54a5e44e785f086c2637a07b5 : SUCCESS in 32m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfb90f5443ea4361a7e154c05563a476 : SUCCESS in 31m 04s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1b81b3b005594adc9b70c9a4819212c6 : SUCCESS in 33m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9aab791ebda449c7adfb03d25ab62b50 : SUCCESS in 10m 50s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/88eb5b48a3ff487f905dc9b85f137776 : SUCCESS in 6m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e8420245f3c4fff875f72f8e5afb8d6 : SUCCESS in 10m 04s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a49ad85e4318461480c2c11fda975ceb : SUCCESS in 10m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2177209fbeb6428888aa73cc1f4fc367 : SUCCESS in 11m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95c52b9fa40a40258ed274d771d303fe : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":8},{"id":"d41a5c486258b8eaf6d5348680b136f564f5ce9f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-04-22 17:29:08.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ace2ec42472cce5057252f2889e5936197484971","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 17:29:24.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"39739d8da2cec3924824db8a16af8407164a529b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 18:43:50.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6d7335131124587855b20487dc92ae0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79d0c657a3e406085a42933ea5934ca : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c586551296f4451185bbc618e7088da7 : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f97e923973c84fef9eade1db0f953857 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9345531147164fa2a64564aaa7b4e07c : SUCCESS in 7m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e6abd94ce419409389fc45f450aec35b : SUCCESS in 3m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8733d4b6d8cf42a6932e0034bad27cd3 : SUCCESS in 5m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c352340f73bf4af9a03e8ef847125627 : SUCCESS in 7m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0a930b7a21214b8aa56bd65b1d7e7567 : SUCCESS in 30m 58s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7a0979546cb84220aa5077d8902b6be6 : SUCCESS in 58m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/940987bd5c9942c1bcdb3a3f095bf16d : SUCCESS in 48m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/431120cbb1714aae907620aef1d84798 : SUCCESS in 32m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9b91e3009f9846edab3925f6edb4e741 : SUCCESS in 58m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2d83e522e837438696aa760d110560ed : SUCCESS in 15m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de02c0d9768f46f59b0c9272a7448d94 : SUCCESS in 12m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/db347cb96159467bbaa50acabd0b0c0e : SUCCESS in 14m 16s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6a52aa54856d412f80878d57bc718730 : SUCCESS in 8m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/230693f8dfe447d1a4d77f7c54924eb0 : SUCCESS in 14m 36s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d2b2d3d6fad94388aafbea0478634ac2 : SUCCESS in 25m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2617908696f54b90906af4e577620ba5 : SUCCESS in 19m 26s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5341c2da58a5414da7c98eb257103935 : SUCCESS in 24m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4d447c2fdc324872b342f4a6e0857998 : SUCCESS in 7m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be8e0d57713a476982719b9313bf186f : SUCCESS in 6m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/37c5325f456b41e7b2d7c30d1d4bc6ee : SUCCESS in 7m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7984fec28eb14e7594ff3566293aa996 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":8},{"id":"3bcb598268bcbc74a109d1566cea339e2ba5fca5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 18:43:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"34cd737a97f77ef11d9bfe8b0009ffb7b130029d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 18:44:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11b6ff68701d409588de6e95f4f4787c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7832bf204c1430d99cffc883d5de496 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"b1eadcaf6ebb5c43213196cda695cb051dfb8bcb","revisions":{"c0c809cd6a9e73167b7280b015db7cb70e9e4dbf":{"kind":"REWORK","_number":1,"created":"2025-04-14 10:49:07.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/88/947088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/947088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/947088/1"}}},"commit":{"parents":[{"commit":"ccdfeb7696aa7790c86097cf6555da00f7597278","subject":"CI: Don\u0027t use new RabbitMQ features yet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ccdfeb7696aa7790c86097cf6555da00f7597278"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:49:07.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-04-14 10:49:07.000000000","tz":0},"subject":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"","message":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"\n\nThis reverts commit ccdfeb7696aa7790c86097cf6555da00f7597278.\n\nReason for revert: These features cannot be disabled in master now, so we need a proper fix instead.\n\nChange-Id: I6af9c6fe5b1ec16e55668dfd83f396386ae326bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0c809cd6a9e73167b7280b015db7cb70e9e4dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0c809cd6a9e73167b7280b015db7cb70e9e4dbf"}]},"branch":"refs/heads/master"},"430eab42d91e7e6b07731fad9be7414ceae9ce2d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-14 10:50:14.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/88/947088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/947088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/947088/2"}}},"commit":{"parents":[{"commit":"c76284f1b4a8c5cc5d4733b4e8a2d526ebcb9739","subject":"Migrate RabbitMQ queues before upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c76284f1b4a8c5cc5d4733b4e8a2d526ebcb9739"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:49:07.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:50:03.000000000","tz":60},"subject":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"","message":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"\n\nThis reverts commit ccdfeb7696aa7790c86097cf6555da00f7597278.\n\nReason for revert: These features cannot be disabled in master now, so we need a proper fix instead.\n\nChange-Id: I6af9c6fe5b1ec16e55668dfd83f396386ae326bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/430eab42d91e7e6b07731fad9be7414ceae9ce2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/430eab42d91e7e6b07731fad9be7414ceae9ce2d"}]},"branch":"refs/heads/master"},"9cf9f9523353c660781faf5085b170016d06ad02":{"kind":"NO_CHANGE","_number":3,"created":"2025-04-14 10:50:40.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/88/947088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/947088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/947088/3"}}},"commit":{"parents":[{"commit":"8da8780a30913c2e598ec199549311c18919d382","subject":"CI: Migrate RabbitMQ queues before upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8da8780a30913c2e598ec199549311c18919d382"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:49:07.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:50:34.000000000","tz":60},"subject":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"","message":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"\n\nThis reverts commit ccdfeb7696aa7790c86097cf6555da00f7597278.\n\nReason for revert: These features cannot be disabled in master now, so we need a proper fix instead.\n\nChange-Id: I6af9c6fe5b1ec16e55668dfd83f396386ae326bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cf9f9523353c660781faf5085b170016d06ad02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cf9f9523353c660781faf5085b170016d06ad02"}]},"branch":"refs/heads/master"},"12a6607935f5e4f8ef78ae009dc3323ed48e02a3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-04-14 13:56:56.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/88/947088/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/947088/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/947088/4"}}},"commit":{"parents":[{"commit":"577e8af7b3ce8a1c037b8da2ac8a27c44db79e94","subject":"CI: Migrate RabbitMQ queues before upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/577e8af7b3ce8a1c037b8da2ac8a27c44db79e94"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:49:07.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 13:56:48.000000000","tz":60},"subject":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"","message":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"\n\nThis reverts commit ccdfeb7696aa7790c86097cf6555da00f7597278.\n\nReason for revert: These features cannot be disabled in master now, so we need a proper fix instead.\n\nChange-Id: I6af9c6fe5b1ec16e55668dfd83f396386ae326bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12a6607935f5e4f8ef78ae009dc3323ed48e02a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12a6607935f5e4f8ef78ae009dc3323ed48e02a3"}]},"branch":"refs/heads/master"},"777a72658f51272df4f9a00fb94527b80cc61c7b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-04-14 14:43:13.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/88/947088/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/947088/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/947088/5"}}},"commit":{"parents":[{"commit":"5176b3e3b14ada443f5b78bc3741f56cd34f0d34","subject":"CI: Migrate RabbitMQ queues before upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5176b3e3b14ada443f5b78bc3741f56cd34f0d34"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:49:07.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 14:43:05.000000000","tz":60},"subject":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"","message":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"\n\nThis reverts commit ccdfeb7696aa7790c86097cf6555da00f7597278.\n\nReason for revert: These features cannot be disabled in master now, so we need a proper fix instead.\n\nChange-Id: I6af9c6fe5b1ec16e55668dfd83f396386ae326bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/777a72658f51272df4f9a00fb94527b80cc61c7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/777a72658f51272df4f9a00fb94527b80cc61c7b"}]},"branch":"refs/heads/master"},"dd80600d5a097919114b0185b189f0df2c84e1e9":{"kind":"NO_CHANGE","_number":6,"created":"2025-04-14 14:45:53.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/88/947088/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/947088/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/947088/6"}}},"commit":{"parents":[{"commit":"72301eee8873b48ef924eeebf170b379bcef2b8c","subject":"CI: Migrate RabbitMQ queues before upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72301eee8873b48ef924eeebf170b379bcef2b8c"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:49:07.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 14:45:53.000000000","tz":0},"subject":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"","message":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"\n\nThis reverts commit ccdfeb7696aa7790c86097cf6555da00f7597278.\n\nReason for revert: These features cannot be disabled in master now, so we need a proper fix instead.\n\nChange-Id: I6af9c6fe5b1ec16e55668dfd83f396386ae326bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd80600d5a097919114b0185b189f0df2c84e1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd80600d5a097919114b0185b189f0df2c84e1e9"}]},"branch":"refs/heads/master","description":"Rebase"},"246c360c3855cf6910259e394c26fc16b240ff8b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-04-18 13:29:33.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/88/947088/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/947088/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/947088/7"}}},"commit":{"parents":[{"commit":"aefec669129611d9cfd08f9d1946957d613cc478","subject":"CI: Migrate RabbitMQ queues before upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aefec669129611d9cfd08f9d1946957d613cc478"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:49:07.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-18 13:29:33.000000000","tz":0},"subject":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"","message":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"\n\nThis reverts commit ccdfeb7696aa7790c86097cf6555da00f7597278.\n\nReason for revert: These features cannot be disabled in master now, so we need a proper fix instead.\n\nChange-Id: I6af9c6fe5b1ec16e55668dfd83f396386ae326bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/246c360c3855cf6910259e394c26fc16b240ff8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/246c360c3855cf6910259e394c26fc16b240ff8b"}]},"branch":"refs/heads/master","description":"Rebase"},"b1eadcaf6ebb5c43213196cda695cb051dfb8bcb":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-04-22 11:49:17.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/88/947088/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/947088/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/947088/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/947088/8"}}},"commit":{"parents":[{"commit":"aefec669129611d9cfd08f9d1946957d613cc478","subject":"CI: Migrate RabbitMQ queues before upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aefec669129611d9cfd08f9d1946957d613cc478"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:49:07.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-22 11:49:17.000000000","tz":0},"subject":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"","message":"Revert \"CI: Don\u0027t use new RabbitMQ features yet\"\n\nThis reverts commit ccdfeb7696aa7790c86097cf6555da00f7597278.\n\nReason for revert: Now resolved by calling the migration script.\n\nChange-Id: I6af9c6fe5b1ec16e55668dfd83f396386ae326bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1eadcaf6ebb5c43213196cda695cb051dfb8bcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1eadcaf6ebb5c43213196cda695cb051dfb8bcb"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
