)]}'
{"id":"openstack%2Fkayobe~947089","triplet_id":"openstack%2Fkayobe~master~Ic1bec4032d60cfdacfa5627fbaf1c7fc0e755ff1","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-04-22 12:31:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic1bec4032d60cfdacfa5627fbaf1c7fc0e755ff1","subject":"CI: Migrate RabbitMQ queues before upgrade","status":"MERGED","created":"2025-04-14 10:50:14.000000000","updated":"2025-04-22 12:32:32.000000000","submitted":"2025-04-22 12:31:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947089","meta_rev_id":"7bde456c53ef46ffb4f1ce1982bb49e4aa4f2411","_number":947089,"virtual_id_number":947089,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-22 12:31:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-04-22 08:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-04-18 13:29:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-22 11:28:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-22 11:28:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-14 12:34:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-15 07:57:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-04-22 08:48:07.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-04-22 11:28:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"3cc050e5b18bd58bab9eeebd03698fd516fb41d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-14 10:50:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"334cd7e23554baeba92b701e0df8fdafb50d602c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-14 10:50:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9be706117e747eea3220acb79fb581c4ecae2628","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 12:34:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93e4b6913eef4a9586fb009e79693f4d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e8166c5ff2c4f0f989f1e139826a5c1 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f42ae6b9017e49148455fadf1869e83c : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f335ea7670824bdc9c08257d9a3ed750 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3a7f17de8864a079fa68c485f0b450d : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f02769cd55764f3cbdeaa29ac18d311c : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd024fbccb6409b82d91d0c6e7ee907 : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2bf03e9777874c4daa6354a8fe982434 : SUCCESS in 3m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a160caeb36df4de3abecc493a76b2681 : SUCCESS in 5m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e6512e9dc2bf4a888a946e2a76962d49 : SUCCESS in 5m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/df27f4e7a7dd4a09a6c0ab36a5a4cff7 : FAILURE in 4m 36s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/61be85860f164f0e8cc3770bf478f399 : FAILURE in 9m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f30e0b5113c414e9e04cce36e5ec4b9 : SUCCESS in 56m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b2b3637334eb420aa72df365a75d11bd : SUCCESS in 1h 08m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6ded19469a714ed187d3a5ad9cbe8020 : FAILURE in 9m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4d9079cf6a6944299683938ba5223f1a : FAILURE in 9m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/729729e1e48d4d55812a531c777cd5c6 : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/958e579dcaf04d4995d106c85f6db290 : FAILURE in 10m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97daabe2c7324cc78d6a31510d19f554 : FAILURE in 1h 37m 24s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/65b196a9b95f4de196ae4ae5cb187c31 : FAILURE in 1h 42m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/21ff8b34eea442c0aaea0c6fcceb2870 : FAILURE in 8m 45s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/12bb26d460ce44cd84982a25fc488568 : FAILURE in 9m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa1d5fea1f8343daab521459c4d8dab9 : SUCCESS in 9m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/621c38aa259a4c399512f09e56019a7b : SUCCESS in 12m 57s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/32870ec654a349d4a7537f45678a18fd : FAILURE in 5m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/53af0ae688844d4286f3f6b94aa45cc8 : FAILURE in 5m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/252775761fe54fe8811bd2757ed8878e : SUCCESS in 25m 57s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bcc80f95a2ee4306a7f3e570928ecb77 : SUCCESS in 28m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/757c158228364023b45e48cd96d5600b : FAILURE in 7m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b84db0b27200405188aa1dbe80dc3ad4 : FAILURE in 6m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/110638dd6372424b872d80e96f80f127 : SUCCESS in 6m 08s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/597906dccf634ed59a552b4d1790ae21 : SUCCESS in 9m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/885f3b6be24545c4940460a6286507d9 : FAILURE in 4m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/234b13cd4f2c47008220edacda0c3339 : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":2},{"id":"c06ff8e4237a00872f75b8a6bff4f5b9a49a9d14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-14 13:56:56.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"56c962bedda78b0a2f21f34cf4081e1e9c80c835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-14 14:43:13.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"a798400d1daab72d2ea49c2f07873c10cab1944b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-14 14:45:49.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"6e3821cfbae4c9bfaaebdb5ce0a53028809f42b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 16:42:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8de60101c094412bbc0d1d1132493da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f33ec42f780a4b8ab788f4ecd111ccf9 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e980f5de2fb4bf295fd235369428b61 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1a811ec339945e0ae008453f52d42ee : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4836a62908b410f92ed3bde0b7433b1 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4c29a53b3584a2c9a138692806435b3 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/401b4fc49dd948f9863a86da5a6a9d38 : SUCCESS in 9m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82c2fd383e5041218151016e02c819de : SUCCESS in 2m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81a6ce54bd3c469ebe84a088a78ed621 : SUCCESS in 6m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5855826d6b14018835a0d06622a1073 : SUCCESS in 6m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c85b7801ea5a44509ba21639538c2b2a : SUCCESS in 34m 40s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/edcdd3d2d05f49b08d0001df85a16110 : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/675d2fe2ed73455b8173c0058447e642 : SUCCESS in 48m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a69f45eb10b945debde118750bc8e4ad : SUCCESS in 1h 05m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/02ffa1b0f935490ea415abc09ae44663 : SUCCESS in 55m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/afd4cadc50d641e8af3564ad535639d5 : SUCCESS in 20m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d19aaa79d3f743ac89d539c7f042c706 : SUCCESS in 11m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/71a0c5511d8f4f26beb86034c34586da : FAILURE in 7m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39f7027aa08a416f9e2ce8eb6d760d53 : FAILURE in 1h 53m 46s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b42fc979b31847b2a80862192e1049f2 : FAILURE in 1h 28m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5d8f43fde7d947bc873271b24bfc398b : SUCCESS in 17m 45s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/d98f139da7d841e482860ea859f31fb8 : SUCCESS in 9m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4ca970f277f40ec930f402884cfc4d2 : SUCCESS in 17m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/02e31f00bb5c4b599053ccf9d1a47d89 : SUCCESS in 16m 53s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d81a3a9b9c744cdfbcb60858c5bede9d : SUCCESS in 38m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a41d41f891584368ad13cccaa3c0aa74 : FAILURE in 9m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b047a3c7492647528a902cac1a65c272 : SUCCESS in 38m 05s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4e5339b711124f61b0cca1f65ae87209 : SUCCESS in 32m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/66bac006e9654ceb8daa3f6f1b8766fb : SUCCESS in 15m 51s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/dc5e51a8dc144a36aca97e33e2c9a329 : SUCCESS in 11m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20192efc83224970a7fe4bf0ac9eaa46 : SUCCESS in 10m 01s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/dd6c666c27584ca486dda2502a258c5c : SUCCESS in 11m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d5dbe23722ed4c4692a90f8f72299a41 : SUCCESS in 12m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9565e33a1af44fa3bc432f4ff36f1828 : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":5},{"id":"2c3e7caddd4f5ecd92d2891e230ad853ffd121a5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-15 07:57:32.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"22af544a9148c7be35efa46a4abd3f00cdef8489","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-04-17 12:47:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f67e5e5b0530942473a384240d45620818d295cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 14:53:40.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09598dde245d40bd8be7b4e578f54708\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2330fadd57c745909203ca7c4d5f85f2 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f09a4dbaa124bec9b2030a651cd0c0f : SUCCESS in 2m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9308453712ae411082f3d29acc598269 : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec2fb008d52843848ab0dacb050297e6 : SUCCESS in 2m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4803c67bbc804e5f99aa602a6c49d5c8 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e044668c7a4448a0a14ffa1354728dd1 : SUCCESS in 4m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d6b077ebd3ad48438dc85766fc9752cd : SUCCESS in 2m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a874c6d616246448460c810263c2214 : SUCCESS in 5m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6245d7c489074c6ca512a089d505143f : SUCCESS in 8m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2c6fe71ee4a24b2083e990daf69ff322 : SUCCESS in 52m 45s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/80288766fee34c7e8171f673e2f18b72 : SUCCESS in 52m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/043314220a6f4085958860ad4a4680a3 : SUCCESS in 48m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d77f47cc09b34b39a8e804983c3623f1 : SUCCESS in 49m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/921713b436664e9ab9d62580d4f548c5 : SUCCESS in 53m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3cf1bd62a8de4dcebfca45b6f46af9d9 : SUCCESS in 19m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/381722f9bdbe46279356568205a6a789 : SUCCESS in 7m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d699aaf1fc9947fa819ab5d38d2ffa0c : FAILURE in 7m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5ce43c862ee40c2bdd17e345c8fb3c2 : SUCCESS in 2h 02m 17s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a8cd5d9e7c57489f8f878e97d3de68df : SUCCESS in 2h 05m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/09eb91bfc6bb43be837b4c9087cfc72a : SUCCESS in 17m 16s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/9b174a19fcb0482e8cee0ac2c6aeac2f : SUCCESS in 9m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67519c50d66b4fa092b0d4226e53511b : SUCCESS in 12m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7e10b4e75311490b9b2001c95af93c2b : SUCCESS in 13m 47s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e88e2de1251948fe880e9565c0140276 : SUCCESS in 21m 39s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/611c0c286c08409daa6ddef5ef6d3117 : FAILURE in 8m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2be3ea4c90c413db19dcbbfb8429b86 : SUCCESS in 26m 32s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/653d2bfe85534b7e9a2f480a5fdc49f3 : SUCCESS in 28m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/824f1e4bac674c168a2b6cdb06a3f67c : SUCCESS in 12m 35s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/0204785f40574216827ff19b886a79f5 : SUCCESS in 12m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/083edcc0857d491991527fbd75fbfcd0 : SUCCESS in 9m 48s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/4daf9d589de043168d14d8efe6653cf4 : SUCCESS in 10m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/436e6365d31641629aaac22260f842cb : SUCCESS in 12m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0115aea46c644a03b7d3ac027cd2bac1 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":5},{"id":"ddaea431070c29c8a41ee90314902c32ba5ba051","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-18 13:29:27.000000000","message":"Patch Set 6: Patch Set 5 was rebased on behalf of \u003cGERRIT_ACCOUNT_35263\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"_revision_number":6},{"id":"415ad38577aec79de444b0111bdecb6afb2dc295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 15:52:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/031d4e7cbf1a42ef9c9d3d10d1f3b066\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb0b61b211b342378ad48e52755902ee : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/842e07cf7d534086899bf2a6d3fdefad : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8c875cff168491cb65fe3461b85aacb : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/758a5d6852e34b8d8fa9fa559f1577e3 : SUCCESS in 2m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2be2e65feaa4fccb5db83ee86279b95 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904009ed9b0b46618a2db7774eb875e4 : SUCCESS in 8m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/53e3b3c8d18548fca15f9a8ac455d664 : SUCCESS in 2m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/94e96aff3e2b4edc8add5f9880c136f5 : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dd8161a793c14e8a8b93808332e9bb8a : SUCCESS in 9m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/21b6faf545044796b7b53cb5664dbc93 : SUCCESS in 34m 06s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9bc0a674da794f5892ef929f886cc0e3 : SUCCESS in 1h 01m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6966f90995841418e3afa4045a38a30 : SUCCESS in 57m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/48b13dc6dbb3432b88a64036e71630c1 : SUCCESS in 48m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/42958ea238284c90b6c493b4f372476e : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dbd087c557494e6f82f5da6777cbd015 : SUCCESS in 17m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02ca07c10bff4c3d9fd0f814d0efd8bd : SUCCESS in 12m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/49e61222646247439135269508ecf18a : SUCCESS in 2h 15m 41s (non-voting)\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c5840f455d744bca7143bbc781a1233 : SUCCESS in 1h 38m 28s (non-voting)\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e4ef817ad3614acd9377e36f4096a0e4 : SUCCESS in 2h 18m 37s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0021dd05f4ba4d40b28b35a02e1f23b3 : SUCCESS in 17m 37s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/59ad23ba04e5478f9ef37eee831218b9 : SUCCESS in 14m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8419114b6a84db6a4dea3d6d09ff3de : SUCCESS in 14m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/18604500c5d64e048311e29aabd06297 : SUCCESS in 7m 37s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fbe053ffc7274b658c6fec874fae4f95 : SUCCESS in 30m 51s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/db7ab5ed3141419f9aa547b2a8963e36 : SUCCESS in 30m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48e3dcc1826849e2a7da05a582d046af : SUCCESS in 30m 02s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ec65eee6e00f486c895032b7b795c887 : SUCCESS in 28m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5e1c59e0b097449fa2f6bd43be41a226 : SUCCESS in 6m 46s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1c429305777644a5bc3098a7fd385240 : SUCCESS in 12m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8709236b9ac4047bfd9e2e2d7b74f5a : SUCCESS in 11m 18s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/86afeaedfb274824aa31fae493e139f5 : SUCCESS in 10m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d214046215e94308bb7fa6e16b3e16a5 : SUCCESS in 10m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97c3f788869c40bcbd64d4d3a6cc5dc7 : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":6},{"id":"172432fa2e6c6bd021b8e9cf098e7d891f66ace2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-04-22 08:48:07.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7a7ea5335c45c0dc553ea9ca3782c9ebe4f9a11f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-04-22 11:28:27.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a075ddba1849129c4384df688a127f8acee352b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 11:28:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0cb53a9c71125c6c2147b825885bf9733e82d812","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 12:31:34.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7ab69e789f74503afb081b96c2eb15f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b068ce5ce8254d18972bc23b5a93d299 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8de0c29b9da841259152fbf84e4e0906 : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/307748cf38ea49bf849c928fa9b04796 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11976d850d0f497a8148a2aa225abc26 : SUCCESS in 9m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c180c06960224ae1baad87320a99d4c6 : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98824b2261ab43cc99bf0fb3adf9fe39 : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/90bbdfc0d7254a09965fe0a48b188ed3 : SUCCESS in 8m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0ad20576589346d1b471b0fbdb721237 : SUCCESS in 52m 48s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/2dd9cb62e4af40dba66c092cd4196b19 : SUCCESS in 53m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/142fe5df82dc48489d970ab4fb7f9c66 : SUCCESS in 47m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/260765b00f5f42ec903c4971b042d493 : SUCCESS in 54m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f641c82923964d99b1b4dfe9109bd23d : SUCCESS in 54m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ad1339f8c0fe4d1d8160fc26018a97b0 : SUCCESS in 15m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2293888fd5a24b7791556d40d12f2293 : SUCCESS in 7m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b0f4cbc8862b4a2092508a7d143e12d8 : SUCCESS in 10m 08s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ead6a0e60c5c4878bf3bc793a83d0991 : SUCCESS in 12m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8704818f8d249ad826f2141e2936261 : SUCCESS in 7m 48s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/37aeee67f5b94b768e8d2e5fe0342744 : SUCCESS in 24m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82be7f0af8f645bb8b35d38626547769 : SUCCESS in 20m 54s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1b3d8878d8f3488e87b723f54c343987 : SUCCESS in 33m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a89b5d2adacf4b29ab8bb5354039c242 : SUCCESS in 10m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ffdaec0e504496aa54dfc74d64dafc6 : SUCCESS in 13m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e4aa416eb95a429380acf70db8fbc9ba : SUCCESS in 6m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab595fd45fd345db8d22646465b33353 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":6},{"id":"c63f032fa51bb1d40cd17fb95dc914243cafcb8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 12:31:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"7bde456c53ef46ffb4f1ce1982bb49e4aa4f2411","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 12:32:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ff5d3a9170c4e9dab01090426efc615\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7435726ae854dc6a9c0b3c8c3c7bc36 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"aefec669129611d9cfd08f9d1946957d613cc478","revisions":{"c76284f1b4a8c5cc5d4733b4e8a2d526ebcb9739":{"kind":"REWORK","_number":1,"created":"2025-04-14 10:50:14.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/89/947089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/947089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/947089/1"}}},"commit":{"parents":[{"commit":"e035946abad20c44c12f94c1aac9a1de32f361b7","subject":"Merge \"Bump up Ansible supported versions to 11.x/12.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e035946abad20c44c12f94c1aac9a1de32f361b7"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:47:04.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:50:03.000000000","tz":60},"subject":"Migrate RabbitMQ queues before upgrade","message":"Migrate RabbitMQ queues before upgrade\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945805\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945846\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946993\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946994\nChange-Id: Ic1bec4032d60cfdacfa5627fbaf1c7fc0e755ff1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c76284f1b4a8c5cc5d4733b4e8a2d526ebcb9739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c76284f1b4a8c5cc5d4733b4e8a2d526ebcb9739"}]},"branch":"refs/heads/master"},"8da8780a30913c2e598ec199549311c18919d382":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-04-14 10:50:40.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/89/947089/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/947089/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/947089/2"}}},"commit":{"parents":[{"commit":"e035946abad20c44c12f94c1aac9a1de32f361b7","subject":"Merge \"Bump up Ansible supported versions to 11.x/12.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e035946abad20c44c12f94c1aac9a1de32f361b7"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:47:04.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:50:31.000000000","tz":60},"subject":"CI: Migrate RabbitMQ queues before upgrade","message":"CI: Migrate RabbitMQ queues before upgrade\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945805\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945846\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946993\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946994\nChange-Id: Ic1bec4032d60cfdacfa5627fbaf1c7fc0e755ff1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8da8780a30913c2e598ec199549311c18919d382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8da8780a30913c2e598ec199549311c18919d382"}]},"branch":"refs/heads/master"},"577e8af7b3ce8a1c037b8da2ac8a27c44db79e94":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-04-14 13:56:56.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/89/947089/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/947089/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/947089/3"}}},"commit":{"parents":[{"commit":"79b1f79338d8b5595a58937f2c2c090fa8da6fe4","subject":"Fix network configure on Rocky hosts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79b1f79338d8b5595a58937f2c2c090fa8da6fe4"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:47:04.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 13:56:48.000000000","tz":60},"subject":"CI: Migrate RabbitMQ queues before upgrade","message":"CI: Migrate RabbitMQ queues before upgrade\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945805\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945846\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946993\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946994\nChange-Id: Ic1bec4032d60cfdacfa5627fbaf1c7fc0e755ff1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/577e8af7b3ce8a1c037b8da2ac8a27c44db79e94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/577e8af7b3ce8a1c037b8da2ac8a27c44db79e94"}]},"branch":"refs/heads/master"},"5176b3e3b14ada443f5b78bc3741f56cd34f0d34":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-04-14 14:43:13.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/89/947089/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/947089/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/947089/4"}}},"commit":{"parents":[{"commit":"c20f52978a0da4e9405a3d74932625cc0ae35669","subject":"Fix network configure on Rocky hosts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c20f52978a0da4e9405a3d74932625cc0ae35669"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:47:04.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 14:43:05.000000000","tz":60},"subject":"CI: Migrate RabbitMQ queues before upgrade","message":"CI: Migrate RabbitMQ queues before upgrade\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945805\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945846\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946993\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946994\nChange-Id: Ic1bec4032d60cfdacfa5627fbaf1c7fc0e755ff1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5176b3e3b14ada443f5b78bc3741f56cd34f0d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5176b3e3b14ada443f5b78bc3741f56cd34f0d34"}]},"branch":"refs/heads/master"},"72301eee8873b48ef924eeebf170b379bcef2b8c":{"kind":"NO_CHANGE","_number":5,"created":"2025-04-14 14:45:49.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/89/947089/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/947089/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/947089/5"}}},"commit":{"parents":[{"commit":"563a850d3d69314429f3611f5ed896de658ac21a","subject":"Fix network configure on Rocky hosts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/563a850d3d69314429f3611f5ed896de658ac21a"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:47:04.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 14:45:49.000000000","tz":0},"subject":"CI: Migrate RabbitMQ queues before upgrade","message":"CI: Migrate RabbitMQ queues before upgrade\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945805\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945846\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946993\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946994\nChange-Id: Ic1bec4032d60cfdacfa5627fbaf1c7fc0e755ff1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72301eee8873b48ef924eeebf170b379bcef2b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72301eee8873b48ef924eeebf170b379bcef2b8c"}]},"branch":"refs/heads/master","description":"Rebase"},"aefec669129611d9cfd08f9d1946957d613cc478":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-04-18 13:29:27.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/89/947089/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/947089/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/947089/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/947089/6"}}},"commit":{"parents":[{"commit":"8ed9f7037a1508e741ec618de9c38c3758ac0d3f","subject":"Fix network configure on Rocky hosts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ed9f7037a1508e741ec618de9c38c3758ac0d3f"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-14 10:47:04.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2025-04-18 13:29:27.000000000","tz":0},"subject":"CI: Migrate RabbitMQ queues before upgrade","message":"CI: Migrate RabbitMQ queues before upgrade\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945805\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/945846\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946993\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/946994\nChange-Id: Ic1bec4032d60cfdacfa5627fbaf1c7fc0e755ff1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aefec669129611d9cfd08f9d1946957d613cc478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aefec669129611d9cfd08f9d1946957d613cc478"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
