)]}'
{"id":"openstack%2Fkayobe~947863","triplet_id":"openstack%2Fkayobe~master~I8ea60974e2ace6876514271be4156e6c889d8776","project":"openstack/kayobe","branch":"master","topic":"fix-openstacksdk-u-c-default","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-05-19 18:07:32.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-19 18:07:32.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-19 13:39:35.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-05-19 18:07:32.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-19 18:07:32.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-05-19 13:40:55.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I8ea60974e2ace6876514271be4156e6c889d8776","subject":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation","status":"MERGED","created":"2025-04-22 18:07:53.000000000","updated":"2025-05-19 18:08:35.000000000","submitted":"2025-05-19 18:07:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947863-fix-openstacksdk-u-c-default","meta_rev_id":"ce252cbc3c42ca1d86cfc96c4d0dbc93c31b8fd9","_number":947863,"virtual_id_number":947863,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-19 18:07:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-19 13:23:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-05-19 13:40:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-05-19 13:40:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-05-19 13:39:35.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-05-19 13:40:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-19 13:36:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-22 18:08:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-04-22 18:08:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-04-22 18:08:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-04-22 18:08:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-04-22 20:45:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-23 07:17:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-04-23 08:47:10.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"722af4a5506ba2491b549c735b63d812fbe9169b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-22 18:07:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"140306f44c28a21c2ff8f8fe153c8478b6ef5991","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-22 18:08:41.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"d6dbeebb78e95cac7d6c44576aa93ebba3278484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 20:45:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f768c855c5c041f0acca3cb05902d0c0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/516a46728313445c8b8bb5e549ab3ed9 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef648ab0f0a240819e6ae0cea7e28047 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69dcb2bd048b4814a9507996b1336c24 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6c5714bb83e4442ba26d141a1d00490 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a09987d2d5945d3af0bda18bc436dc3 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56c507d38c144a4582c8fcdefc316e24 : SUCCESS in 7m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a885cf211a124cf5b045c2a76d2169fb : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ce267fcc9c443f1a31670ff3b19cc69 : SUCCESS in 11m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c02b0ce412445c18673a9db5aa3401d : SUCCESS in 11m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ffcdd3dbb3f24d3aacf04f8441a019e6 : SUCCESS in 31m 08s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/df10a302c4804096804fe0ba5f65c142 : SUCCESS in 35m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cff97081a40488b97b1717bb481dadc : SUCCESS in 49m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9683b0fa4c7840c2a40be3b5c3789198 : SUCCESS in 31m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/11e97f8cee9442b2a63b0c26c2250f43 : SUCCESS in 36m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/237196acad604751945e1b3e683a451f : SUCCESS in 16m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a581fa06eb36447b9d41e0692cd6439e : SUCCESS in 12m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d8d5f9e7264049e5a9f80da6a501cc5d : SUCCESS in 2h 00m 03s (non-voting)\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18edc9c55af24766aa56b2d4fc97cdaf : SUCCESS in 1h 54m 55s (non-voting)\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e403500fa21544bdac2d8bb62360779f : SUCCESS in 2h 24m 25s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d2de7904dc5c4fb9ad10e600ce040ed0 : SUCCESS in 17m 53s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8253e8bcf29347139b51b634677e3a9f : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6bae93d9fa640c1acc6238f8ba61b37 : SUCCESS in 14m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fcd5ee0cf253449680f364052d80dd3f : SUCCESS in 13m 15s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/aceca3fd94e04ed6b0e325a8067534e5 : SUCCESS in 40m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5ae86f41e8c3444d9e7c2680e698d028 : SUCCESS in 30m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c56992bbac65448bbe1b12fd224e4699 : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/40bfab1a9840479ca65004bea89eddd2 : SUCCESS in 28m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9adf3d1cf735431bb0147b3afc42bee2 : SUCCESS in 10m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/73a07d5a35464e1482fc47680865d065 : SUCCESS in 11m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f9579e78ad84b32a69668366f8ed03a : SUCCESS in 10m 35s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/dc06fbb359674863b72cf84232a08f18 : SUCCESS in 12m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/64b5e7123ee94e269faef9eb411f6406 : SUCCESS in 10m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e111ab136dc34ac3b126c00c4852c2f5 : SUCCESS in 13m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ac68b510f33c305ab7e2296846898bcbe95bb8fa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-04-23 07:17:28.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"17159fec5abba7a8d99085a4bab0b57eb69dc339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-23 07:44:03.000000000","message":"Patch Set 2: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"94cad883e23d9a0119cefcdda43fd0912869d3de","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-23 07:49:12.000000000","message":"Patch Set 2: Backport-Candidate+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b9ef8fef3250ac1793486e79f9e4694539b44082","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-23 07:49:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ab3a4e982a253c73c2a1b1e9e77d7a456e5f081","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-04-23 08:47:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a0bd214f89397d2c78cd92fd1ba987e9a58f49c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-23 09:09:17.000000000","message":"Patch Set 3: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"89e52b1aaad66ca138636ed93e20d6dbeed77ed5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-23 09:09:27.000000000","message":"Patch Set 3: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef27a07d78bffcbf67d61b87209b5d1796e5a7dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 11:30:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a01f7fa32a9d4ea68b377219281dbbaf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad96671ce1a64367811bd0b82b68431d : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e6ce4a6c4d430898061ca60d9864e1 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43a23c0297914367bb531785f16ccaf6 : SUCCESS in 6m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f53852a888bf4593b7dd54b4404c06d3 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f62e3e6444744bba87d4048a54efae60 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c3d9d3de7649c499c8e45ebd81d690 : SUCCESS in 7m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/224927bedf3e4491a36a55d8aeca58cc : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/246ae7dc677b4db6ac553a789d329dfc : SUCCESS in 5m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4111ccfe0b504ea9b4d880af3d06a456 : SUCCESS in 9m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/834b02240ae14db6a329844f1a492118 : SUCCESS in 1h 00m 04s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/68ef2df749fb4185860019e7d4f5d7af : SUCCESS in 1h 01m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb99b96ae1a645589a97f4d9e3f17a5c : SUCCESS in 30m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/640f9f3e0f584da1bccc50f36fbc0fcf : SUCCESS in 53m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/017a336942694158bc18ef3db4f16251 : SUCCESS in 1h 04m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e7ec84f910164648a418fdf952672928 : SUCCESS in 16m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/803bd53ddf4544e8b3107c3ece8a3b0d : SUCCESS in 13m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/676d2f8f7e7046568089dba9a610362e : SUCCESS in 2h 08m 20s (non-voting)\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e667ea2123b481d83d5878233948b5e : SUCCESS in 2h 07m 08s (non-voting)\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9066c11afc5f433daeea59c2bce06f2d : SUCCESS in 2h 14m 09s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bf88cc8bc7b14ddb9acfe89c86e4cf38 : SUCCESS in 15m 39s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/c8002222bfdd42b0a7915aca4a188e92 : SUCCESS in 15m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eaa79b30012246bb8f6a7dab54516ef1 : SUCCESS in 13m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d48bd2989c1e43729be5445615859fa2 : SUCCESS in 12m 53s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8017daf56500464c8d407b117fef7ec0 : SUCCESS in 35m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9d76883cc9284106ba1183eac6c5f00e : SUCCESS in 33m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35e94e6c596747a799f243cba41c7147 : SUCCESS in 34m 42s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a7a47cf221124141ba9a3e4abd6d6bf2 : SUCCESS in 33m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/97e694ccebbc46b9add4116e71254374 : SUCCESS in 11m 30s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/739e4b47feb5410e8991313a70c9ffce : SUCCESS in 12m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4945c82ce604a7590f3b01f632d6d5b : SUCCESS in 11m 25s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/6ca0d355a20a498cb8a2dbd7d45eae0a : SUCCESS in 9m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a9e4b668889a4844b84e14c59297df0b : SUCCESS in 10m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b5d4903892a412a9262426a26162e8c : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":3},{"id":"fb7f08e172ea5af3d5715203a3cbdcfc5f9239ef","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-04-23 13:20:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad1b47cf9c12cb6d939b38f88de39efe34b9582d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-23 13:58:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"51959bc21e88867d722de7cd025ef94af0662731","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-04-23 14:26:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"379038f10aee2c56e36b3c9243bf06bcb1bffeb0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-23 14:47:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9ebc630647318f8e2d2ba41a16909e51681fdaad","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-08 20:31:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da4aac621f5614ec3cadb2f1b701eb0198cff1e2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-12 12:16:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"39f644383ab86e21edd133cea76588765ff4a64e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 12:50:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"948bdbef2f6e8f006230448e6ae0f8633cfe3ab9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-12 13:42:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6442253513ba959a2125663414f942d919f1dee5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-13 13:56:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f2a172ca8b6291ba198186152fa670f99b208371","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-13 14:36:41.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"68050c460ffb8cae71cf3fced3d9da16c6bbc990","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-13 14:36:51.000000000","message":"Patch Set 3: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb83416cfcfed56b1e63aefe8a6defe4e8100359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 17:05:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/104f8a3f2da64c26a89011f058145a29\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c1c923630d545748f78d6a333894248 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71377782c3af47c0a637d6830ae7135e : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90d80d3d476743c19f796037120aa757 : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c27f65b4b3847b3b157915b04fcb7a3 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bee9d7b6aa024a539e625b9d7b1e217f : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4940f6492f4b0bb3f9d150bb4964a8 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f68fc973b7d8431b92f88a4ca8dfd5ea : SUCCESS in 2m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b97e9b9d475456fb2964262f551b3d3 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3b159d6df9da49219fa28e4d7a6e6542 : SUCCESS in 7m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94f49b0a070546e88acc452d20211521 : SUCCESS in 10m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3263e0f6b1964f09b18b934855d02d8d : SUCCESS in 1h 06m 05s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/afda0094112a439b91ed21bbf9d63a32 : SUCCESS in 57m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06407eb656b3477cb79fe414f9bec817 : SUCCESS in 1h 00m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/364dc9b0a5c44bdfb682977a4e6eb750 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/166eb68aa08b4e328cd7a2fbe7c7638a : SUCCESS in 1h 07m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5acb58375e5b4874acc895fd6787a8cb : SUCCESS in 16m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/442c69bc0ab24eb2a6ace6330efd9907 : SUCCESS in 14m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6e0ae06d25284263a9de6bee2f0a660e : SUCCESS in 2h 10m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3bfb3c5bfa2447f912cfb5ffa0e20f0 : SUCCESS in 2h 15m 57s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/517b68d844b749a7bbd54f57480d29f8 : SUCCESS in 2h 03m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d60073117ccb4909bb3c407ce529020a : SUCCESS in 2h 19m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3b031a359b2042108d539a5eca29b622 : SUCCESS in 16m 58s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/a418c75a974c426f9ba912d7f2a2c06d : SUCCESS in 15m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/064acb09764b4fbd9f1deff29cffadd2 : SUCCESS in 14m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/83b647440ae94f6c8cee7286bb4597be : SUCCESS in 13m 18s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8f222664d2d84fd4961543ac87dd70f4 : SUCCESS in 32m 48s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5d3e766f607d4a7db09ebb87138520e0 : SUCCESS in 30m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59a43e44dea9457ab75a82b35a1dd486 : SUCCESS in 27m 57s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e0268c206a7d483199cc4ecf872aad29 : SUCCESS in 30m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4cae1475aa7a428bad704af2524dd60f : SUCCESS in 13m 00s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e337d96d3d1142a49be880b6aea88b53 : SUCCESS in 13m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/840a5fee656b45679c5f01ad5cc5c2b1 : SUCCESS in 10m 13s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a4a2fcfdca054572a983a377ea47e852 : SUCCESS in 10m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/68e5bed6d7e44dbb82770bd18a1bd5cd : SUCCESS in 11m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/454363135727410fb9ad4b0c10ea55b6 : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":4},{"id":"52f978e0fc10edc467e654bfb74b1cd16cdc34e6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-14 08:57:39.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"809791f41da2e176d5d3778b66b67ef01af5edbf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-15 09:32:53.000000000","message":"Patch Set 4: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"44ebb2b2deee96ca370aeddfa20fce0338679788","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 10:51:43.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d395cbd07d25d1098786ce5f60c41f6f5eb4e049","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 10:51:59.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"8763bff537d0e629e8d57be6944161bedb3e9296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 13:00:36.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"e291a14a05302aa15f8df07315890ae8276097f6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 13:23:35.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"210aef3273807c735339e933954ace3685631247","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 13:36:44.000000000","message":"Patch Set 7: Review-Priority+1","accounts_in_message":[],"_revision_number":7},{"id":"3fa29f4f4c4de82877d8e2fb57af221d70799262","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-19 13:39:35.000000000","message":"Patch Set 7: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bbac19a510061bbb082f2b8ee88421114461d101","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-19 13:40:55.000000000","message":"Patch Set 7: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"4b36794403634f966e666e68bbe830029d7ab701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 15:39:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cefaa6f59b6744a88ae8dfd05324bbb7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d13eaebd7f7f4e5d9b1ed23a6f9a64e5 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1d91d2b4c24863b98e63ddffba7835 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab6d27e8b1c640f7bd52896e5f3b50fd : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0bbd09dc918a42c09c8f79a8ea1fe593 : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d875df3afa74f859399c1fd8317aca0 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d3b36bc81b4a1e8debbdcfa22cae43 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69b6b90e1536496f8540054cbb5e4bee : SUCCESS in 2m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca1670383b4245559fd0e24de978cd26 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/79afbbf1b1a34fc1b7ca30f036ef5cc4 : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cf3b460f058a4b319fdab98642c4a230 : SUCCESS in 9m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ced325f6f7ab4ee8a78decf7a98c9efe : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a94148ea75924ff89fedb674dfa26863 : SUCCESS in 55m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbfdb4194e8043a6975a9156d574ef5c : SUCCESS in 1h 01m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b0fbcf7573ce4660a1dbd9e19b8a00f6 : SUCCESS in 56m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f5bcff27c2474249a326cd76e0c1bfe8 : SUCCESS in 1h 14m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/936c44f044024d75980246020d0c4b89 : SUCCESS in 19m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9360220510345409aaf565b51d1850a : SUCCESS in 13m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a88bdcc88023486d8eeb430e60d0f5ae : SUCCESS in 2h 24m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38c78b228c9a494aacfe26f67f24f4c3 : SUCCESS in 2h 16m 20s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e29ce5fe92d2484fa3d505899ba73abb : SUCCESS in 2h 30m 46s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/172076552e314bcdbbb721c08891c314 : SUCCESS in 2h 22m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/df510f07eb094514b4c8bf61e3fe4b6d : SUCCESS in 17m 07s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ad44e6986ec04f018cc4d484672b7e87 : SUCCESS in 19m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31e20987cb6a4d689c8346efd156d412 : SUCCESS in 13m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7da63cb97f8947ffaa133d9a0049401e : SUCCESS in 18m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/05ba3a6c40264c619bae8c0e42696c60 : SUCCESS in 40m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/37b1e3b854c845a6baf942e33d5e292e : SUCCESS in 38m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55d35f1fd28847c09a5b863e921f9e04 : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3c5761dfd24a47a0b507e828c6eae734 : SUCCESS in 34m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5636eb303ccc4dcf9dc074598b80c34e : SUCCESS in 13m 16s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/fe22003073354082a2bd44086e200870 : SUCCESS in 13m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f73c06fb3214588924dc0fb9aed4a70 : SUCCESS in 11m 45s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/69035d9e7dca4a148f08e4a3d65e36ad : SUCCESS in 12m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2e912926628345ad988e6284c37261b6 : SUCCESS in 12m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bb6c878080e4016814641382f490651 : SUCCESS in 12m 11s","accounts_in_message":[],"_revision_number":7},{"id":"8f3d854e1ad0546e307de4975a19b9688f591d82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 15:39:53.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"72699d86c3db6d597856f1d1e5906b32fa1d017c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 18:07:31.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6709694c25244702b4d91ed0c053e61f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c029464a154e33a27d4b944c0cdf37 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7864e4a6d494ac9bdf6d93fb44c81e0 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9eb0c3c82f884443b45dfbcc0d302b98 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6756127ecda04796b571984087403067 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5efb480f6f54f5fa0fb1637b2486fb1 : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb88f14ffba54d80b3ff403b96a5b86f : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b578ebd2a94420793263f64e486b210 : SUCCESS in 10m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d0511314770a436a99d4f432c8a92d8e : SUCCESS in 10m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9dc9ccc40ad84b7789b97b2b5458191c : SUCCESS in 56m 37s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/66c01295a9e6429ca33b22f686571351 : SUCCESS in 1h 02m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d605ac70a7db4bc3923fc12633738a63 : SUCCESS in 56m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/567a20b4df7f4ae69747dda63a5fd269 : SUCCESS in 54m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6abe9dbdf63d4d0db15d3320cf185e0d : SUCCESS in 1h 11m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6860adb3f16f47e1923ca87321bdeb70 : SUCCESS in 17m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e3e1f4b40c440b2aa736a2e38379d10 : SUCCESS in 15m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7142fcfa62ce4c4dbea2815b2e104dc8 : SUCCESS in 2h 11m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef9bff793db0435b9c900090bec74b56 : SUCCESS in 1h 56m 15s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b96a6012542a436b863f4da103564597 : SUCCESS in 2h 25m 12s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da75a087bdbb4d44b4bf8c29e4dbdcde : SUCCESS in 2h 14m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4d0759dacf0f43db83f6ac0b8910d0c0 : SUCCESS in 16m 06s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/c1c5262f07da43ac9f83bea3b1efd504 : SUCCESS in 16m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b15cc4289b540d29b868a611b0a5db9 : SUCCESS in 10m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/726b00b6c6214712891bc6a4190235df : SUCCESS in 12m 34s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a5fcc6e060da424ca15e4da4edf05ddf : SUCCESS in 30m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cd02671bed44b99b3d948f818686a98 : SUCCESS in 33m 55s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e14dc1a42c084a1cb455edde5c3a77ef : SUCCESS in 33m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6e9d69edcfb84b06ab4fdf0f8266c803 : SUCCESS in 13m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51de3bd0ad5241368bdb2e9bfd9f5258 : SUCCESS in 11m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/908fc57715a546bc932ccebe7b164d07 : SUCCESS in 11m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9356e5f664e42e38a33301687d73d5d : SUCCESS in 10m 39s","accounts_in_message":[],"_revision_number":7},{"id":"7b5342cc6b658cd50cf867eb223415a229cb7727","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 18:07:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"ce252cbc3c42ca1d86cfc96c4d0dbc93c31b8fd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 18:08:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab8875d2bda1484bb1732b130679063d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e50ee48706540e3b087751de6f4311d : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79d827e7182b4fa78b8d7f1f6d498905 : SUCCESS in 31s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"81affd272f65624e04ffcbe812437087cb5255fb","revisions":{"dcd3e4f07619ee5920ffb0b2db55bde1180891fe":{"kind":"REWORK","_number":1,"created":"2025-04-22 18:07:53.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/63/947863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/947863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/947863/1"}}},"commit":{"parents":[{"commit":"e035946abad20c44c12f94c1aac9a1de32f361b7","subject":"Merge \"Bump up Ansible supported versions to 11.x/12.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e035946abad20c44c12f94c1aac9a1de32f361b7"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-22 17:56:04.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-22 18:07:09.000000000","tz":180},"subject":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation","message":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation\n\nUse \u0027pip_upper_constraints_file\u0027 for \u0027openstacksdk\u0027 installation\ninstead of the hard-coded value in the inventory. This is very useful\nto leverage the need to override the\n\u0027openstacksdk_upper_constraint_file\u0027, which is not described anywhere\nin the \u0027etc/kayobe\u0027 nor documentation.\n\nTrifialFix\n\nChange-Id: I8ea60974e2ace6876514271be4156e6c889d8776\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dcd3e4f07619ee5920ffb0b2db55bde1180891fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dcd3e4f07619ee5920ffb0b2db55bde1180891fe"}]},"branch":"refs/heads/master"},"402364d495ec697616c3909ed7caff4f86c1809c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-04-23 07:44:03.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/63/947863/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/947863/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/947863/2"}}},"commit":{"parents":[{"commit":"e035946abad20c44c12f94c1aac9a1de32f361b7","subject":"Merge \"Bump up Ansible supported versions to 11.x/12.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e035946abad20c44c12f94c1aac9a1de32f361b7"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-22 17:56:04.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-23 07:44:03.000000000","tz":0},"subject":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation","message":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation\n\nUse \u0027pip_upper_constraints_file\u0027 for \u0027openstacksdk\u0027 installation\ninstead of the hard-coded value in the inventory.\n\nTrifialFix\n\nChange-Id: I8ea60974e2ace6876514271be4156e6c889d8776\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/402364d495ec697616c3909ed7caff4f86c1809c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/402364d495ec697616c3909ed7caff4f86c1809c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"580437dfbf938594ef1ddd87bce5ce30cc918724":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-04-23 09:09:17.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/63/947863/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/947863/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/947863/3"}}},"commit":{"parents":[{"commit":"e035946abad20c44c12f94c1aac9a1de32f361b7","subject":"Merge \"Bump up Ansible supported versions to 11.x/12.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e035946abad20c44c12f94c1aac9a1de32f361b7"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-22 17:56:04.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-23 09:09:16.000000000","tz":0},"subject":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation","message":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation\n\nUse \u0027pip_upper_constraints_file\u0027 for \u0027openstacksdk\u0027 installation\ninstead of the hard-coded value in the inventory.\n\nTrivialFix\n\nChange-Id: I8ea60974e2ace6876514271be4156e6c889d8776\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/580437dfbf938594ef1ddd87bce5ce30cc918724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/580437dfbf938594ef1ddd87bce5ce30cc918724"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f31549da7dec0e457f2b622e07177341d0b81260":{"kind":"REWORK","_number":4,"created":"2025-05-13 14:36:41.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/63/947863/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/947863/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/947863/4"}}},"commit":{"parents":[{"commit":"c2ce28256c22a0187a0409c2e390a776dce29858","subject":"Rename docker group to container-engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2ce28256c22a0187a0409c2e390a776dce29858"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-22 17:56:04.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-13 14:35:31.000000000","tz":180},"subject":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation","message":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation\n\nUse \u0027pip_upper_constraints_file\u0027 for \u0027openstacksdk\u0027 installation\ninstead of the hard-coded value in the inventory.\n\nTrivialFix\n\nChange-Id: I8ea60974e2ace6876514271be4156e6c889d8776\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f31549da7dec0e457f2b622e07177341d0b81260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f31549da7dec0e457f2b622e07177341d0b81260"}]},"branch":"refs/heads/master"},"c54a2bdabb465f9b56354f27fbef657ca13a5b24":{"kind":"REWORK","_number":5,"created":"2025-05-19 10:51:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/947863/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/947863/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/947863/5"}}},"commit":{"parents":[{"commit":"c2ce28256c22a0187a0409c2e390a776dce29858","subject":"Rename docker group to container-engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2ce28256c22a0187a0409c2e390a776dce29858"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-22 17:56:04.000000000","tz":180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-19 10:51:35.000000000","tz":120},"subject":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation","message":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation\n\nUse \u0027pip_upper_constraints_file\u0027 for \u0027openstacksdk\u0027 installation\ninstead of the hard-coded value in the inventory.\n\nTrivialFix\n\nChange-Id: I8ea60974e2ace6876514271be4156e6c889d8776\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c54a2bdabb465f9b56354f27fbef657ca13a5b24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c54a2bdabb465f9b56354f27fbef657ca13a5b24"}]},"branch":"refs/heads/master"},"021c092fb94d4b92aa9c0502030266faf9aa4fec":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-05-19 10:51:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/947863/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/947863/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/947863/6"}}},"commit":{"parents":[{"commit":"8530480926aaf91813ce1126a73af69af04f012a","subject":"Merge \"CI: Improve readabiity of ansible output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8530480926aaf91813ce1126a73af69af04f012a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-22 17:56:04.000000000","tz":180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-19 10:51:52.000000000","tz":120},"subject":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation","message":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation\n\nUse \u0027pip_upper_constraints_file\u0027 for \u0027openstacksdk\u0027 installation\ninstead of the hard-coded value in the inventory.\n\nTrivialFix\n\nChange-Id: I8ea60974e2ace6876514271be4156e6c889d8776\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/021c092fb94d4b92aa9c0502030266faf9aa4fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/021c092fb94d4b92aa9c0502030266faf9aa4fec"}]},"branch":"refs/heads/master"},"81affd272f65624e04ffcbe812437087cb5255fb":{"kind":"REWORK","_number":7,"created":"2025-05-19 13:00:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/947863/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/947863/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/947863/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/947863/7"}}},"commit":{"parents":[{"commit":"8530480926aaf91813ce1126a73af69af04f012a","subject":"Merge \"CI: Improve readabiity of ansible output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8530480926aaf91813ce1126a73af69af04f012a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-22 17:56:04.000000000","tz":180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-19 12:59:59.000000000","tz":120},"subject":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation","message":"Use \u0027pip_upper_constraints_file\u0027 for openstacksdk installation\n\nUse \u0027pip_upper_constraints_file\u0027 for \u0027openstacksdk\u0027 installation\ninstead of the hard-coded value in the inventory.\n\nChange-Id: I8ea60974e2ace6876514271be4156e6c889d8776\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81affd272f65624e04ffcbe812437087cb5255fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81affd272f65624e04ffcbe812437087cb5255fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
