)]}'
{"id":"openstack%2Fkayobe~948452","triplet_id":"openstack%2Fkayobe~master~Icdc3808a5f9d43007a83a0854d07d4a486857917","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-04-29 16:14:24.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Icdc3808a5f9d43007a83a0854d07d4a486857917","subject":"Use uefi instead of bios for Tenks","status":"NEW","created":"2025-04-29 15:06:52.000000000","updated":"2026-02-19 15:28:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"698ec725de68e07ae41aa0256a8670b5a258be24","_number":948452,"virtual_id_number":948452,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-29 17:34:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"reviewer_updates":[{"updated":"2025-04-29 16:00:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-04-29 17:34:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-19 15:28:54.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"CC"}],"messages":[{"id":"984aa110743b1e7e6874fd8d5e59e57782323cd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-04-29 15:06:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99bb2847b13b9e384d2e1aa237eec76ac43c653e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-04-29 16:14:24.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"8e5340f67d2902318a42affcaa808e65d909033b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 17:34:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bf265ba9b924056acfb89bd72ab4e69\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4db825f185ed454d8f304a02aa60cb37 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65f48dd89ee0484aabf4625acadf9c3b : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cee052a43504450af25d93ee3c7c79d : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d6f4129b74348a691e378ec2872fcce : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7edc3d21bbe94967a9c98ead422a5f0f : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba180164f9224cedad334b04196d45bd : SUCCESS in 7m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a458afd9e30548999a5eb3a98ae086e5 : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/04a963114efc49ae9eb325ceca52a8f1 : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d888932d4c6f4c7f948f2f5ae14155ef : SUCCESS in 8m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/41a1ed1f83854a3e97bd5e62c39077e7 : SUCCESS in 55m 51s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/d32bd5c393de4e599f752fde2857e0c7 : SUCCESS in 58m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d31aebe01c24e19b0df2a2e24937d35 : SUCCESS in 55m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4e29baf79001432cb418e729ae8e619c : SUCCESS in 53m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/93bdc4e8a1e6467599bc90b9e232c182 : SUCCESS in 1h 10m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fe0d29ab94254bcea1c799bf13c35e85 : SUCCESS in 15m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/400e6309badf4b9ab8ae42d33145fbb6 : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8c20737312fa4244aa8ac594ca0c5718 : SUCCESS in 2h 20m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ab8184ae2f94316ae0b64c1a437f2af : SUCCESS in 2h 03m 48s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1f902c523c3248c68fe2e32cd13c4320 : SUCCESS in 2h 14m 30s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c076d04244594416818a631e8be00d4a : SUCCESS in 2h 08m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/80caa0ad35484c63add53556670fd5ba : SUCCESS in 15m 12s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/a5dfc85f192a4e8b842e48fdd128406f : SUCCESS in 17m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32c634887aaa4589a16de4ab281526f9 : SUCCESS in 14m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/20f258f41599475196a341cb5b04e263 : SUCCESS in 13m 44s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/121b6ab36d724da39085a392af8459dc : SUCCESS in 34m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/537b66a5ace34ef090a24d14fb3a1320 : SUCCESS in 33m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35bcd59a0f1a4b988fab8ed2c08bbbc7 : SUCCESS in 30m 29s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/06543b271689494fbfbc5926f963ebe3 : SUCCESS in 31m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4b53859377f141ac83ac104febc27a6c : SUCCESS in 12m 45s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/a11c70eeba21443a87bc91c30c357011 : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35ba745cc6ab48e9a92a161f795d273e : SUCCESS in 10m 07s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/3f7b003f43d74b7ba07fc4f3f0968665 : SUCCESS in 10m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/326b41c502d04439947f1f59612f9f9d : SUCCESS in 11m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/142a932292674d448c6902c3c32e3e0e : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd2aa3973b679793b447b32bfd77d39f7d069b6d","revisions":{"dd2aa3973b679793b447b32bfd77d39f7d069b6d":{"kind":"REWORK","_number":1,"created":"2025-04-29 15:06:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/948452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/948452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/948452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/948452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/948452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/948452/1"}}},"commit":{"parents":[{"commit":"6567440e46610fd0420c40e0a259258fda19d82f","subject":"Merge \"CI: Enable SLURP upgrade job for Ubuntu Noble\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6567440e46610fd0420c40e0a259258fda19d82f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-04-29 15:05:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-04-29 15:05:50.000000000","tz":120},"subject":"Use uefi instead of bios for Tenks","message":"Use uefi instead of bios for Tenks\n\nTenks virtual machines fail to boot from a Rocky Linux 9.5 image when\ndefault_boot_mode is set to bios. Switch to the default value of Tenks,\nwhich is uefi [1].\n\n[1] https://review.opendev.org/c/openstack/tenks/+/827553\n\nChange-Id: Icdc3808a5f9d43007a83a0854d07d4a486857917\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd2aa3973b679793b447b32bfd77d39f7d069b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd2aa3973b679793b447b32bfd77d39f7d069b6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
