)]}'
{"id":"openstack%2Fkayobe~949249","triplet_id":"openstack%2Fkayobe~master~I515d720d79e078e781f12748ce5cfa33573a2d33","project":"openstack/kayobe","branch":"master","topic":"support-list-tasks-ka","attention_set":{},"removed_from_attention_set":{"35010":{"account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"last_update":"2025-05-13 16:32:54.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-09 18:48:47.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-13 16:32:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I515d720d79e078e781f12748ce5cfa33573a2d33","subject":"Add support for `--list-tasks` against `kolla-ansible`","status":"MERGED","created":"2025-05-08 18:53:13.000000000","updated":"2025-05-13 16:33:51.000000000","submitted":"2025-05-13 16:32:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949249-support-list-tasks-ka","meta_rev_id":"684abb86f61a419928c0b2f1ddf99e215686cb9e","_number":949249,"virtual_id_number":949249,"owner":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-13 16:32:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-05-13 14:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-05-09 18:48:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-09 13:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-05-13 14:09:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 19:08:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-08 21:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-09 09:50:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-05-09 13:39:56.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-05-13 14:09:05.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"13a286ead60426ad4bfa749834ffe4174a7ed71a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-05-08 18:53:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7e5c775fadd5df839372458c1496d5e38b967c1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-05-08 18:54:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78bf56fb002daf7a9417bcf6b99f67db9ed0cb1e","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-05-08 18:55:40.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"a8f055500df4e9558b46fa8ab08e41b0d4af4549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 21:25:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d583f4e42174de6adcbabe79a85917e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad12294dd35643dbb6d98c8170a4e503 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/635d1cd9d1244a10a67ebc804084470e : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bef2691044ec496ba27ea7cb1c6acb0d : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d52feb7ab7841c1aa4b41b188cbaaf6 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c3fb3eff5474418b1a16398509bb4b9 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20d810e00404a65b29df7319c767dd2 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d01f533c29d44819ffe88bfb6487207 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/31245c45cdcf4f298d83aa2e2e4713d0 : SUCCESS in 3m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c761c06a3864c32983ae0987fa409b1 : SUCCESS in 5m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fea25773c7564641817d2a2ce080c54d : SUCCESS in 8m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/844649abe9ec47f48a99171fb00962cf : SUCCESS in 57m 58s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/4d91fccec130437a8b95987502e2bd7f : SUCCESS in 58m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5aa0022eb4f43f59314d2d1532245ec : SUCCESS in 54m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1b07680b33ef41e4a374b64ac8dbb46b : SUCCESS in 37m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9f6cbdb052b140de91ef101d1f7d8bbc : SUCCESS in 59m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4a3a1c9bf6a242fb8c085a6578439a1a : SUCCESS in 17m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/632e5bd667cb4d2abf08ee0e164edc11 : SUCCESS in 17m 20s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/de1901ae9570400a86f0b68556122666 : SUCCESS in 2h 28m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dd5f8c03dd0430b9ceb0192804a5f38 : SUCCESS in 1h 48m 43s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/20a6c051dfc143f48d8d415d3d455e48 : SUCCESS in 2h 15m 07s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f274c72c5fa4c83bd2cd0c16bb220e7 : SUCCESS in 1h 47m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7f5efc421ece438dbc599549f0e7cf6f : SUCCESS in 20m 20s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/2232361c3a8f4237a8b964d819378dc3 : SUCCESS in 15m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/236dd6d6ffbf4bf6bd89551245edbf17 : SUCCESS in 16m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b75254d7491b41ab87120dd556568d3c : SUCCESS in 14m 21s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/11ecc51eebcc4a4fa339ccb905007ae2 : SUCCESS in 36m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1fc45b53fb3e4ac286b8c08a5b07dfdd : SUCCESS in 32m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d10e5b65cfe416c9d6c44383a30de26 : SUCCESS in 31m 27s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0a58931e1098469c870d82b8cfe3e5cb : SUCCESS in 33m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e468cbdfe72240789a60b7f60da199df : SUCCESS in 12m 24s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e2f904cb45ca429e941db0b90e3195c4 : SUCCESS in 10m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/715e6ae0c0a841ac9224e1bb32b3f6b3 : SUCCESS in 9m 21s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/60e369d5a6b143808e4ba770e32e2a66 : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a39da4b90d2346938b8fa4ec4afca45f : SUCCESS in 7m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b47ce6725c046f886f03db8ae807261 : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":2},{"id":"dbc812b6f75b9831a8938fcd986f31fd009abb86","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-09 13:39:56.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0699557f93c69055917869829bd10ca6947aced7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-09 18:48:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9008f8998b9e19de1fb1f962845a3968a41ebdae","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-13 14:09:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"21b71d69a778bb600e375d4b7b581ff1b9367424","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 14:09:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c9bb7c0c7b1b85bc36b88602dd4885d95984d1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 16:32:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1846c03408f843c98b5afdb5323774a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d325c8437f0e4e10bddf2477fb360e5c : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f44e420e40ce41f0b4616ea96fa11ad5 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b69c87c5f734453d8844ff5cf1ee6fd5 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2198d5fe2284a0885d7386f0cba6c43 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7316f7c3242b495681252341866ed483 : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fbe08a0dfbc74687b5c2d4562fa93a9f : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5232c7a429ac4fd3b5816c0b4e4b60cf : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fbf71f6c88764599a4a9c4077a207f58 : SUCCESS in 6m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d07210e32f9e4bce9ce326699c20469f : SUCCESS in 56m 03s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f50593cd56ac447f91dc101994a690de : SUCCESS in 55m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28ad9a0fb59347ee8232cba9344889ff : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a047f7608bc742ee913c08fa4c5a1200 : SUCCESS in 57m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3295445b978a4b7c9089330cab74a7d4 : SUCCESS in 1h 02m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/87a15ed11ab245d4a0d65ca9393fe234 : SUCCESS in 11m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47ce0d9a403f4daf853ca8dc5d993421 : SUCCESS in 14m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/76c9f50981274e67a98a7a0f02e95b8e : SUCCESS in 1h 54m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2f5ef494fc1415294d0e3f21b0be363 : SUCCESS in 2h 23m 19s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/08d32a752ddd4c03ae4fd4ab31c922e9 : SUCCESS in 2h 07m 18s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2908d1e8b06440caeea208fb8157dc2 : SUCCESS in 1h 06m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9f23ae9682c0482481b80191f203a88d : SUCCESS in 14m 17s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/32a8b2ac4d34418295f49c1a7dc81ee3 : SUCCESS in 17m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/351356a6cb6c49efb0d5840f5a11b0a9 : SUCCESS in 14m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3105da3dc4254949b3b2f6921f0df800 : SUCCESS in 7m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c732197263b34b879d0cf0f7bc12c980 : SUCCESS in 27m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5655b6b246d14e348aec52d9ea466d3c : SUCCESS in 37m 39s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/dab7a19e9d634637810f021cefebed47 : SUCCESS in 30m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1c4aeafbb65348d0b24521fcdd410cc2 : SUCCESS in 14m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b73eb17e45654c72917788fc88e2f304 : SUCCESS in 10m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3b2737ecc662466c996798711807cf6d : SUCCESS in 11m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5df689f0a2d5456c92eda96c1509f660 : SUCCESS in 13m 43s","accounts_in_message":[],"_revision_number":2},{"id":"1b95e4f2364a0d040af4b3c556778344d0333ec8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 16:32:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"684abb86f61a419928c0b2f1ddf99e215686cb9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 16:33:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f9ebac723524c36b63946ab06cf2753\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2445f46caa0425a9d9290b036f7cc54 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76c4425afb85401cad223e1a6ccf25c3 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"28667735bdb97c8ce4018d721e73c58ed0143e22","revisions":{"86c98024bf75d1a24b00374de6a95f7f20adce69":{"kind":"REWORK","_number":1,"created":"2025-05-08 18:53:13.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/49/949249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/949249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/949249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/949249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/949249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/949249/1"}}},"commit":{"parents":[{"commit":"6567440e46610fd0420c40e0a259258fda19d82f","subject":"Merge \"CI: Enable SLURP upgrade job for Ubuntu Noble\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6567440e46610fd0420c40e0a259258fda19d82f"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-05-08 18:34:01.000000000","tz":60},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-05-08 18:46:16.000000000","tz":60},"subject":"Add support for `--list-tasks` against `kolla-ansible`","message":"Add support for `--list-tasks` against `kolla-ansible`\n\nSupport the use of `--list-tasks` when running `Kayobe` commands that\nuse `kolla-ansible` such `kayobe overcloud service deploy --list-tasks`.\n\nChange-Id: I515d720d79e078e781f12748ce5cfa33573a2d33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86c98024bf75d1a24b00374de6a95f7f20adce69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86c98024bf75d1a24b00374de6a95f7f20adce69"}]},"branch":"refs/heads/master"},"28667735bdb97c8ce4018d721e73c58ed0143e22":{"kind":"REWORK","_number":2,"created":"2025-05-08 18:54:50.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/49/949249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/949249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/949249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/949249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/949249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/949249/2"}}},"commit":{"parents":[{"commit":"6567440e46610fd0420c40e0a259258fda19d82f","subject":"Merge \"CI: Enable SLURP upgrade job for Ubuntu Noble\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6567440e46610fd0420c40e0a259258fda19d82f"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-05-08 18:34:01.000000000","tz":60},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-05-08 18:54:43.000000000","tz":60},"subject":"Add support for `--list-tasks` against `kolla-ansible`","message":"Add support for `--list-tasks` against `kolla-ansible`\n\nSupport the use of `--list-tasks` when running `Kayobe` commands that\nuse `kolla-ansible` such `kayobe overcloud service deploy --list-tasks`.\n\nChange-Id: I515d720d79e078e781f12748ce5cfa33573a2d33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28667735bdb97c8ce4018d721e73c58ed0143e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28667735bdb97c8ce4018d721e73c58ed0143e22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
