)]}'
{"id":"openstack%2Fkayobe~949322","triplet_id":"openstack%2Fkayobe~master~I41ccedbd98b0088b92d3eb8671955423815d599f","project":"openstack/kayobe","branch":"master","topic":"fix-ironic-journalctl","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-09 22:09:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I41ccedbd98b0088b92d3eb8671955423815d599f","subject":"CI: Fix Bifrost log collection","status":"MERGED","created":"2025-05-09 10:18:44.000000000","updated":"2025-05-09 22:10:11.000000000","submitted":"2025-05-09 22:09:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949322-fix-ironic-journalctl","meta_rev_id":"244948e4caeaae60eca45c2b03427d1ac8fca131","_number":949322,"virtual_id_number":949322,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-09 22:09:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-05-09 13:02:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-05-09 18:46:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-09 12:42:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-05-09 13:02:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-09 10:41:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-05-09 10:41:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-05-09 10:41:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-05-09 10:41:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-05-09 10:41:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-05-09 10:41:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-05-09 10:41:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-05-09 12:37:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-09 18:46:49.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"7f796779095224d6e1d2bddaeef17d940e95d8ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-09 10:18:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5337e24516331e569d3a2e0ba0c86cdb8af4350b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-09 10:41:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9f353bc286c723c9051c77b1d9f03ec0edcd113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 12:37:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41c3612988e74d29b06bd9dfdf1f5321\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04cecb9cf3da4f51907501c59abf449e : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/587acc568b2345d790cc177382a5aec7 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a39ac05cb5c9440a9e2f079aacd4426c : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5de1f6fc93a469e8907bb2a1bdba85d : SUCCESS in 3m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cda39a00e6a4231a676c3c837a680c5 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48656dbb4d9b40128072b35509965b7e : SUCCESS in 9m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/60782f3475014da89973b06133b88f06 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/33dcce49cd1644368a4287980b7902ec : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a8d71be5fc634c85adf130340479c080 : SUCCESS in 9m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/865d45a79ecb4eb69effe4c25d2d16ed : SUCCESS in 55m 39s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/3932278fd8b44d00b1cfc6ac105bd3f3 : SUCCESS in 55m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e69b60b77a854461b963f1fb0703904f : SUCCESS in 50m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/429832b3318644b4aa613917c89965e3 : SUCCESS in 29m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b63a836a4ad4404787913bebe9347353 : SUCCESS in 1h 08m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b8210cacc44b40b7bf88bdae510af402 : SUCCESS in 13m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b96fa2664ccb462795a6521a1f24f45f : SUCCESS in 15m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/250e2fbed35f4584b0f7795aed077696 : SUCCESS in 1h 44m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b087009cc109480283b2840a80d85b72 : SUCCESS in 2h 18m 29s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3d3d2e67a57643bf83004e424eb46a8f : SUCCESS in 1h 48m 38s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7af97c0604c045f89c728c4a80ff8773 : SUCCESS in 2h 15m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f0a52980d3154f488ae48e6b06ae5f8b : SUCCESS in 15m 41s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6bd49d91375142a7a3d0c6468ba88422 : SUCCESS in 10m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/120226b95be14a11b51c0ebb51f24872 : SUCCESS in 16m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fa84ff19c56b4717b3ae353ed2d380b7 : SUCCESS in 12m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e6a69fa26bc444a6b5fe78f0e4936535 : SUCCESS in 18m 02s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7ec9f7dc4bf046eb82fc03af832c2e1b : SUCCESS in 27m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e68dafa98ac84dda8a8e439317aad2c9 : SUCCESS in 31m 33s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9b0cb42b47354272a3127ac677140afc : SUCCESS in 38m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/84d54ab12228404eb4e76a3d4ec11049 : SUCCESS in 10m 33s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3ad36b35c27a4d7099f841404d3b4c15 : SUCCESS in 12m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d765822ff6b74f42bf16e4db6d6653f5 : SUCCESS in 6m 50s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a041a3bd98024efda2da26b675c76324 : SUCCESS in 10m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e7b3ecce4b0d4a56b6142ebf42e11b76 : SUCCESS in 10m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5bfe1ae1024441caae6cd2c76de17eb : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0bcf6c963d6c0c58d1832e0ed563e17dcd2427e7","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-09 12:42:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9439b3b89ca502f9bcc338e6d089366508a1b33c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-09 13:02:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c2cb42f184221bb80eef9532934ba1891b769247","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 13:02:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f04bdda854acc4f3a957636486c33f8220a9e5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 16:14:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2530160dabd4767bba3a35ab13e9f75\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ee78d2bc6d84a939ef5592ee223ac24 : SUCCESS in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2e4f04e05e34eb88d132d37766d50dc : SUCCESS in 2m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37b86047436d4ed8bb8787e616e280cd : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbcb18901bdb4eb1bf18a0d12951bef3 : SUCCESS in 7m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d227d8986554eff8cd9e23caff85203 : SUCCESS in 2m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f9c28415ab74470e85d61c57c9fbc701 : SUCCESS in 7m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9eaaf7fd94b347398cb2ffee8dfe4b39 : SUCCESS in 5m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b18b56f318d549b8ada868c9e07048ae : SUCCESS in 46m 53s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7f4e7e623af749b4bf46842ba3607cf8 : SUCCESS in 31m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/561b3ebf579748a6b0125751ebdf5329 : SUCCESS in 46m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e0ccfc605c444d9bf36b5762cbe88b4 : SUCCESS in 29m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3b12dd4b421b40df940d34cd6a06c6c5 : SUCCESS in 37m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/37af9238f75e436a878d9052efb598b7 : SUCCESS in 14m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb09c4774b234c438e5fec1c2a3abe5b : SUCCESS in 12m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/94ef39cf7b004d889a0b190fb5c7f868 : SUCCESS in 1h 13m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/261e32e7e8a946448afc56966e6a101b : SUCCESS in 1h 24m 25s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/85ef14485e9a4a69b2b9c3ceb0aed8c0 : TIMED_OUT in 3h 06m 26s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6265ec4a1c4b4a36b54a6ade69e845e4 : SUCCESS in 2h 23m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c1085c2baf604d448942d67b55042d05 : SUCCESS in 12m 24s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8c41cb69473347c88684ff51aecc86c4 : SUCCESS in 13m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/205532de7d3548468bbbcbcbe1bd145c : SUCCESS in 11m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4daabf21f823496d9c6bac644343506e : SUCCESS in 13m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ca216992714542c580c98157c25fd6ce : SUCCESS in 33m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe33e226b312477ab7fe9b732105e42f : SUCCESS in 24m 21s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5c7bb926eff6465e91ca0a7ab2e4f5e7 : SUCCESS in 24m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d02121475b5746a988878d56ae259ab5 : SUCCESS in 8m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24d92eee25d14529a4ea3dfc86e51b19 : SUCCESS in 10m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/55788ba6049b4e21a7daf938da5bf024 : SUCCESS in 13m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/182e6cc309f045d98a7bb457c99d2fe1 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6f07fbb6f81be7178707fb33b3cc215685dafa77","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-09 17:15:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"538a737cb9044dd8e32c7b868b240d0608f24454","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-09 18:46:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"675be7189d3899a12c469c0a672bbc857a386561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 19:18:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a510f4afc88544e79f96c37f9a6ce15c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffb4ce43de29472bbfb7170f42187f97 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cac321ccd12c40a1a87b9a908877a8d8 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be196d98bb9d4516a3b853f5812de4aa : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa685c4a36894a6bb6c090d4fa3f10f4 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/955c6dec72554804857e60b3805a2fdc : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38fdf6afd248413686da624ef781b0c0 : SUCCESS in 8m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/162ce0090f414669b9265951062130d8 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b402978c100b4f80adf12c97eb71cf2c : SUCCESS in 9m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba4f714da555477d9a238add8afa5bfa : SUCCESS in 8m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bdc258be5b614c82a54c7f5e7336f70e : SUCCESS in 51m 39s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e66d03f02bfd4d42b5320ad4f817d520 : SUCCESS in 53m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d33f1653b2e4d71a1249dc4e6f28be7 : SUCCESS in 44m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c9a64cebef804027be9c5d39937d0f99 : SUCCESS in 53m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1ccb8c5b024f43a1ad9338748a7db169 : SUCCESS in 48m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7562684a893e4f71a9bba5347c8fb78e : SUCCESS in 16m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7c868edadf84b9898701c2e506d2af2 : SUCCESS in 12m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/49efde7a1b464f25aa55a3af2f6d4a94 : SUCCESS in 1h 57m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c4e46a16a1944109627077f87d6ab49 : SUCCESS in 1h 43m 55s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0e22ba8c5a9a43139ec85ad0559aedfb : SUCCESS in 1h 13m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfb1346958dc445db3c984a54c96de17 : SUCCESS in 1h 19m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/70197bff31da4e1a9e4ca26e9ea50edc : SUCCESS in 14m 44s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/08ecbf33a42d4835b3725a388c299413 : SUCCESS in 9m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d6a4890e8dd41469ced74f3f5846cdf : SUCCESS in 13m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/63f89e35d3e4462f909ee0b84189ead5 : SUCCESS in 11m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/18072cf1f28944fba3ebfecfd47b80c4 : SUCCESS in 31m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fef02032276740359e7cdbe9d3fb7d76 : SUCCESS in 29m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24bdaae94d9747549b0a5f67bef63307 : SUCCESS in 32m 26s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a747f1e67b6145daa9c0443e8b9a1980 : SUCCESS in 29m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4162eadba89d460ba7107226921752f3 : SUCCESS in 8m 43s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/711769f466ac48c8b157d4399761b05e : SUCCESS in 12m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cef9aeb8c1c444179f2606c1ba142cea : SUCCESS in 6m 50s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/c5d506fe999b49bfa22f1a93d29c3298 : SUCCESS in 10m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/59de331fdfc74ba0be13ee33139b9542 : SUCCESS in 13m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cad270eae854e559e95bda3029058e6 : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e95eb7bc634ee5bb1687a68829bf1ac736f1018d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 19:18:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"032bc824430da644c958d88afd9d3b5739de088b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 22:09:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3c6559a3d324834a45358acc03c4353\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07685f3970b04b42879c618f4edc74b8 : SUCCESS in 2m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31b959c35e2b4968aae90bc7092de0b9 : SUCCESS in 1m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/944f6e82d6bf429e891b6d6dc0780066 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e61c1dddb8be4c48a273d31496e6dd79 : SUCCESS in 9m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/02d6d5cecfe34fa991c470a4609a8dd4 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/63858034bfc945f2bb4f8722be9a9406 : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ebd7b476a99145be979c82e1ccbe8319 : SUCCESS in 10m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a968ddafc54b40358cbac7d0c626f3cc : SUCCESS in 59m 22s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/4148b0d0658847a6adfd20059cac4e81 : SUCCESS in 59m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad9542affaee41b9a8c46e1f79fae245 : SUCCESS in 55m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/94719561d4594dc5a933e1ebfd12970d : SUCCESS in 52m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2ce7ceb2c47245459dcbdfaa3e54c622 : SUCCESS in 1h 10m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7cc000fd0fb54b199fb2960c143db634 : SUCCESS in 20m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3da40ef7ea1496e90681e6781abd0ab : SUCCESS in 12m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3a9addc30b0743ec90edf2a800dc1432 : SUCCESS in 2h 16m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebbb166da8be4556ac9a4bc20540acbb : SUCCESS in 2h 22m 59s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7ed0ee686b8247c5be5594c11eb7863c : SUCCESS in 2h 34m 11s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7e13d99f7324ae9888ce18ad22b9c57 : SUCCESS in 2h 15m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9dc4f50d9e36423f82c1ad4a5583726d : SUCCESS in 18m 48s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ceb3adff59194bd59e8529ee6520b80b : SUCCESS in 19m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04eecf6898de4be8adf04680be26c5cd : SUCCESS in 13m 51s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d1945d8c3267449d97de4e4451b9fdf4 : SUCCESS in 8m 10s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b25d315feb674217be5916d2659dcf12 : SUCCESS in 32m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffcc0caf916a46f6bb4ba1e63cba355f : SUCCESS in 37m 14s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/966780fc4174450288341e2ef4c79197 : SUCCESS in 37m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4898e1acd21e4082a1dd5e1dcb1b3d3a : SUCCESS in 12m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4fe668018ff496986d8c07b1bd1c311 : SUCCESS in 9m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b29fa66e6bb5463dace9f38a1225c70e : SUCCESS in 13m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50ed2edc1d5a41b38345adf788549097 : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e57a1c8bfa1326262db54f14ed5713a9e3efcd39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 22:09:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"244948e4caeaae60eca45c2b03427d1ac8fca131","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 22:10:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d2bf0dd7a114799a604cd4303f754b3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e0d164e341f4e3fabaa5fbbb42abcc1 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c11c62be242029b511cc1292b3636f965c509e56","revisions":{"c11c62be242029b511cc1292b3636f965c509e56":{"kind":"REWORK","_number":1,"created":"2025-05-09 10:18:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/949322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/949322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/949322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/949322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/949322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/949322/1"}}},"commit":{"parents":[{"commit":"53ece53da627698e18f5556199ceace26c2296d6","subject":"CI: Make kayobe-seed-ubuntu-noble-podman voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53ece53da627698e18f5556199ceace26c2296d6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-09 10:18:06.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-09 10:18:06.000000000","tz":120},"subject":"CI: Fix Bifrost log collection","message":"CI: Fix Bifrost log collection\n\nBifrost changed to use a combined Ironic service instead of separate API\nand conductor services in the Yoga release [1]. Update copy_bifrost_logs\nto get logs from the combined service instead.\n\nAlso remove log collection for RabbitMQ which was disabled in Train [2].\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/820390\n[2] https://review.opendev.org/c/openstack/bifrost/+/645093\n\nChange-Id: I41ccedbd98b0088b92d3eb8671955423815d599f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c11c62be242029b511cc1292b3636f965c509e56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c11c62be242029b511cc1292b3636f965c509e56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
