)]}'
{"id":"openstack%2Fkayobe~949489","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~I41ccedbd98b0088b92d3eb8671955423815d599f","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-05-13 13:49:30.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-13 15:03:24.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-05-12 12:00:29.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I41ccedbd98b0088b92d3eb8671955423815d599f","subject":"CI: Fix Bifrost log collection","status":"MERGED","created":"2025-05-12 11:10:07.000000000","updated":"2025-05-13 15:04:22.000000000","submitted":"2025-05-13 15:03:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949489","cherry_pick_of_change":949399,"cherry_pick_of_patch_set":1,"meta_rev_id":"9facafa6813efaf49b526cd3006db64d5fa5c2c5","_number":949489,"virtual_id_number":949489,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-13 15:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-05-13 13:49:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-05-12 16:17:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-12 12:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-05-13 13:49:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-12 11:10:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-05-12 11:10:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-05-12 14:20:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-12 16:17:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"335fc64e270d04b1bef3dcdd179e4f301f9b6b31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 11:10:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"264cee1799e3c99b147e9d2f11bdc429afa3fac8","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-12 12:00:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"35e556e41d15dec9ed57e55537cae24846b2aa1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 14:20:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aae25a906234443f9571d671d9863972\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a550855a6564c1eb2e7335f5785a3cc : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221884a2e12045898c4b370970ee3eac : SUCCESS in 6m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/06c296dbde0f4828a3dab0649b478bba : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d85f859de15b403fbe16359fdad2ad9f : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f3cf139c0b414364b596ac52e53c3af4 : SUCCESS in 10m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3f24691cf7914b63b4f85509cb1d1b40 : SUCCESS in 50m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6e34d23188664324abf5aa6c423ae752 : SUCCESS in 1h 02m 02s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/193f7fd160f5479c80e72b3b87b1bf1a : SUCCESS in 58m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13d242aed0f84fd89dfc12b151891bfe : SUCCESS in 57m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/03c9a94829684f5f8579f425e48725c0 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6f5d3cc929ba4fff8851f09be9884b8f : SUCCESS in 19m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/be9a00ac75db4eb49c61af08146f923a : SUCCESS in 16m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/325e74527f5e475391cac14a0287144d : SUCCESS in 10m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0842d46fd1d4f8a8fe685647882f00b : SUCCESS in 9m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4231bdc5be8240d28d4c31a4faccd944 : SUCCESS in 1h 05m 09s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da6511be49024687a98fc1ad383974d6 : SUCCESS in 1h 30m 24s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a9df9690f7d24ed8bab4d63abcb9b275 : TIMED_OUT in 3h 03m 47s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b285d32d5f44b90b7732e4898a51c16 : SUCCESS in 1h 55m 15s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fce1ae2aaba24cf3ac082518d784ba1c : SUCCESS in 20m 07s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b67772c76fa4663b7c9644347bb40cb : SUCCESS in 19m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a101d8bfb07459d8a5a404bc481b18e : SUCCESS in 18m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6f90907d46a849ac81cc8b58a56047d6 : SUCCESS in 40m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2a197654b6194ca48af32b35bd9f6d00 : SUCCESS in 24m 58s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3149073830d74eb7be70512496fcafa4 : FAILURE in 14m 12s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c499ba9677e3486dbceb1d7e7a787eac : SUCCESS in 26m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e11061aad43840438b5f382103dd4d66 : SUCCESS in 23m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/afff5d467a7047c0aed0f10556c7db34 : SUCCESS in 17m 14s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/977232faabba49f4812d06312fc9e3be : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01196b8460124507a6eb363afdb05724 : SUCCESS in 13m 26s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/7e3be5bf349b491bb57ee389389d946f : SUCCESS in 9m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c09b4ebe415f484fbf960f73b2b520ba : SUCCESS in 14m 39s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/99e5f393424a4137aa87c5bf16afd979 : SUCCESS in 11m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/54a8d25f78f249cf9867d25f68d11f9b : SUCCESS in 9m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72433e901fbc410fa844e29c0b8abe6d : SUCCESS in 13m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e325dd6649c2425aa81a193747eb445a : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f3ed5d3d88c48df9b8d5da63658c694 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48e0efa569534d06b15be8070f02e078 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c68f3cd2e52437ba4f668524298cae7 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/446e27b55af64d808f6d3e956259e181 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e13c01b288d2c9127ae45207470c61903afae9b5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 14:54:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a24576486c797cd3b228c2a4a501fa566280f584","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-12 16:17:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f386afb50db3e466d04cd5b086638953ef5b6564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 17:02:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4b4f7c1fc5f4e9fbcfb654b080de2c6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b53f39899b58475f85953b9a07b110a8 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5069c26b4e5d419f9a72e6cb1689d2d0 : SUCCESS in 7m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/561cd403cf794f86a991ac509ec1f895 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b775d484a2944d39e51175e9bfb3ba4 : SUCCESS in 14m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d69e47a7e39545e0a7fdf3b981f55a30 : SUCCESS in 8m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9fbca344f53c412ea9378a5a1e82cb90 : SUCCESS in 59m 57s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7de455e5c28c4135832e1032588bba19 : SUCCESS in 1h 06m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12518daf9bc146008bf2a0c62b01c2d5 : SUCCESS in 58m 59s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/860b652f1fbe4e0ba564529185953609 : FAILURE in 14m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7b3d51ff81c14efbb536644cb293dd7f : SUCCESS in 1h 07m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a77e8eee55504689bbcde037c7efb6f9 : SUCCESS in 17m 22s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/98f5120cffb84e7b932826e7ee5eb4d0 : SUCCESS in 18m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9631ea399c34142bebd2ba09a4b7bbc : SUCCESS in 13m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23b93ff2a60644cc82e40c5fa42abca6 : SUCCESS in 14m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9b4eafae677a4ebfa25aab041a724d00 : SUCCESS in 1h 39m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbdba497037f415383bc62ebc618b23b : SUCCESS in 1h 33m 51s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3d2cc8d2d23a46fba73e89ae1a184cb8 : SUCCESS in 2h 07m 09s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a771abfec2894d71af83a4549feb3832 : SUCCESS in 1h 55m 24s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7c2043ed4ace4aa68f12b9dcae02931f : SUCCESS in 19m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82a72cfb5ef24316b4c90f4788c8a427 : SUCCESS in 15m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ba68b4ba62b43dbaa4e09a3274b1d0a : SUCCESS in 14m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1872d3d9ba844ea1bd6935c097e6761f : SUCCESS in 44m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8ebf07ee325b45be89a5fe8612c66e5b : SUCCESS in 30m 07s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87a7a6ec4a144c908e0d758adeeb5bfe : SUCCESS in 27m 53s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b475c7c7df074849856ef62d8dd0532c : SUCCESS in 29m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6678206ab474ceb8269855ed1f33d7e : SUCCESS in 27m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/faf784d5d69f4c07a44098288d0f73a6 : SUCCESS in 12m 21s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/cb8bfb07a271439d92545289a01bcf12 : SUCCESS in 11m 53s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6795981ad16b4d4a895fa610263fb14d : SUCCESS in 8m 53s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/21ee7d43d7274f88b89f74c3f51fa66f : SUCCESS in 18m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce4d046c7af6466a922c54b66820a5a3 : SUCCESS in 11m 06s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/19f53fdbc3504f0a88ddaadad7948488 : SUCCESS in 13m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a6319d7274f24b1eb22b8e6f8aa0c667 : SUCCESS in 16m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e79b6b3a9e4040709d8643eb26f0c417 : SUCCESS in 13m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09d2cf493fd94d9a987a49da9d9e7086 : SUCCESS in 14m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff0224be9fcd4e9db62e6329a92d9b22 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f4558292e6e4697a8fb7db33a5ef51e : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4c5ef351ee24cca8f954e955899ce13 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6bae389892c4ccb85c63cad9c52dc77 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"737cefa7d7f6b071536227a8110be3b65138fd08","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 17:04:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29f8bd5b4c6c61720fc62f86e077ba347639ffd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 19:21:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0163a1a137834c989ff7dad5e6f70ad8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af324c5b164144f384257854dae489ad : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/253b77742aae404fafdf5c49cff7c7d7 : SUCCESS in 5m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5217298a7394464b9ba0d0938cd1905b : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/44eabb731e3e4fdf95a0299836d4974a : SUCCESS in 5m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3b1b95a9996742188376d6b4c0113c88 : SUCCESS in 10m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6e514e77492a4e9b94e3af81c273cf5c : SUCCESS in 56m 51s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8264bc4f695a47f98dc97a76c6a52571 : SUCCESS in 44m 09s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb39af2b86904e0bb570a626a964c0a7 : SUCCESS in 52m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f12a1b78168b4b35888cbae617ec3aac : SUCCESS in 57m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e0fc4cc72430434cbe25b7e240ce7650 : SUCCESS in 1h 09m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/39e2bb905eb54597a4c81b6eed3cc698 : SUCCESS in 15m 30s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/eca1644566334bd1b11c191016011212 : SUCCESS in 20m 29s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d625153168864bf8b218dc2e28a311ce : SUCCESS in 10m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/496a9a82d9e4428b931d7836c77e70a9 : SUCCESS in 14m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/05a5a05154c740349eb2409957bcb62c : SUCCESS in 1h 35m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17bf3c7b7e0646b686dd9ff5e43c9e70 : SUCCESS in 1h 43m 16s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/961bf390d0b246c8ad5110add3ff78ba : SUCCESS in 1h 04m 19s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/560ad06a3393487a9fe72f93dca19826 : SUCCESS in 2h 01m 01s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/79e39b39380048d9b0130f040ee3ba84 : SUCCESS in 14m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02eaa80b36284a2792ba72f2e5fd5dac : SUCCESS in 17m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8924066d58de48c1a314badc9aec27d6 : SUCCESS in 18m 41s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/01af080937184f1798c79613765b10fd : SUCCESS in 42m 32s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5c83b0295a2941f4b8e962cd3f4a8cfa : SUCCESS in 31m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8634f750f77344a79d370f3d56b80401 : FAILURE in 14m 17s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/61337e70e34f474ba252fbbecd269384 : SUCCESS in 25m 35s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d4e70cd4f8244beaec8f0e416de545c : SUCCESS in 23m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/92a9b6930abc44fd9a4271c8b169a209 : SUCCESS in 14m 54s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/00126bdc05f743b58ec72026822dd925 : SUCCESS in 11m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a383742a75ce4985acd1ea6562e210cb : SUCCESS in 9m 05s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/20bd6b5b8a524f4e868e84fc9f8ed470 : SUCCESS in 14m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea347b336449401d923dd767ab8aa246 : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d35b3ff5e9be4462afc9cd18a49a0a87 : SUCCESS in 14m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7b45a3d4442f4d738b6cd68efef2669a : SUCCESS in 16m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f5e5bdd79e4462da54bf7d191473174 : SUCCESS in 13m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/492ffe4591fa47afbfdc1bc4c9525f89 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e447cc8459046a7b424bc26c8d37c27 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f4cf3821c4b49d3a2c1a726c46960c0 : SUCCESS in 2m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5da22518793f46d491bb495ba0185134 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9717491fba22492bb6d1231b620322ee : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"84a337939898bbf02cf2f5184cdddeaab29b0c6b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 20:14:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63108b069059726ef0a86cbf7f8b40cb8d56fc1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 21:37:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c595b32b53c640049d4257a37ff34fe6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4aa201edfc2e4894aee0f82c372a4c88 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17b4ab30bf34316abca770330af5067 : SUCCESS in 5m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/26c80eca1acc4ef5a85fe501bc569f4d : SUCCESS in 3m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f2b9bf460347430c99af16d850630799 : SUCCESS in 6m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50bf3745d1544cd5972cb542457a286f : SUCCESS in 5m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a675d35bb73d43dbba4c4e0fead49690 : SUCCESS in 1h 03m 04s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d035858a30364eb48a2950682a204356 : SUCCESS in 1h 01m 42s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e68aa223fa149cf9f3fc3a1cd80ef3e : SUCCESS in 50m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91de23064eb44b0eb598b72342f08881 : SUCCESS in 56m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/aab1dfeef9964340a2bab88a0782752d : SUCCESS in 1h 10m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7be9ab7f19b8440c8a06cbbd8f6a3eb3 : SUCCESS in 14m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/76c49d75b86145429626a3f95751fbf3 : SUCCESS in 10m 32s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2068b25ed860468797bdb16bad2c6ba3 : SUCCESS in 10m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/845e6d5710684850a6105f9137c1b270 : SUCCESS in 12m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5e3819aa93b548cd9ad41f7cdbc3204e : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8c6cd5f2d574a15b0b1c63064db867a : SUCCESS in 1h 21m 35s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1ea63da2ae864a0cbc041d14b683a9d2 : SUCCESS in 1h 03m 15s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52d0fca396024c5b99ef02dc84a1e70f : SUCCESS in 1h 07m 58s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c9721ab967834717ae31fddffc2f7fc3 : SUCCESS in 16m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0de1233462a445eb9773633e1d658382 : SUCCESS in 11m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed88a4281ef3469a9f20c3313f3dd293 : SUCCESS in 10m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0f49a6fda8384a2c857c6d7b5f438bd3 : SUCCESS in 42m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/57eeed74dd6d4d0888089a4ac374dbf0 : SUCCESS in 23m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23730196a85241198dd9ece38426c750 : SUCCESS in 15m 41s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/21c45972410f41368fcd3f71282b121f : SUCCESS in 21m 06s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9161a8ca27df4e9c8238835f6a9842b9 : SUCCESS in 15m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9ffac97667d0452a846ddacc4ca90139 : SUCCESS in 16m 09s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/aec826ff623d429e805aacdcd5c5c0cb : SUCCESS in 10m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb0ff46bf48c4864996e66085b916548 : SUCCESS in 12m 34s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/ea331a5bf8624df48245c80e81e9727f : SUCCESS in 10m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a431965a31bd426dbe59256b2bca27a1 : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/b3b020fc902f4f519027af6ca8832c9e : SUCCESS in 14m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e31ee869b20d4b4bb9664c3c3e94089b : SUCCESS in 9m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51e518e7f33f49438ef0e081be3c5575 : SUCCESS in 12m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f944f5aee7d443d8665aa01c622c740 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57271f10f7cd47198014dfb637d24b97 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3778d9ec7d334340b33a8be931d9ffec : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b291d10b79184daca292428319a6e81d : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6efcfe6fc96d473eb5fc88e5b514a681 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"fbf23362737747e1b71f1d5a449ad11a147c6f81","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-13 13:49:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a665d734064fa3df0dd5d315322d6413eedb729e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 13:49:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cbb9c24d97d3d7c71a3d85094fb07042526b256","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:03:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b43870da6ceb4bb6a6ed897738e66f35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/130a9fe853274758af0acef47d3b7b3c : SUCCESS in 7m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/def6fcab0775453580a1021b9c28147d : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e739ca00e4cc45729854330921a3c44e : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1c9fe14060e14a90a13e10e1fe85a3cd : SUCCESS in 9m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/facd93f711534fe9a4f04f37515de2b0 : SUCCESS in 1h 08m 39s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f57a5948b5904f58a2faa460864bdbdb : SUCCESS in 46m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/999cc80cef6f49e3bf7d5a3ff1d028bc : SUCCESS in 1h 00m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6cf1b50ceaeb41139fef4081462333db : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/89e313458e874d1dbfc7a4b0718d51c3 : SUCCESS in 18m 12s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/913c5599f03345fbbe8ac752c6fa572b : SUCCESS in 7m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac21f2398a764033b60b8b3c570d3d23 : SUCCESS in 7m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/aecba722259548a49ce4fcf04ea78291 : SUCCESS in 1h 06m 14s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3ea378a2eb8464dbe48b89e0b471823 : SUCCESS in 58m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/00cc935500b14321b05f0ec7edbd1703 : SUCCESS in 17m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/096b7f9dbad04162a3406410c88dae5f : SUCCESS in 12m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca5422d76854476aaa36ad7dfa5cfba7 : SUCCESS in 18m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a2f4ecebd6504ba396ae00eda32ac1e8 : SUCCESS in 28m 57s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/176ed9fa432b49bbbc9283ae5095bbf0 : SUCCESS in 31m 11s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/baf1b98b45764fb0988a0b0b01001b9b : SUCCESS in 25m 41s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82c6010586924fbb8ab575f683f3f69e : SUCCESS in 25m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5c7c453d195e45278e3edffbebd6197c : SUCCESS in 10m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a3287c8504e4ce9bc9d93ea3095013a : SUCCESS in 15m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d8b5f340db64fedb492226daa626968 : SUCCESS in 15m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/25c3de9955fb42658cabc568deb52d1f : SUCCESS in 15m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a9c896b54dd43a381dfd938402e1829 : SUCCESS in 14m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/370330ad26e94cecb11b6e12807628bd : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2cbde9735f656b2a4a475984eaf1385a9a947d9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:03:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9facafa6813efaf49b526cd3006db64d5fa5c2c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:04:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9fbeb8613b54c37abe7ca98b6dc71f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d8fa65628c40cba8151993a8d79c53 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea57251983fd13df0eddfbc56d06d8dc15a63986","revisions":{"ea57251983fd13df0eddfbc56d06d8dc15a63986":{"kind":"REWORK","_number":1,"created":"2025-05-12 11:10:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/949489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/949489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/949489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/949489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/949489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/949489/1"}}},"commit":{"parents":[{"commit":"dc5b0e0fc99a7c448c2cd43734466ab1aad2549d","subject":"Merge \"CI: Fix ipaddr invocation in network bootstrap\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc5b0e0fc99a7c448c2cd43734466ab1aad2549d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-09 10:18:06.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-12 11:10:07.000000000","tz":0},"subject":"CI: Fix Bifrost log collection","message":"CI: Fix Bifrost log collection\n\nBifrost changed to use a combined Ironic service instead of separate API\nand conductor services in the Yoga release [1]. Update copy_bifrost_logs\nto get logs from the combined service instead.\n\nAlso remove log collection for RabbitMQ which was disabled in Train [2].\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/820390\n[2] https://review.opendev.org/c/openstack/bifrost/+/645093\n\nChange-Id: I41ccedbd98b0088b92d3eb8671955423815d599f\n(cherry picked from commit c11c62be242029b511cc1292b3636f965c509e56)\n(cherry picked from commit c4427e161b23257b79618b1adc3b95509b0b71cf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea57251983fd13df0eddfbc56d06d8dc15a63986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea57251983fd13df0eddfbc56d06d8dc15a63986"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
