)]}'
{"id":"openstack%2Fkayobe~949498","triplet_id":"openstack%2Fkayobe~master~Ib0a4b02f08cd43f0293f73aaaabfe732e2f6406a","project":"openstack/kayobe","branch":"master","topic":"kayobe-release-2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-12 17:23:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib0a4b02f08cd43f0293f73aaaabfe732e2f6406a","subject":"[release] Sync with Kolla Ansible for 2025.1","status":"MERGED","created":"2025-05-12 12:18:45.000000000","updated":"2025-05-12 17:23:56.000000000","submitted":"2025-05-12 17:23:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949498-kayobe-release-2025.1","meta_rev_id":"9907cfd9a29064a8fe5411567cb2e71de0fe9d27","_number":949498,"virtual_id_number":949498,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-12 17:23:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-05-12 12:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-05-12 12:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-12 12:33:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-05-12 12:58:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-12 12:21:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-05-12 12:21:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-05-12 12:21:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-05-12 12:21:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-05-12 12:21:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-05-12 12:21:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-05-12 12:21:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-05-12 12:33:37.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-12 14:42:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63017c6ba1cab78be183b8bb4f328c8306d2808e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 12:18:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c657c8b7def67e03a15e38270e701db819823c24","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 12:20:13.000000000","message":"Topic set to kolla-ansible-sync-2025.1","accounts_in_message":[],"_revision_number":1},{"id":"9a248c38cd6d11025228e9dd454cdc96a0b726bb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-12 12:22:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d57f36b26cdc15af443b2d2eeb64448dd949e899","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 12:27:09.000000000","message":"Topic kolla-ansible-sync-2025.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"b0018396444da86e2b6ceb0c9992b88efbf7d531","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 12:27:12.000000000","message":"Topic set to kayobe-release-2025.1","accounts_in_message":[],"_revision_number":1},{"id":"e73d75f1402408c5069e5321583ac62bfb9d777b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-12 12:33:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3a36289b80dec4c8ce757d2297f2f783c2c0a390","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-12 12:58:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b74383df24a28f53a4b43d850a8f2999317f870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 14:42:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e2e5f2d3b2b468ab948dbbbc8e59a9c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d55f6d59320c44c1bfd18a6e05747062 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef3a6484e82f47babcee4ca678f8b33d : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e114e1127b8446d96418f4ad6231f64 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63ddd1b3285b435cbcd39a358b349943 : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/487ce9f472f742c085fef92a17dba09b : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902ddd55df7949c79502576a807c0a99 : SUCCESS in 9m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b77da950972343a58b0bc3ae79da9d7c : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/09c704c5f2c541ec8f0d41b5a10b950b : SUCCESS in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ccc7192758647de9ea657db4b78ef01 : SUCCESS in 10m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ad520500100241df8c3b7ff4d7f1e41e : SUCCESS in 1h 03m 16s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/5ba630079a3f4928b0fd23d7465cb94c : SUCCESS in 49m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6bf538a502341cba1b0618644af16e9 : SUCCESS in 1h 05m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dd0b52e233b3471e95b772a09a41b6db : SUCCESS in 45m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9c1a2c4bf6bf494a8063a3146f49a7ff : SUCCESS in 52m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b36d940aa9f04e489db3594ab5277246 : SUCCESS in 14m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93b83b648f344700a2ca5acbbaf5b595 : SUCCESS in 6m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a856e0fe94ff492ebc824345c74e5868 : SUCCESS in 1h 49m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d79fa5fc9d946d89082845d313d41db : SUCCESS in 1h 49m 35s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ea7efa237f5a46ff8e171352d38167a3 : SUCCESS in 1h 47m 34s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9672087fac94f24ac815e981bf7adfd : SUCCESS in 2h 18m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/637de03d63a54335bf7593e364ec25ad : SUCCESS in 18m 37s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/83f1290e41a14bf19e67287781303fbd : SUCCESS in 16m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b21cdc0c3434e33948f00ff99f68957 : SUCCESS in 16m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7bb5782f35674fcf9396177b692ea51b : SUCCESS in 6m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/dd011dfd985a48c5bec02025b04f5902 : SUCCESS in 41m 25s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fbd4aeabc0504fc795c9883d8afff870 : SUCCESS in 20m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34ee8771840e4cf293040cf71f9529c2 : SUCCESS in 30m 34s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/641be685785f48de8f7ba99277e4f282 : SUCCESS in 32m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c17366e8291f49b3b1beb8624c43ef4b : SUCCESS in 6m 38s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b69416b326dd48f5bd0f78aee50d84a5 : SUCCESS in 12m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ae984b7044345e7a8cac2a22fc54b98 : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/0a43578ad4d04590aed9cf563dbffabc : SUCCESS in 9m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a3edc33d5e4a40189f1a2c28fe75ea84 : SUCCESS in 11m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33635b651a0c41368dad2ec24345a846 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"10bf7abc4e85fc53c53222170ba0b6dd6779837a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 14:43:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77293fec9630ac8647203862a24fe5373ee838ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 17:23:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbb87e7f8fa24507a5e42ebf41b7a8f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b60726a91f14539af9f70ea061e74df : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea4d5ab13b3a465cb2018120ae6353a5 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03d0f9b361f2454bb689e514950040b8 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c8279859a78411d9c8b5ff2fababafb : SUCCESS in 8m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/161de408282045e7846e496a5b2ad5a5 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9e4308079c3f4b20b52a27a86b75231e : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/86e5e9c371f343f3af414286fe3931fb : SUCCESS in 10m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f31e002fb57249ee8a85bd723a969948 : SUCCESS in 58m 57s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e7da92aa4108491f9af407d647e2c0c8 : SUCCESS in 58m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e681067b3597438b95101f07d5f936c9 : SUCCESS in 58m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0a5730871b27445fae293955e8c7294c : SUCCESS in 56m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/37780cf10d46479f9aa4b4691753f580 : SUCCESS in 1h 11m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9599314589ec4f6aacc416035aaa24ba : SUCCESS in 17m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61b25eefe1c846c390510f227db7116a : SUCCESS in 13m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b5cf35915ecc475e9bc0905651643e99 : SUCCESS in 2h 22m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad1a051d624b4a86b448ce650528fee7 : SUCCESS in 2h 26m 40s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9ed93d077d624e7986ca8be1c4d7f642 : SUCCESS in 2h 25m 26s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/457000e387414cf98e361079832125f1 : SUCCESS in 1h 41m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/df5789b8ba5b4626aadcb643fb3f1996 : SUCCESS in 18m 34s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ebc5f957e6364eeb97d7fb54f00f4926 : SUCCESS in 8m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d695d06a099846da89bfe6c68cb08914 : SUCCESS in 18m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/db107496c2544d6cb5e5ad9f8454c1ab : SUCCESS in 16m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d7a54c09055943a98310d0f8980ab4b5 : SUCCESS in 36m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edf80e639f9f4f51ad798f00e8c7dace : SUCCESS in 30m 23s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/40cc0cdeb1bf405b8ee49f91394717da : SUCCESS in 23m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ce509c7f01374e9fae28bb2557420026 : SUCCESS in 14m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0b18bac7fbf417eb0bb8034371c2c3b : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/660b138466c64f4a8095803f4478d4b4 : SUCCESS in 11m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f84ae07533943f2ba76bb323065980a : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ddc873f18849bf7d1ebda322e687f6f4d5a2657d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 17:23:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9907cfd9a29064a8fe5411567cb2e71de0fe9d27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 17:23:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c49de4963d4494fa3bee90571087901\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b04bbac13845519c34da2300f9d03d : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"741511f84126468cda1658a647918ee0fbe7f6c3","revisions":{"741511f84126468cda1658a647918ee0fbe7f6c3":{"kind":"REWORK","_number":1,"created":"2025-05-12 12:18:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/949498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/949498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/949498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/949498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/949498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/949498/1"}}},"commit":{"parents":[{"commit":"f0609a8386e12de1b8849024a7e376dd2aa866db","subject":"Merge \"Use pause module to prompt for deprovision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0609a8386e12de1b8849024a7e376dd2aa866db"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-12 12:18:24.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-12 12:18:24.000000000","tz":120},"subject":"[release] Sync with Kolla Ansible for 2025.1","message":"[release] Sync with Kolla Ansible for 2025.1\n\nChange-Id: Ib0a4b02f08cd43f0293f73aaaabfe732e2f6406a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/741511f84126468cda1658a647918ee0fbe7f6c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/741511f84126468cda1658a647918ee0fbe7f6c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
