)]}'
{"id":"openstack%2Fkayobe~949525","triplet_id":"openstack%2Fkayobe~master~Ie4414733ac915faeb9fd10b02f46693651c0d706","project":"openstack/kayobe","branch":"master","topic":"fix-ipa-kernel-options-default","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-05-14 10:49:51.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-14 10:49:51.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-14 10:49:51.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-05-14 10:49:51.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-13 16:33:42.000000000","reason":"removed on reply"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-05-14 08:15:56.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Ie4414733ac915faeb9fd10b02f46693651c0d706","subject":"Fix ipa_kernel_options_default when ipa_collect_lldp is false","status":"MERGED","created":"2025-05-12 16:47:26.000000000","updated":"2025-05-14 10:50:45.000000000","submitted":"2025-05-14 10:49:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949525-fix-ipa-kernel-options-default","meta_rev_id":"b3d4f4a99d4b6aec1d3ffa75578ab1e5463983c4","_number":949525,"virtual_id_number":949525,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-14 10:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-13 16:33:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-05-14 08:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-05-14 08:15:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-05-13 19:29:15.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-12 19:24:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-12 21:10:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-05-12 21:10:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-05-12 21:10:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-05-12 21:10:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-05-12 21:10:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"21dff2f2775e341fc1a7343ff3a0f72079e53793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-12 16:47:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03601247fde1f3a9dfc67a6f9d9644cf60b0f9a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 19:24:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebcc891399de41e18873aba99b8055aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1388f5d0637e4a8a9159bbe1e348ca6e : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5a47ad5e7f347c8b8bb3f94863a970d : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64d85dacdb1d4221af4d3183fb9a0a27 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4d83fe656534ea99a1a3e84f2c3b5c1 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c2222a1d4854fbfa8c442c66dc0e3b8 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bfece409e9d402fb667b45c32df3747 : SUCCESS in 8m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b793979180cd45f389b2c8a21e528dbb : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5247bd42164b45bca4c562f983222573 : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c0b901859f24d609fbc4eca137dd6a5 : SUCCESS in 9m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f1f74ec45d214ba89e699694b0cbdf61 : SUCCESS in 54m 56s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/d63b5485d5c744689031f739a2e42a57 : SUCCESS in 59m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79df41ef49124888be4b3db406a8e46d : SUCCESS in 49m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aa2ce177fe0f4741ab3289a1dd0368ce : SUCCESS in 54m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1a7ece3f4dae4a5c84aa200de310bc43 : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/89e8b98438c043c9a6e0aaeb86ead91f : SUCCESS in 15m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffdcb0f4730f468bb6c76f91d3de689b : SUCCESS in 12m 56s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e1166c0316164601b82a100d3b73dbc4 : SUCCESS in 2h 19m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e28d3bbb405144e0a284bf8a228daa03 : SUCCESS in 1h 26m 46s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/df4604d32ec94875a61ea585c23ead44 : SUCCESS in 2h 17m 35s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a2044e471454d46a046dfe89b35a608 : SUCCESS in 2h 00m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/db4cf58dce7a4d64aa2605566588816b : SUCCESS in 15m 37s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1bee2d0b374348558cf674ea53a3039c : SUCCESS in 16m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47410c3de76b4fae9dc37b80a333b804 : SUCCESS in 15m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f638e501c83b453aa3473c6c5cfdf94d : SUCCESS in 13m 31s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2383dcc9b3ab451899eaa17cb0b877e6 : SUCCESS in 36m 38s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/367c19315a144962aa64247eee3d6ae7 : SUCCESS in 31m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3292dc83cb0d482990a3e35f147c71ff : SUCCESS in 36m 45s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1c115f0f972a4cee928f41d015f2dadd : SUCCESS in 31m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fa7dd480ac494aaca3dab930fbb52040 : SUCCESS in 10m 12s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/06c9e4adca604f6c828d9ab392b0c3ba : SUCCESS in 12m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e2c729f2187484e9aaea94d368b4cc6 : SUCCESS in 10m 53s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/73c233285364427ca5b92befd48955ee : SUCCESS in 11m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/df9e33202dde4683a0921b0308c55798 : SUCCESS in 8m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b290f4bac4254ac7b9463f76764c3b83 : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e249bff99b579ab66003f8d211417e518bf631a4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-12 21:10:13.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"bb047669f1b0a33fddf26e2722739c40e1e076ce","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 21:28:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e31b5cf6e5caff57ded91120fcee3a1293d62e3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-12 22:20:10.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c61f9a397e593d5885ac4647f863d172f46a25c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-12 22:31:24.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ffea0ff6383e33cc880a190585f37140cfb505b2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-12 22:32:02.000000000","message":"Patch Set 2: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"724f8a016cfa054b66412b4808130d742be2900f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 00:56:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a9651c6c8ee43dcbacccf3b8dc00e97\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d21a854bd9ab423da243c637b479e567 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99ce1ae3d8a04f65b1972834eabb680e : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f9d883b9f904f068248848ed7e35355 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c8f1ca3128c46c5bcc80e4513cc074f : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7ad2f09c7a244e8b4ea9c2851cc27ed : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86f7942b3ec408d9efe7244823b0063 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55174469f35341f6947ecefda623eddf : SUCCESS in 2m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/34e3d2bfcc7845849e34d157cbf46b73 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c9e9634c68f4fa4bf91e263b208b719 : SUCCESS in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52936ee4d89c47e78291cb791592b319 : SUCCESS in 8m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e1d1efc933e64256af70d88d5d2d73d7 : SUCCESS in 1h 06m 04s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0a9dcc5fc57e447eb761f9bbdb349ff3 : SUCCESS in 31m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0959d114f094b51be86bd308e32e6c6 : SUCCESS in 41m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/172e2226bf3043978460f66127cab72c : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1558f75080174ff6a0070b3feabf3cfb : SUCCESS in 1h 03m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/633565bd7a37444185a7fa9447d5ad27 : SUCCESS in 9m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b018cb0f20349b48170a99bb3b40513 : SUCCESS in 11m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5fada8a03d864c83b200449b1d85167f : SUCCESS in 2h 10m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49f769e294474f61be31f85fb584887d : SUCCESS in 1h 11m 12s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/60617aa104114a80b11dc70dec063e22 : SUCCESS in 2h 23m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f594f6b48cf41df8b6110126885aab7 : SUCCESS in 1h 37m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/784258ade83c4f0dad4aed75f41ae09a : SUCCESS in 16m 34s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6e6a5de664d94c63b76ed8c40afd4f6e : SUCCESS in 9m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4e869d73ecb42fe8a28c59c073e076b : SUCCESS in 8m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3e7c17e6a7484eb882a6deceb675178d : SUCCESS in 8m 09s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/09148e73708d40b1abc1d1d9aeea8971 : SUCCESS in 20m 58s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b78459fc30e74904b44c1618b2d3df04 : SUCCESS in 22m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e76dcb6c8ac4568ab30710883c4fa64 : SUCCESS in 20m 41s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/58c698ad4a1d4beda90b3c26c2e9faff : SUCCESS in 32m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a73dedd89e1f423a8c5902d84fafe7da : SUCCESS in 12m 58s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e0ffecad7e1a4b7a945df2a003438e3e : SUCCESS in 12m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4782b376cb8740c3a7be182e6f3847f2 : SUCCESS in 12m 09s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/69aeacb429064e27bee20235064f929a : SUCCESS in 10m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/58fde39bb630486e9a52b95d55dfde93 : SUCCESS in 13m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f36e3d97423044208c5bd6c653779830 : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4f3bcb3358c67b714086da720b8325d1211c1aa1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-13 15:55:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bea4a3a546eac85081ebc61527d50c560701b0ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-13 16:05:06.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2d0a75e84db7f2610f5f2d074a78bf49e12354b9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-13 16:05:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"deb6b37d43fbe333811d4c8bbaec401931063cab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"real_author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-13 16:09:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_15197\u003e\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"_revision_number":4},{"id":"00518845438dfab3743e2774ac9016189fee7a56","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-13 16:10:17.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"199e86d883844b7c6f2359018a5396c30152397f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-13 16:12:18.000000000","message":"Patch Set 4: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aa8ba6698d0a8a1fb9bcc2a81e708b8f9dc891b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-13 16:12:39.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"e7191d31ecb43d2a3f9a24afe7125084ae7f806f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-13 16:33:42.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"995166e085049c2196cbaa08fbc08a97b21c3176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 18:38:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41e1d75d5d2e43a899cddf9f3e1e8a8a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5b6a919f5c44724ab78d3403a95176a : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54cfa70a37584e9b9fef1d5ef755c2ca : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d63d54105654b8f8abe92672d582a2e : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a8eace22fbc45108e2fa81cc3970cd8 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/257e1d2db74b4c9f9697b5ee3308ea90 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77b860ebd4d840709bf5537576010398 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2a350a12ddf471690d52317f92d4443 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0460bb6dd50e4eeca082aa4c8297eb8d : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b3d9b56c33f940b3a607cfa918b9a23a : SUCCESS in 9m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3fa5659df8b7489db1dc8fed9f504879 : SUCCESS in 11m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fb249113c36a4ab0b39f79a90a2773c4 : SUCCESS in 34m 03s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0b89e9f5ee514141acf6b7d89375937c : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1415f648c80442b2995d116ff17dc621 : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4f10160290884c679365697246aa9cbb : SUCCESS in 53m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1c2f7a9f35a940e2a463bba5291e2853 : SUCCESS in 1h 03m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d9a017e0e5e64d9a99cc7d5fd1558246 : SUCCESS in 19m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb0f5fbba6184122a234d31749810e1e : SUCCESS in 8m 43s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/53d4c2ee40d54ee8a7e49e604ae2654c : SUCCESS in 1h 24m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/505393850f3a48609fe4b71cc2c58fd2 : SUCCESS in 2h 00m 00s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e666cd95faf948a293e01db6069ffc46 : SUCCESS in 2h 21m 20s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cdceee13b4148d19be17a89f7292134 : SUCCESS in 2h 06m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a14255ef3482424eb4c61a48dfcea2e2 : SUCCESS in 9m 04s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/a712970171fb421e80a80fd7cd26d9a4 : SUCCESS in 16m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a9eb8d273e34995ae0abd65bcea6b6f : SUCCESS in 13m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6cef198b59ce479d8745a5140d57f439 : SUCCESS in 7m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a06205c591e446c3b31c1ee6c5f83636 : SUCCESS in 36m 35s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2a99cbf0b6534ff5a304cf0631f34b65 : SUCCESS in 36m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa7ba06135a148e5ad7c666c81c394a8 : FAILURE in 14m 36s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5502b6ee5a1d487881d749a0c01711c8 : SUCCESS in 36m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d194c9ffcc904a0687e3bfca7d6f0683 : SUCCESS in 6m 39s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/66d7e711e57f4d40aa2f24e4a91d9015 : SUCCESS in 13m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82ffe58cea044382b167144b1104fbb7 : SUCCESS in 10m 39s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/627f96bea228437ebc267a733f76c54b : SUCCESS in 11m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4c865a71299847dcbb3940bf41dcb73b : SUCCESS in 12m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daf6ada3512b48d79abeff07a8207014 : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":5},{"id":"0dfab1f72ef5663f715763c96db982463f77063f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-13 19:29:15.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"78c468f0766b482ae592eaa5600c44cf50f01ec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 22:04:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c397ba90fd6b422ba482abb04967cf1a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7221234fb0c467b8d29defac366c68b : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d97615d99cec41d0860fdc8df5f6c4d2 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4887d87e42984e3988c25cf077750d72 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e94383d45a54ab586dabaa7562c9986 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31a2b70d29e741758a833406d04be173 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fbdbed6d5b8430eb30536ef3f8aef02 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/119091b6c5e74621b522bbdf981d5da5 : SUCCESS in 2m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c6ccf64cf9549d289552d683f56f285 : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6e2ebc7956be44d5b37f25c4d30fb55e : SUCCESS in 6m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3545c80f58014f85b87e1ebf2b70a22c : SUCCESS in 10m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e4a1646659594bdeb5130c407f5497bf : SUCCESS in 58m 35s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1f55ef3e86e744c7869c9c58cb1cb2ca : SUCCESS in 1h 07m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4123b1dbc0e43f69939242b23908eaf : SUCCESS in 53m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/59e21c673392478aa3cbcb91f3ca74ad : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/506fffee553a4a2d8288045ce9577ec9 : SUCCESS in 1h 08m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d39fe7e3c2b14e3ebaa9473cc00b6b49 : SUCCESS in 20m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12cc5518df314535827f126a05794367 : SUCCESS in 12m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/056b5ade931c43e7837aa076acd3817b : SUCCESS in 2h 27m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a22566a510b9483d880be5c23d672664 : SUCCESS in 2h 09m 48s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3e96038d057946de80714e3a05dda026 : SUCCESS in 1h 14m 15s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29ed457f6b9f4ec3b90d3e3520832b12 : SUCCESS in 1h 09m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e48030c1eab848289792a8b7b675c49e : SUCCESS in 8m 26s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/a30a7f933e884e6bb0f234b1dc5d2cdc : SUCCESS in 9m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/196c74600b3d45058f66c8b278f51e88 : SUCCESS in 15m 01s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7fedb4457f484552919b7b586384dec3 : SUCCESS in 13m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/46a066c6edd6444ab026033e9bc4beeb : SUCCESS in 18m 23s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7d825ad7110441f0896d2bb094a5fd0b : SUCCESS in 21m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/264113744f1c4fef84e27c8d6eda5568 : SUCCESS in 19m 26s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/61996a772817479c97a66a9430fc8573 : SUCCESS in 29m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/add74a9c19064f189d9b4c7f7ccee90c : SUCCESS in 13m 12s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/f8c138cc51b644698822debfc2e349db : SUCCESS in 13m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f69650ea9094bc8a7c9f74270f9a0e5 : SUCCESS in 11m 29s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/edcbfbd03ae94f708c756989bee56c55 : SUCCESS in 13m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/551aa1f1aa68438e996fbb8ca2f7de5b : SUCCESS in 14m 08s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1f330e548c3443284f0422ea961f2e2 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":5},{"id":"4562f7563e293a97141fff20b2e243b6389ab728","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-14 08:15:56.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7727492ba762ac84e3a62c0ce7e4be059c14c706","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 08:16:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f6a6e0b357e3f783be04375c112edb562547d144","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 10:49:50.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c219f366b20425f96fde3de76cd5cb7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e74aa0b49e1411bbc7d52044fb5899e : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4dd419c7962e4a2da650c62a51004350 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1bff73fceea34ff2adb74053c9f1aaa6 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e4b00931bec472ebfc5e1fed5333dac : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91ea302b502f46c6ad58205d737b1d1f : SUCCESS in 2m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b0079c8c7ab47aab6567937d583dc0d : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b39b538a669c4690989e630098af49f8 : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/057781df8ba04891869e3703d05db2e2 : SUCCESS in 9m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/35659a1134ef41d88dd663789e814352 : SUCCESS in 54m 22s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1dc916194a8e49e284d27602501f1c0c : SUCCESS in 56m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5f8c4c35d8e4932af2983818a2652f8 : SUCCESS in 53m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/489bf0b1b50b4e41bdebf65f9d7bbbef : SUCCESS in 1h 07m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5ed85b91558f48eea3a9671c88d9523a : SUCCESS in 57m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c0310731adf0482a8cef49600dbdbb1c : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6668c9ea8231415abc82b607376af0e1 : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/031faf524b38425a972334416e6902bf : SUCCESS in 2h 11m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/636da058c9214357a17ee2c245b621d4 : SUCCESS in 2h 04m 59s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/df519ea3c1e14fa8a2dfe612bcf98fd7 : SUCCESS in 2h 32m 16s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/069ccf18410e49c4819f961633b8b2e9 : SUCCESS in 2h 03m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4dde23db70f84b7685f16c480260a66d : SUCCESS in 10m 27s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/34f32418dd1d4024b41e333b483674a6 : SUCCESS in 16m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f10e3155376b492ebf33a0a63ef43786 : SUCCESS in 14m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6fc69211e82f460dba14a870beb4f699 : SUCCESS in 14m 53s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3e526cada4f74c00ab0f9a0b2ee7e42b : SUCCESS in 32m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fd4d35dbab94f32ac9a93285af9d576 : SUCCESS in 29m 04s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/099ac9f838c34bb5a3770b30a4768f60 : SUCCESS in 33m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/68f3a2dc5056474ab736487f99413bd5 : SUCCESS in 13m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d22c04c2dff4b3e848e2a889c6c204f : SUCCESS in 10m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2a076b27c26448828fe2a2a916e0362a : SUCCESS in 11m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e253a823262441c86971893858b38bb : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":5},{"id":"84c03e65dcd2e82b45fc55034b7df1ab71aede6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 10:49:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"b3d4f4a99d4b6aec1d3ffa75578ab1e5463983c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 10:50:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/463c8d499be94c70bb06aad4ee7893d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f3084e61b0464a9964677d87b07f4d : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da39e14161df467da4e277b8ba322b4e : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2087a51c6fadf828a7fe0154853b18fa4a3b0bda","revisions":{"752b6635f46b29f90a067c51f50511f0d2fb9b22":{"kind":"REWORK","_number":1,"created":"2025-05-12 16:47:26.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/25/949525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/949525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/949525/1"}}},"commit":{"parents":[{"commit":"f0609a8386e12de1b8849024a7e376dd2aa866db","subject":"Merge \"Use pause module to prompt for deprovision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0609a8386e12de1b8849024a7e376dd2aa866db"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-12 16:45:06.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-12 16:46:45.000000000","tz":180},"subject":"Fix \u0027ipa_kernel_options_default\u0027 when \u0027ipa_collect_lldp\u0027 is set to false","message":"Fix \u0027ipa_kernel_options_default\u0027 when \u0027ipa_collect_lldp\u0027 is set to false\n\nTrivialFix\n\nChange-Id: Ie4414733ac915faeb9fd10b02f46693651c0d706\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/752b6635f46b29f90a067c51f50511f0d2fb9b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/752b6635f46b29f90a067c51f50511f0d2fb9b22"}]},"branch":"refs/heads/master"},"7b0a493ed348bda088fa9aae6a3100061f039304":{"kind":"REWORK","_number":2,"created":"2025-05-12 22:31:24.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/25/949525/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/949525/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/949525/2"}}},"commit":{"parents":[{"commit":"3c916b3e95b403f6aef12dda32a055935c9be477","subject":"[release] Use OpenStack 2025.1 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c916b3e95b403f6aef12dda32a055935c9be477"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-12 16:45:06.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-12 22:31:13.000000000","tz":180},"subject":"Fix \u0027ipa_kernel_options_default\u0027 when \u0027ipa_collect_lldp\u0027 is set to false","message":"Fix \u0027ipa_kernel_options_default\u0027 when \u0027ipa_collect_lldp\u0027 is set to false\n\nTrivialFix\n\nCloses-Bug: #2110505\nChange-Id: Ie4414733ac915faeb9fd10b02f46693651c0d706\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b0a493ed348bda088fa9aae6a3100061f039304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b0a493ed348bda088fa9aae6a3100061f039304"}]},"branch":"refs/heads/master"},"66a06f21f87ccb8b0af675d63b0db0a6c7eedef5":{"kind":"REWORK","_number":3,"created":"2025-05-13 16:05:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/949525/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/949525/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/949525/3"}}},"commit":{"parents":[{"commit":"3c916b3e95b403f6aef12dda32a055935c9be477","subject":"[release] Use OpenStack 2025.1 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c916b3e95b403f6aef12dda32a055935c9be477"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-12 16:45:06.000000000","tz":180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-13 16:05:01.000000000","tz":120},"subject":"Fix ipa_kernel_options_default when ipa_collect_lldp is false","message":"Fix ipa_kernel_options_default when ipa_collect_lldp is false\n\nTrivialFix\n\nCloses-Bug: #2110505\nChange-Id: Ie4414733ac915faeb9fd10b02f46693651c0d706\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66a06f21f87ccb8b0af675d63b0db0a6c7eedef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66a06f21f87ccb8b0af675d63b0db0a6c7eedef5"}]},"branch":"refs/heads/master"},"a7be79ee9e3044c59b4c08a907f3e2bebe9d5f17":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-05-13 16:09:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"real_uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/25/949525/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/949525/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/949525/4"}}},"commit":{"parents":[{"commit":"c2ce28256c22a0187a0409c2e390a776dce29858","subject":"Rename docker group to container-engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2ce28256c22a0187a0409c2e390a776dce29858"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-12 16:45:06.000000000","tz":180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-13 16:09:13.000000000","tz":0},"subject":"Fix ipa_kernel_options_default when ipa_collect_lldp is false","message":"Fix ipa_kernel_options_default when ipa_collect_lldp is false\n\nTrivialFix\n\nCloses-Bug: #2110505\nChange-Id: Ie4414733ac915faeb9fd10b02f46693651c0d706\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7be79ee9e3044c59b4c08a907f3e2bebe9d5f17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a7be79ee9e3044c59b4c08a907f3e2bebe9d5f17"}]},"branch":"refs/heads/master","description":"Rebase"},"2087a51c6fadf828a7fe0154853b18fa4a3b0bda":{"kind":"REWORK","_number":5,"created":"2025-05-13 16:12:39.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/25/949525/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/949525/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/949525/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/949525/5"}}},"commit":{"parents":[{"commit":"c2ce28256c22a0187a0409c2e390a776dce29858","subject":"Rename docker group to container-engine","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2ce28256c22a0187a0409c2e390a776dce29858"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-12 16:45:06.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-13 16:12:00.000000000","tz":0},"subject":"Fix ipa_kernel_options_default when ipa_collect_lldp is false","message":"Fix ipa_kernel_options_default when ipa_collect_lldp is false\n\nTrivialFix\n\nCloses-Bug: #2110505\nChange-Id: Ie4414733ac915faeb9fd10b02f46693651c0d706\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2087a51c6fadf828a7fe0154853b18fa4a3b0bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2087a51c6fadf828a7fe0154853b18fa4a3b0bda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
