)]}'
{"id":"openstack%2Fkayobe~949574","triplet_id":"openstack%2Fkayobe~master~I8778247e6fa8f4cb429fc163feaea66f4307582b","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-13 13:22:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8778247e6fa8f4cb429fc163feaea66f4307582b","subject":"Rename docker group to container-engine","status":"MERGED","created":"2025-05-13 08:39:27.000000000","updated":"2025-05-13 13:23:48.000000000","submitted":"2025-05-13 13:22:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949574","meta_rev_id":"e847c597ea9207218a29ff82d03c170755919669","_number":949574,"virtual_id_number":949574,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-13 13:22:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-05-13 08:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-13 10:38:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-13 08:49:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-13 08:49:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 08:45:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-05-13 08:49:19.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-05-13 10:38:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-13 11:04:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"524df231644fd54a25f8517d00ce334b4ae0dcfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-13 08:39:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ead2102b844583bb33e5d9d3ed3c693233d015d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-13 08:45:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fcf8d3382cf7fe82066e715ea7260622ba12938b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-13 08:49:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1529df7644dae8c3e40b60bf7cfbeb8c11a2b12e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-13 10:38:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"02b1962e503dadbae10f12af10ddfa1cb40138aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 11:04:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f42d3a4203d547c3b6dbca231ee17aa4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ceec1f6d19e4b498838dc9d75e31219 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5cab2f4d665493eb586f1d30555ee7c : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/422d205631434f579c3348d23d0e84b9 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f70bcb352bba4a2294acdeef77f3daaf : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5dc3b1becb494e6a87108b11ec0ac4c7 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1e7a7c364a4876863ba942e9cd32b3 : SUCCESS in 9m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/06217b7318694d85aa2ea58291b51a1c : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0191d73a9f17434c908ef18c824db3e3 : SUCCESS in 10m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0b26100ec2e146efab660f5edb8f9576 : SUCCESS in 11m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a8d89a7703f14da7ba35cd13ff8961a5 : SUCCESS in 1h 04m 57s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/df65c45235364ec0ac2af24071269485 : SUCCESS in 1h 07m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7af34004a8b41c593297fa3ef0adb95 : SUCCESS in 1h 03m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5cfdbf2cfd2a4413877ce83df2f521ce : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4be9cab0d74345bb97141fc3f5f48724 : SUCCESS in 34m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c0ec0442c86f45d5be9826703dc0ddbb : SUCCESS in 20m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a75af7aba9224455803c103c34981744 : SUCCESS in 7m 56s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a0fe95b5c4aa4464b705294e83e4c464 : SUCCESS in 2h 23m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94f8679a9374497e80a1ab8b80649d36 : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ad4504f506564b69a43b4449b79ff9da : SUCCESS in 1h 16m 23s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae2015350f804219a7d921fbf945dc8c : SUCCESS in 2h 21m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/19084812b5084af5b3b42f8615d651ce : SUCCESS in 19m 40s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/3407b7de307242279c5c966a63375c19 : SUCCESS in 8m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3b6e20fd55941258e2a852cf669333c : SUCCESS in 14m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d94d97398d364727a790d0d2663f4a54 : SUCCESS in 8m 31s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d93bc8a54e484dc29d65eb8b552d3e2e : SUCCESS in 34m 31s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b5120e80a7644b88a2729daf00fdeca2 : SUCCESS in 33m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f7b1ae398bb45fc92e48067af2354b5 : SUCCESS in 30m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7e4d6b95e2a54d459e60bdb9a241fdec : SUCCESS in 20m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8eeb18b2e0ee49568baf8ba218eadb98 : SUCCESS in 8m 03s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/c5a7fdcf588d48138ed1b2ee068a1dac : SUCCESS in 11m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3c70ce5236d4d198e65d6dabf026d37 : SUCCESS in 12m 33s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/8af5b3ce329e4549afb1378dc7d5240f : SUCCESS in 10m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/535e09e6d5104b72b4b276483a85a4a6 : SUCCESS in 11m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e082ca4f8444df6b0f6ba0fc364b166 : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f4c88df6888a651661ecde8d7719385364960f5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 11:05:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88518e601a2caf7202d08f52abfa8e552819678b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 13:22:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acaeb5be007448bd9dd3ec2805575304\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea177f044d54623b8a41c93fe601552 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a119559f950b4f4db2fe2b60ff820b0d : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60542a4d5e17469daad217bb6cecfa67 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463855d345dc49a3bda1fbfd9b4a1eaa : SUCCESS in 8m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/396893a7024c41aa8388bd2e78fdafd4 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8aac363ad134ece98f020862c3bbe89 : SUCCESS in 10m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/294ab7ed4b264c6fab89134fe699622f : SUCCESS in 10m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0a6f9a14b4244d73966ad023e51aad03 : SUCCESS in 33m 01s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/2737f9789f554977a3c3d7db107c0bc0 : SUCCESS in 59m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b689fc6ab434a5689293758f8b13ff4 : SUCCESS in 34m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f59bec1284e14ec18d278ba14ff234fc : SUCCESS in 54m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a9819c040b7e4877b7f37cedcb943e21 : SUCCESS in 1h 07m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7fd90bf9cd534259a06c63b3e247fc9f : SUCCESS in 11m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d79b771ea104e7fb79e63255cb54db5 : SUCCESS in 13m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/45fe5c0d0a7d4b6290d87c45d81bb85e : SUCCESS in 1h 21m 57s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e0c9423881948e382d97d062c8a5a0d : SUCCESS in 2h 13m 32s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2d012d44e958421f8888ac737f073bce : SUCCESS in 2h 11m 13s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2179fba02e848b9b3608b9ee9b934d0 : SUCCESS in 2h 01m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/af6652e1fd294de3b8b0d8948e1a7166 : SUCCESS in 15m 53s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/7d476ea12efa47a4954075b5a1531996 : SUCCESS in 17m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b6a55fd618f4d21b061133ce05a3d05 : SUCCESS in 14m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2dfaa48df1cc4b79b9cb300997d618c2 : SUCCESS in 18m 05s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2cdf54e8b8bb4a99af451081e2fd26b3 : SUCCESS in 31m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/969c641797dd434f9a807ba355b3ed6c : SUCCESS in 33m 36s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8e630c2500e84a7fad2e217ad1b2ac28 : SUCCESS in 33m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9fce15f40ccb4c67b7dfc421af470dab : SUCCESS in 12m 09s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adc80f1fae14459fbec225b64e150e25 : SUCCESS in 10m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/34b532d33a934cc296f4a8f34374230d : SUCCESS in 11m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6977a3409dcc4856996ae54fddb99c79 : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ea344831519a9240ad44598d3fc1e315a41327d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 13:22:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e847c597ea9207218a29ff82d03c170755919669","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 13:23:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc50c6c963f947ff9f975c45c1bc1512\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4666b057552a41e38b69a83ff6f2ea9e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2ce28256c22a0187a0409c2e390a776dce29858","revisions":{"c2ce28256c22a0187a0409c2e390a776dce29858":{"kind":"REWORK","_number":1,"created":"2025-05-13 08:39:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/74/949574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/949574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/949574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/949574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/949574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/949574/1"}}},"commit":{"parents":[{"commit":"3c916b3e95b403f6aef12dda32a055935c9be477","subject":"[release] Use OpenStack 2025.1 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c916b3e95b403f6aef12dda32a055935c9be477"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-13 08:39:05.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-13 08:39:05.000000000","tz":120},"subject":"Rename docker group to container-engine","message":"Rename docker group to container-engine\n\nThis matches the same change done in kayobe-config-dev [1].\n\n[1] https://review.opendev.org/c/openstack/kayobe-config-dev/+/910017\n\nChange-Id: I8778247e6fa8f4cb429fc163feaea66f4307582b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2ce28256c22a0187a0409c2e390a776dce29858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2ce28256c22a0187a0409c2e390a776dce29858"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
