)]}'
{"id":"openstack%2Fkayobe~949946","triplet_id":"openstack%2Fkayobe~master~Ia8777550b00bd1cf0391a1c7bddbbf99fe7c0d20","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-19 16:49:29.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-19 14:25:55.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-19 16:49:29.000000000","reason":"Change was submitted"},"36238":{"account":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"last_update":"2025-05-19 16:49:29.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-05-19 16:49:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8777550b00bd1cf0391a1c7bddbbf99fe7c0d20","subject":"Fix authentication failures due to missing system scope","status":"MERGED","created":"2025-05-15 20:36:44.000000000","updated":"2025-05-19 16:50:26.000000000","submitted":"2025-05-19 16:49:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949946","meta_rev_id":"d8cca0d667748f9fab4a0a0c043a7d8bf52c7d54","_number":949946,"virtual_id_number":949946,"owner":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-19 16:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-05-19 13:43:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-19 13:23:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-05-19 14:09:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-05-19 14:09:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-05-19 14:25:55.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-05-19 14:09:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-19 13:36:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-15 21:00:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-15 23:15:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 04:54:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-05-19 10:54:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-05-19 13:23:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-05-19 14:09:07.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"144fb67b879e36b328ee3c2d53adbaa2ce3f238e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-05-15 20:36:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37cbb373ac71f98ef4c7ef6be4676b5b5203dee9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-05-15 20:37:34.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ecf6ad14a6a1a34da52699755b859ef28d0eba55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-05-15 20:42:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"697d6c00aa071029ece83121ae7166fcb3525d0c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-15 21:00:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ce34c6aa004f65e036af0242ee1ccff3f1e7db68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 23:15:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/657fffe9a8624bf09e805bdbe69d0490\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41139c53ffd84b6b8555c1d415685b7a : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1141854831fc4338ae7f7df2e7c85de4 : SUCCESS in 2m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1be2d419f84f472f9690bf903906b818 : SUCCESS in 2m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/379fcfe916094dc0b70da1d3ccacdf6d : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8c90ecf1e974f64ab03d91c739553fe : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/945caef910cf4187a0d58ffadec56672 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3d8dc3258684d2b8fc34a8d82e60393 : SUCCESS in 1m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e31fb8cb14f64e9387248c6f68f4a885 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8779fcf2e55144a8a46b62c782e51fd5 : SUCCESS in 9m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/708709f39afc42d5993ff4ac2e7f6c3f : SUCCESS in 8m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ff6b9941b5464863b2c1e1a68ca854a4 : SUCCESS in 36m 34s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/33116dc6e4284c87bee3f41a3f23b4b6 : SUCCESS in 1h 01m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a67a6b519c924c72b08bff986db08d39 : SUCCESS in 34m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/87d5427f15d549849c3d7bc1944c7965 : SUCCESS in 1h 01m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f0df26c2da3e4f4382f670b8290b194a : SUCCESS in 37m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/17333abb5dce47c4b15fc2fb7293cf64 : SUCCESS in 18m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbba698efd03496493a3a0ec0b88c71a : SUCCESS in 8m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/98b46d0f642c4076a50820dec812f24b : SUCCESS in 2h 24m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27a5001adc1540879c1a9d3aa442efc7 : SUCCESS in 2h 24m 34s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/47fec6e39a4c4938bb1698908cfe25c6 : SUCCESS in 2h 32m 18s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d92e5998d9d4326b1c4ad8cbbed6218 : SUCCESS in 2h 23m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3f7d7cd4413c4ec0b18a3ca5ce8efabf : SUCCESS in 20m 35s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/14d3d39709d34a3c9476b61d2b4cd969 : SUCCESS in 18m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2446f3bd3c444eee9200df3a3aab2af8 : SUCCESS in 15m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5751f34ee6b947219d9990c4becd616d : SUCCESS in 16m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/955ca106a91142b4a439f0ab837a4737 : SUCCESS in 38m 49s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/06fcaf0f8f7845dd8c27cf40cd8c719e : SUCCESS in 34m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2d2b9fde0bc4c87936484cb3990d346 : SUCCESS in 32m 12s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3a0e1363ac214148b6a399d903f5fe0a : SUCCESS in 20m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f5749f76626241b390fa4be11d978265 : SUCCESS in 12m 36s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8e4e19ea8bdd48a6af48f19d849c8e16 : SUCCESS in 11m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26bc02eef3d44c1fb40c3153ef2d805f : SUCCESS in 9m 59s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/7a2c4be8081745f193af6fb90ff62894 : SUCCESS in 9m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5bb07f00a52d44f99da7d716453ec2b6 : SUCCESS in 12m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44d598306acb4ee6a0de35745b4acd0a : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":3},{"id":"fa0eb4c8d463ac774ec07febaea9619a393ecdd1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-16 04:54:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f3b92a19f8f53825e8cc6e25b6ff9ff04776fd8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-05-16 13:09:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7fd49c4268c505f6e7bc5e495d28a85c56c66a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 16:00:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/679c03b3b069415c9ba6e1d54b353105\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2abd805deea74603a9d4e7941b132c2b : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15ef89b838a640fab56f074a470aba21 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32f23cb2aef54e728ec6d54df3edd943 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31f297323bc24bc7a74e1106de88dfc7 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/865cbed10dcc4ca5a8c44688a55b0efe : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e033d670149475bb7d82ed7a7654847 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e9c7a1d84c54f89be27dce827294656 : SUCCESS in 2m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/603a9e780ee04451844322a5a20d73e5 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9b8b814b9c274ede987983146dcf6a17 : SUCCESS in 9m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/513933e5735546ad99738660b213be0c : SUCCESS in 5m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4e68df1082014c58ac741c4be087d8d6 : SUCCESS in 59m 40s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7bfee25a08c044d09d035b79ad574095 : SUCCESS in 1h 04m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/789509eb20b642d594d41814893d5d67 : SUCCESS in 36m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6672adae9967484e97641a3c4df47059 : SUCCESS in 30m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5f5fbe1c3b2e4bc2a44e56b0c4884969 : SUCCESS in 1h 08m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/96ba96f5157f4210bd0f398d0bb09dd2 : SUCCESS in 16m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9bb6012340f4e2a8dc9945f3248f6e0 : SUCCESS in 13m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d072ca1256a24d7db595e2b77701197b : SUCCESS in 2h 14m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6176e06174e64c3aa6b23db3bf9f7e21 : SUCCESS in 2h 23m 19s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c4bd1f580e804eb9b9d4c649061f0efe : SUCCESS in 2h 44m 36s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a0dedd285264b278cb8975bc6e10208 : SUCCESS in 2h 25m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/67c5d73f67de4df6bd8efbd252ee9afe : SUCCESS in 19m 44s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/7419ad60c4554196830c7400caf1dcde : SUCCESS in 20m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76587fdc960549a280888b77bafe8cd0 : SUCCESS in 19m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/da7f7a514cbc4dd0929a45550623282e : SUCCESS in 16m 14s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/87f634b27d6e4376b206cce0a683aab7 : SUCCESS in 35m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/75d4558bde4f4c1781aa5a39a62a70bd : SUCCESS in 35m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61248eabd41645c08f1ff8f005e36e1a : SUCCESS in 29m 03s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/831416e0cb4842808d70fbc2c10e1116 : SUCCESS in 40m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3d52d3db2a1f46b0b976b378ceed7853 : SUCCESS in 14m 14s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/7b70751a6f0540789200af51f184fc5c : SUCCESS in 15m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db95210a9fbe45a7a800c3e1a09c2440 : SUCCESS in 9m 40s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/878ac2c57d2b4f8c87fe1ea174d4a7c9 : SUCCESS in 13m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/badbf43bdb964b71a1cd40b35c8cffb6 : SUCCESS in 10m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8711c406d24141799c380b9756b3c15e : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":4},{"id":"430eb26f630ab08a60d888a7d83a41c8718c9a86","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 10:54:22.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a9e33eea3ca8fc6ec05bb1b9a2b5e6f3d510b4b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 10:59:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"723eb8776a28aed350704b6025758ebcaaf11cbf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 13:23:38.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8ad99005e82a2a6e1e0d5e14e61c566e1d9dfab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 13:30:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5471beb0e0d4daaaff2a274d1e834fb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e84e7d57f1c8413d8c0ffe2c487d6b43 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ebd92e7f07946188bec93b68bc6ba0e : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92a5987dd2ad4bbf99ca17335d8f5df7 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f11cdfc4a06a42ad932b9f9125be71c0 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b55d26fbdef440c3a52b681745d924f1 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e4eeb7b5584c86a4dbd4ad0d2330d7 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c987734e7444ec2bc1af6d4d2cb085d : SUCCESS in 2m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d6c7221fd6843e5aa0a13ba112fe108 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e9110a33c4bf43c996b3a86e54c026ab : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/793514548de34318b8ce1658ed96fcfb : SUCCESS in 8m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9186646b461c4ad9b473a741c8c36841 : SUCCESS in 54m 12s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/c7750adc00d541a391607eb00cf78eb8 : SUCCESS in 1h 10m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3aef27ca60043fa9054344b63e0a895 : SUCCESS in 54m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a5065ee696f64ef38de6f536445246f0 : SUCCESS in 1h 06m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4e1cb4edb2a545fbb0b0ae181c432cbe : SUCCESS in 1h 02m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6745e0f32022432389f1f1cc02d1ddd8 : SUCCESS in 18m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1704850bcd6411c82940cd92fb2dc98 : SUCCESS in 17m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1f7c3c84829f4caa812703f4728c8e25 : SUCCESS in 2h 10m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b017f2bd3e864a90b237c4441f6e8315 : SUCCESS in 2h 27m 43s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/fd1f460d985b47a1ad9c9f4a40e88be2 : SUCCESS in 2h 10m 41s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be152bb3df194d969b5c793b06b51218 : SUCCESS in 2h 24m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/24dd7b1e88144c739b350ba083252ce2 : SUCCESS in 19m 06s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/77c58305ff134ecdb13e1d07b6b7b395 : SUCCESS in 18m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ac33837e1f24568958e347c717699ec : SUCCESS in 18m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/556aa93c75e242fea93cee92667df160 : SUCCESS in 17m 22s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6c04d56613c24054830912cadcf173c9 : SUCCESS in 36m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1f718b6954de4367a99debb695e9f94a : SUCCESS in 35m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e869091396424b3b85f7ed464bd82c81 : SUCCESS in 35m 43s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ac882e0ca5a244d6858e7cf1c4bed18d : SUCCESS in 34m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d154848a32c447349d9850e2f9599871 : SUCCESS in 13m 33s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/bdc4f167a5cc4d41a2542a8052dd7e56 : SUCCESS in 13m 26s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3437b21e0f2740758db51994fb006e32 : SUCCESS in 10m 59s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/1b297a4a3aaf4b09983d0bcfe0b26c63 : SUCCESS in 12m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fe826ec54d614c32917ff581873be283 : SUCCESS in 12m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68ba30097ff5496ab84311cba9181bef : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":5},{"id":"9c79a34c43edfaa2b388ad5a82efbd96db74197a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 13:36:41.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"ab389b16bf63ec9776936cf480ef0c793b8a92d7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-19 13:43:39.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4182363d85bf540dcc76253601c2bde9da6ec7f9","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-19 14:09:07.000000000","message":"Patch Set 5: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3b7a9d2397ca6560bc873fb0f648c18adb54fe8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 14:09:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"96dc546d8b4279c32cc4608a4e8f868ef96454a9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-19 14:20:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b34963b0f8bd6eb8e0acea4e3d5a8ad7a90a6a3d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 14:24:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"267069f9df55e2eefbb20afc54ea0a64852e5c59","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-19 14:25:55.000000000","message":"Patch Set 5: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ce6e858a326cacd28867f0999993bd9859263144","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 16:49:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1273332b00a5412fa5d168a3dfa40d71\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d5a86322320460fb8d59e4e3c6cf9bb : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cc0d3f3e028435eaa857715c988f891 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/594b8c95c5e94fb383bf696e1bf83501 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f83bcc127546e5bf2a68e5b2f9470d : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c4861926cf7415c86fe31f90a0328b7 : SUCCESS in 2m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a803fa9e30ed46d1bd8b04de1dc039d3 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d7c4842467d4fc18e8523d74917ac77 : SUCCESS in 9m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/46aeed92a4094fd0bbd727bf0847767b : SUCCESS in 8m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4d4c675a6f044a679f41ca0d663ed01a : SUCCESS in 57m 46s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/acac4e1b338a4df6a3ef9f62e58f5932 : SUCCESS in 56m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9dbc64040c354a2c8de08a213b5d9f4d : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/13bb01d1934f474aab84b31a36dc8dbe : SUCCESS in 1h 03m 26s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/91aedce8ec524f16b21ae30b31243e33 : SUCCESS in 1h 01m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6bfacac20cf046b48047a78c34963ac5 : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a847e86178f47f09ee42563f823bbb7 : SUCCESS in 13m 45s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1a0d075785e4459ca3c5d491947e09ec : SUCCESS in 2h 03m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ead82c4be0c648c380b86ed0c37999d1 : SUCCESS in 2h 11m 34s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/52c6e3220157455b9367896ccce18ba1 : SUCCESS in 2h 24m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/790741fe75414793b1401e4c30728e4b : SUCCESS in 2h 10m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/92f352a1052f45218e1337afc252d02d : SUCCESS in 16m 15s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/dfd86c276efe4066a26bf42ff100b3dd : SUCCESS in 10m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/058d0d8c163548278f67ee896ffaa1b5 : SUCCESS in 14m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fe77dbea71a64263b45139db3e31513d : SUCCESS in 15m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6879f41c41234022bf9a9a9cbafa41f2 : SUCCESS in 29m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2445657c44234b22babde50c9e8a375b : SUCCESS in 29m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/98079389d03a482e8edc79c5371d11a1 : SUCCESS in 34m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c7010ad5e7c9441bb5aa9a81b3fd8157 : SUCCESS in 13m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/679d5fdfe8794ffaa29f39afb90fa76c : SUCCESS in 10m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cf11f2993e24400998e21aa918d7dc81 : SUCCESS in 11m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72f82ba843204bc498a9760949d2b0d1 : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":5},{"id":"da89014bd8016b059436c88a1a6014aeee956193","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 16:49:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d8cca0d667748f9fab4a0a0c043a7d8bf52c7d54","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 16:50:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37697846dc2742bbbf8b6a17ed64238e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f5c1ef9ad24e83ac3bfae5ba949848 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d56b12bc39724dc4a487796063373d72 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"73d57e5930191ccf8149498e0f93ac4c711b904b","revisions":{"1634b9e5913e89da253481c3490b1506540e43fc":{"kind":"REWORK","_number":1,"created":"2025-05-15 20:36:44.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/46/949946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/949946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/949946/1"}}},"commit":{"parents":[{"commit":"b4733da80a3c532cbd45b76494f72eef06a9c8ba","subject":"Add the ability to override the system\u0027s default repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4733da80a3c532cbd45b76494f72eef06a9c8ba"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-05-15 20:30:32.000000000","tz":60},"committer":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-05-15 20:36:20.000000000","tz":60},"subject":"Fix authentication failures to missing system scope.","message":"Fix authentication failures to missing system scope.\n\nEnsure the OS_SYSTEM_SCOPE environment variable is present in\nopenstack_auth to prevent authentication issues occuring in\nbaremetal-compute playbooks.\n\nChange-Id: Ia8777550b00bd1cf0391a1c7bddbbf99fe7c0d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1634b9e5913e89da253481c3490b1506540e43fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1634b9e5913e89da253481c3490b1506540e43fc"}]},"branch":"refs/heads/master"},"8bb1ae913499cc4e50dc552d7706f62a7806b505":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-15 20:37:34.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/46/949946/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/949946/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/949946/2"}}},"commit":{"parents":[{"commit":"b4733da80a3c532cbd45b76494f72eef06a9c8ba","subject":"Add the ability to override the system\u0027s default repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4733da80a3c532cbd45b76494f72eef06a9c8ba"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-05-15 20:30:32.000000000","tz":60},"committer":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-05-15 20:37:21.000000000","tz":60},"subject":"Fix authentication failures due to missing system scope.","message":"Fix authentication failures due to missing system scope.\n\nEnsure the OS_SYSTEM_SCOPE environment variable is present in\nopenstack_auth to prevent authentication issues occuring in\nbaremetal-compute playbooks.\n\nChange-Id: Ia8777550b00bd1cf0391a1c7bddbbf99fe7c0d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bb1ae913499cc4e50dc552d7706f62a7806b505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bb1ae913499cc4e50dc552d7706f62a7806b505"}]},"branch":"refs/heads/master"},"9835bdf6ac0febf1b6603a1d27f6f16af39df1ba":{"kind":"REWORK","_number":3,"created":"2025-05-15 20:42:22.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/46/949946/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/949946/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/949946/3"}}},"commit":{"parents":[{"commit":"b4733da80a3c532cbd45b76494f72eef06a9c8ba","subject":"Add the ability to override the system\u0027s default repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4733da80a3c532cbd45b76494f72eef06a9c8ba"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-05-15 20:30:32.000000000","tz":60},"committer":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-05-15 20:42:14.000000000","tz":60},"subject":"Fix authentication failures due to missing system scope.","message":"Fix authentication failures due to missing system scope.\n\nEnsure the OS_SYSTEM_SCOPE environment variable is present in\nopenstack_auth to prevent authentication issues occuring in\nbaremetal-compute playbooks.\n\nChange-Id: Ia8777550b00bd1cf0391a1c7bddbbf99fe7c0d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9835bdf6ac0febf1b6603a1d27f6f16af39df1ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9835bdf6ac0febf1b6603a1d27f6f16af39df1ba"}]},"branch":"refs/heads/master"},"49ece9888cd8ca7fbfd0a15f3045f3d154505109":{"kind":"REWORK","_number":4,"created":"2025-05-16 13:09:07.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/46/949946/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/949946/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/949946/4"}}},"commit":{"parents":[{"commit":"b4733da80a3c532cbd45b76494f72eef06a9c8ba","subject":"Add the ability to override the system\u0027s default repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4733da80a3c532cbd45b76494f72eef06a9c8ba"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-05-15 20:30:32.000000000","tz":60},"committer":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-05-16 13:07:47.000000000","tz":60},"subject":"Fix authentication failures due to missing system scope","message":"Fix authentication failures due to missing system scope\n\nEnsure the OS_SYSTEM_SCOPE environment variable is present in\nopenstack_auth to prevent authentication issues occuring in\nbaremetal-compute playbooks.\n\nCloses-Bug: #2111103\nChange-Id: Ia8777550b00bd1cf0391a1c7bddbbf99fe7c0d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49ece9888cd8ca7fbfd0a15f3045f3d154505109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49ece9888cd8ca7fbfd0a15f3045f3d154505109"}]},"branch":"refs/heads/master"},"73d57e5930191ccf8149498e0f93ac4c711b904b":{"kind":"REWORK","_number":5,"created":"2025-05-19 10:59:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/46/949946/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/949946/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/949946/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/949946/5"}}},"commit":{"parents":[{"commit":"8530480926aaf91813ce1126a73af69af04f012a","subject":"Merge \"CI: Improve readabiity of ansible output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8530480926aaf91813ce1126a73af69af04f012a"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-05-15 20:30:32.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-19 10:58:59.000000000","tz":120},"subject":"Fix authentication failures due to missing system scope","message":"Fix authentication failures due to missing system scope\n\nEnsure the OS_SYSTEM_SCOPE environment variable is present in\nopenstack_auth to prevent authentication issues occuring in\nbaremetal-compute playbooks.\n\nCloses-Bug: #2111103\nChange-Id: Ia8777550b00bd1cf0391a1c7bddbbf99fe7c0d20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73d57e5930191ccf8149498e0f93ac4c711b904b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73d57e5930191ccf8149498e0f93ac4c711b904b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
