)]}'
{"id":"openstack%2Fkayobe~950203","triplet_id":"openstack%2Fkayobe~master~If1157a56110e26c3668bd54cd0293a9d3f936fc6","project":"openstack/kayobe","branch":"master","topic":"fix-overcloud-deprovision","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-19 13:48:52.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-19 20:24:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If1157a56110e26c3668bd54cd0293a9d3f936fc6","subject":"Fix execution of kayobe overcloud deprovision","status":"MERGED","created":"2025-05-16 16:32:45.000000000","updated":"2025-05-19 20:25:07.000000000","submitted":"2025-05-19 20:24:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950203-fix-overcloud-deprovision","meta_rev_id":"773a7cbe150ebe2c81b0a1c993de3615efc55701","_number":950203,"virtual_id_number":950203,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-19 20:24:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-05-19 17:58:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-05-19 18:36:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-05-19 08:22:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-05-19 17:58:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":-1,"date":"2025-05-19 08:22:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-16 19:32:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 18:57:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-19 08:22:15.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-05-19 08:27:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-19 17:58:13.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"925c35b5ba291695198bbe8a9483ae6ed4a95e00","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-16 16:32:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60d104197adaad3a762afb6e355313d12a4f2585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 18:57:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5704731f5d114c5ab47657e95a5c2c1c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4bdc9a821d8433780df24028d5015bf : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bb353ae342f40288055d756f64c6426 : SUCCESS in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1b176d59b8142ea8403d60a19ff5032 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b7679a3158841d5a45776b21a3e6b01 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de26388ab95a453591f64ef003a7c8ae : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec1399763ed47af96c374aead3c0873 : SUCCESS in 9m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6e027f4f4fbd47d28c44182b919326b4 : SUCCESS in 3m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/75de8f0cdc5e4a45aab3e9933616b591 : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8829b84c13c44cae83a21ece47b5aa46 : SUCCESS in 5m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b2ad7741dae24935927db228f387ee6c : SUCCESS in 28m 59s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/4f45706df90c4b65a3891a62a7c432d2 : SUCCESS in 1h 01m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a727a3217a884f9bbda1c857cabd392d : SUCCESS in 56m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6212172bdad847cb9b1f3e84f2613aa6 : SUCCESS in 56m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/021c810a792743728d19fd1ba02c72a8 : SUCCESS in 1h 07m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bd04cc73b97042ec815abac236b911f3 : SUCCESS in 17m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5eee4e6925d74aa5be4b3c158bf950cb : SUCCESS in 13m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ad1009daef434b9b8351c0187c94e8e0 : SUCCESS in 2h 16m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce6dbc8d30f747e984c6853c62739d8e : SUCCESS in 1h 59m 27s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/56e3d58cab524716a58b53e466554bc1 : SUCCESS in 1h 21m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6d75fdf51b84d1992d347908eefbd3c : SUCCESS in 2h 24m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a42f27c8393f4f5c8c91ee033e6bc7e2 : SUCCESS in 20m 27s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/0ca49780938d40d4b643e66d4c12177f : SUCCESS in 15m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e2b81b399da423da9e3aa7c19bb4cc1 : SUCCESS in 9m 45s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/80786fb7e5f74d6e8896e70e1d811c82 : SUCCESS in 11m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d7f52846e76743faba17a2386746ec6b : SUCCESS in 34m 02s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/496f1d31458849a2adf74f79d6395da3 : SUCCESS in 34m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b73c65872764197b448bfab05fa1175 : SUCCESS in 23m 12s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/59407b2c22ec428790b63bc296f2f3d8 : SUCCESS in 30m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ab7eeae9b1046c0bfd31fb65c4f5388 : SUCCESS in 11m 46s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/0cc984e22352435b9cbeaafd38af2dea : SUCCESS in 14m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d71b0d683b554e4d9467c915fbd02f0c : SUCCESS in 11m 12s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/2885189fdc2b48b4b5510160447048ba : SUCCESS in 11m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c9ee78794b714cf6877887627048c5cf : SUCCESS in 11m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef176fa3a6bb4663a0ce026abc332984 : SUCCESS in 9m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ecbb7554dfb527003de31a533fbbf5b877d38f7b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-16 19:21:49.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b1382a1f0acc01aa7c4eeb96b6b5184b07081c80","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-16 19:31:18.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"98443820e0bd11356fea7ea3aebbfe6b93568f38","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-16 19:32:05.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"1cf3e6ea1ca58c764b258c70d1d04bbe5f2605aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 21:38:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bccf2a1e626f4a5a8a631f3059880f84\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eeca32ddde244b1995a2bef3e344a38a : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd2700f5a2241869354087ea74e5c35 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1aaf120f382e4afa995346255b05466d : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2aea923b89b4e63a75b98519f64e977 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cccf2d82a5114c97a54564e9aa58f635 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d7c25a60c4440a3941f7ef24f24b238 : SUCCESS in 7m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d55e1cb77afe4354ba24d20330713936 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07c8464537154ec5b8c51cd7cf09a020 : SUCCESS in 7m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1cecdc926c0428dbe149b87e35463f5 : SUCCESS in 10m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/408730b0c13a4bc58b4029e888fe6c71 : SUCCESS in 30m 39s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/103d27ec2e41449a830029036cf91b50 : SUCCESS in 55m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa60ac5d07e14a7c9e3736d2710c9935 : SUCCESS in 28m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/51bfc8c9680542979e5c51f5bd8f8f6a : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/31e1efcacf3c4009b8d3f3b8b479a4d0 : SUCCESS in 1h 00m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4926ed37769e474b982e83495cf6af5e : SUCCESS in 15m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af828174382c425f97e86d3a20976891 : SUCCESS in 11m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/675c4545308e4aac94f48230c6a451ff : SUCCESS in 1h 57m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5c4a9ea9e9d48599ad0070b9098eed2 : SUCCESS in 2h 06m 11s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/62957c59200a463287b8b128614a856c : SUCCESS in 1h 25m 02s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f88cede22e5444ebd8db3d0f9a982c4 : SUCCESS in 2h 16m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/19fd7fa385c84aa69f7698c90348c750 : SUCCESS in 12m 14s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1ec5435167694f7ea7c3ec8d1df19f93 : SUCCESS in 14m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bb724c9d61248c59f8cf073906da9a7 : SUCCESS in 15m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/870de2b295cb489dbef011eefa22d02c : SUCCESS in 14m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/21dba6a6cd2b409fbfae74c4dde20a19 : SUCCESS in 18m 51s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/15e3b97e132541b88c0f320755784787 : SUCCESS in 23m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d38e85898e6c4930941faabfb5ebdd32 : SUCCESS in 31m 35s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7f86a7bb19b84e6cad2243b4164da9b1 : SUCCESS in 32m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bd545f8035f74033b80bd04b4b1d36d5 : SUCCESS in 12m 10s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/fc9cab73caa945419fce50375ded6bb8 : SUCCESS in 9m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b7d6b4d4a0c496bad7c852a237905d6 : SUCCESS in 5m 36s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/7860a6a222c445f8b2a2f1e132b3726a : SUCCESS in 11m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ccb1cf849b614b0f815d096f105e90c1 : SUCCESS in 11m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0b80b5d6d3e47fea8e73dcb3ff316cc : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":2},{"id":"c38e4a2091025998b4310d0c7d2329338ad17564","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-19 08:22:15.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d49a93d2ec7dd813f6dc4263dce76ff10d26de4d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-19 08:27:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca1596d957b177a624bab9369c1f83bf75579e16","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 10:47:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bff9aec64ac679ed525f86f02f27139f44eeaa87","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-19 13:48:52.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7a470c623f1813a1e63e6e878f3447a9e033ca2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-19 17:58:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f6d4488022d12910b571ddf309c25746fe161a43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 17:58:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"acf4fa75afcb9a98710714103d996aea8530092d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-19 18:36:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cddf79f939260287610fc8621628cd3a500ec7c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 20:24:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a801921ac7e74b149c288dc6b5e7a590\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe9a25cd08874422a5d0e00381f3c87c : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f11c1063907423aa06d84ebb431aef1 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8705d5aa3504721bd06f33045ae1f05 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ea32c369ca44ba8f92b8f2e22eb05c : SUCCESS in 9m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fdd749641aa84174b539f4aea7565e36 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/39e5e4bff0bd4a10bc37e1700694323f : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c82a1a288137412083cb2aa51d5c5bd0 : SUCCESS in 11m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/77133c649b7742fb8c2ccacb1e2771af : SUCCESS in 59m 55s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/55b5cecb6bb64dbc9aba5e00bc7cabd7 : SUCCESS in 38m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c92b4c91857a43e3b486b2a98d5acd80 : SUCCESS in 58m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c64178101f80480dbbdf7425a10780b2 : SUCCESS in 58m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/745daeccf67e482e8466609c8cfa7358 : SUCCESS in 1h 05m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/33dc7876da4048e0adc2688aeedd11c3 : SUCCESS in 18m 54s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5a3a83fb98749408444fa5275178d83 : SUCCESS in 12m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/77e6401acc764d08888e8ed9af211f5a : SUCCESS in 2h 18m 20s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ff2907a100e42fc8dd9800d0e714ebc : SUCCESS in 2h 23m 22s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/516209259554462e8196678deaedc6b0 : SUCCESS in 2h 03m 48s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4bc9816d64b4956a09c74d6d1a8669d : SUCCESS in 1h 51m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/44a062e3f80d4a7a8aae6e87873d1250 : SUCCESS in 17m 42s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/e33797e94d3f480db8b968016df44923 : SUCCESS in 15m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41cae87225b24ba4b1bec495949fd418 : SUCCESS in 10m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/61d8d9f4cc7242b0a5894c04c275800d : SUCCESS in 14m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/45c664f74a044aecafbc7e2419081f9d : SUCCESS in 38m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b555204b292448508c6c11d2310f9518 : SUCCESS in 21m 22s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bdf59b2e9dfc43f498cad52881ab3f23 : SUCCESS in 23m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/57c28ff276804afb815d5522e9191224 : SUCCESS in 15m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aedd680bc4314966849fd10082e88479 : SUCCESS in 10m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/079b3324ad264ef18d02b515cbd5cf06 : SUCCESS in 12m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8ec80f5aa754a189569271787e46067 : SUCCESS in 11m 04s","accounts_in_message":[],"_revision_number":2},{"id":"17b89cc7c9764781c5577a2bcf9d069b1f55d264","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 20:24:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"773a7cbe150ebe2c81b0a1c993de3615efc55701","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 20:25:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a7dff84772d4aa7af433dff7e4b94be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a2551a429946a3ad6e4a4f336bb9eb : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9172c7784e00a4156d13deef4795b5064fa1c75b","revisions":{"19dd74cc9f257b7903e9357a4e1fb6da75ce03b5":{"kind":"REWORK","_number":1,"created":"2025-05-16 16:32:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/950203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/950203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/950203/1"}}},"commit":{"parents":[{"commit":"b4733da80a3c532cbd45b76494f72eef06a9c8ba","subject":"Add the ability to override the system\u0027s default repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4733da80a3c532cbd45b76494f72eef06a9c8ba"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-16 16:32:17.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-16 16:32:17.000000000","tz":120},"subject":"Fix execution overcloud deprovision","message":"Fix execution overcloud deprovision\n\nWith `kayobe overcloud deprovision`, the task `Get PXE MAC address` runs\nthe command `openstack baremetal port list` which fails with:\n\n    Traceback (most recent call last):\n      File \"/var/lib/kolla/venv/bin/openstack\", line 5, in \u003cmodule\u003e\n        from openstackclient.shell import main\n    ModuleNotFoundError: No module named \u0027openstackclient\u0027\n\nFix by using the `baremetal port list` command instead.\n\nChange-Id: If1157a56110e26c3668bd54cd0293a9d3f936fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/19dd74cc9f257b7903e9357a4e1fb6da75ce03b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/19dd74cc9f257b7903e9357a4e1fb6da75ce03b5"}]},"branch":"refs/heads/master"},"9172c7784e00a4156d13deef4795b5064fa1c75b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-16 19:21:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/950203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/950203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/950203/2"}}},"commit":{"parents":[{"commit":"b4733da80a3c532cbd45b76494f72eef06a9c8ba","subject":"Add the ability to override the system\u0027s default repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4733da80a3c532cbd45b76494f72eef06a9c8ba"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-16 19:21:09.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-16 19:21:09.000000000","tz":120},"subject":"Fix execution of kayobe overcloud deprovision","message":"Fix execution of kayobe overcloud deprovision\n\nWith `kayobe overcloud deprovision`, the task `Get PXE MAC address` runs\nthe command `openstack baremetal port list` which fails with:\n\n    Traceback (most recent call last):\n      File \"/var/lib/kolla/venv/bin/openstack\", line 5, in \u003cmodule\u003e\n        from openstackclient.shell import main\n    ModuleNotFoundError: No module named \u0027openstackclient\u0027\n\nFix by using the `baremetal port list` command instead. This does not\nappear to affect 2024.2 and earlier.\n\nChange-Id: If1157a56110e26c3668bd54cd0293a9d3f936fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9172c7784e00a4156d13deef4795b5064fa1c75b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9172c7784e00a4156d13deef4795b5064fa1c75b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
