)]}'
{"id":"openstack%2Fkayobe~950358","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~Ia8777550b00bd1cf0391a1c7bddbbf99fe7c0d20","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-20 10:46:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8777550b00bd1cf0391a1c7bddbbf99fe7c0d20","subject":"Fix authentication failures due to missing system scope","status":"MERGED","created":"2025-05-19 19:20:49.000000000","updated":"2025-05-20 10:47:17.000000000","submitted":"2025-05-20 10:46:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950358","cherry_pick_of_change":949946,"cherry_pick_of_patch_set":5,"meta_rev_id":"79fb0577012c47fe6a476102d4716c83fc7cd08e","_number":950358,"virtual_id_number":950358,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-20 10:46:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-05-19 19:37:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-20 06:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-05-20 08:39:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-05-20 08:39:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-19 19:20:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"state":"CC"},{"updated":"2025-05-19 19:37:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-19 21:16:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-20 08:39:27.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"9dc50503bdeed3a1d7dc71c091a4486c915e5549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 19:20:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e7a3b6c8c07318e36d92b0b47c5bfffc022e9e62","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-19 19:37:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f7c227ed3fd1f63abb0008bad59cf283567e7c60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 21:16:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13b887c262b14883a33b687236799538\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/927474153da94b5f91a777f1fc0a4e26 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af174d6ae5bf492788ad2efbcc1c41c3 : SUCCESS in 1m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a11e280d10643159ac4fae6f7fbacf2 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b904b52fac5c4a55877359862872a14b : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de7b6ba944134713bd7684c6bff3bb54 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a96f41e5bdf4e7c84a510d28178ec65 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cef5538b203f4e0c8c9fdb0cbae6aa27 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28b2cce4b1034a638851c0b25599556f : SUCCESS in 10m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91db65d1f6dd4f46912e31597e393859 : SUCCESS in 7m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/30369ed6b822497cb6629c3ea121615b : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2511b9bb82cd4522870b2e39783ac129 : SUCCESS in 1h 06m 33s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bff1553139ff44d78945f5ad55a0076d : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a44d1cf40716492ca51dc799d9abc7e5 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ccc9cb44f51f4aeea29ec9ebd2f580b5 : SUCCESS in 18m 05s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8de1e70c398348dfb3b1e51ccad35b3d : SUCCESS in 18m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d0f4fceb4ec4030ad33ba136e4ee90b : SUCCESS in 18m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a380d679658c45dbba621e1c2c4a111d : SUCCESS in 1h 50m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e8029682b564cc1b549ac89e802380e : SUCCESS in 59m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dc9f087c745f4bd894cae02bb684e6bc : SUCCESS in 21m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77480901a9fa42fb9697e1bce1ea81f9 : SUCCESS in 12m 42s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2712294a04314312bedcc8230c9dbc08 : SUCCESS in 44m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f537a257262743f88361ce23498ca4f2 : SUCCESS in 32m 10s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d44173e1637456d968383eb20a8873f : SUCCESS in 27m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5178bb835d2c42a4991313bf9fa31d8d : SUCCESS in 16m 47s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/419d6fdf873746e4845bab6cab0fd913 : SUCCESS in 16m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01d64a1a50234018b7541c0e6562f29e : SUCCESS in 10m 06s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/b52008db304d4b6b9744b790dd6dfdfe : SUCCESS in 15m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1f8d4c2f3ce94e31839b34258d967d49 : SUCCESS in 10m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0c1fef874dc4ee1a3047ff7c33fe29b : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ef685267238c4c7bac6cbf8dbacd0c8959090578","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-20 06:28:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2fa0ace0c5a8a0f763240f1822ea4c885c702e85","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-20 08:39:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"756782585c17a2639d0b302edacd1be088e89cd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 08:39:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7d0decb49da5d39694801d6ec42089931e50537","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 10:46:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47ceea8330da468e8c0c72ddec0d0c38\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35fdd39ef9344b98865e26c7b877805 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86422181442042d6971224fd50f21e99 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/541ef8970abe4b7a994944c215964c7a : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77954e95d53a4b64899643fa04094198 : SUCCESS in 1m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/995e736c9a3e4299afc5001b36889b2d : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7bbdb02757904fa789cd63ac2a2b064d : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/62b5b94859454dc7ad9ab42ed8389a96 : SUCCESS in 10m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/003d53ef1d2b408f86506f7bd9630261 : SUCCESS in 1h 00m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a23547988fa54ffe8383c84a0d21f006 : SUCCESS in 1h 04m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/61782b416294400fbf2057041f8bb79e : SUCCESS in 1h 11m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b4063ca41b10436c8b6a6d03c3852f36 : SUCCESS in 17m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19f2594529a64dbdbaf69436c7d0bcf1 : SUCCESS in 9m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/59674b858edd477eaeefcbe3e901e05f : SUCCESS in 1h 55m 49s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1246e45f1e0e4f9daf992fbfdf8353ed : SUCCESS in 1h 51m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1a216cbef219493b82fd2f5dfa992511 : SUCCESS in 20m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38ba88c614ae4177abc0dead73e3bbf8 : SUCCESS in 20m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f67b8ed2c750487986209bee7acd9217 : SUCCESS in 32m 11s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54c4fc0db11a4660840ab13734306717 : SUCCESS in 25m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/209f8375ec004c3f83a0aa8236c2fd4a : SUCCESS in 20m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/783df91d4615454f9968476bfdce0d24 : SUCCESS in 14m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f766fd95ec01469690fc926828964945 : SUCCESS in 17m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7420337219cf4cd6b48f02d4910012f4 : SUCCESS in 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1c55be8bf35d899a01a164e41325e3154d4aeeae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 10:46:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"79fb0577012c47fe6a476102d4716c83fc7cd08e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 10:47:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ceb7573c4e54017a01acbf6cc9d62cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2d5bbc74f34280b03d3b8304bbd57b : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c98a966ec61548e88e8f2b8c394aff24 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb444bd9667f3575b66d4d08bdb2bdf81bf00752","revisions":{"eb444bd9667f3575b66d4d08bdb2bdf81bf00752":{"kind":"REWORK","_number":1,"created":"2025-05-19 19:20:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/58/950358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/58/950358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/950358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/950358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/58/950358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/58/950358/1"}}},"commit":{"parents":[{"commit":"5fc01870def4bdc0b1a0503464b55fa7e2eefe9e","subject":"Add the ability to override the system\u0027s default repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fc01870def4bdc0b1a0503464b55fa7e2eefe9e"}]}],"author":{"name":"Jake Hutchinson","email":"jake@stackhpc.com","date":"2025-05-15 20:30:32.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-19 19:20:49.000000000","tz":0},"subject":"Fix authentication failures due to missing system scope","message":"Fix authentication failures due to missing system scope\n\nEnsure the OS_SYSTEM_SCOPE environment variable is present in\nopenstack_auth to prevent authentication issues occuring in\nbaremetal-compute playbooks.\n\nCloses-Bug: #2111103\nChange-Id: Ia8777550b00bd1cf0391a1c7bddbbf99fe7c0d20\n(cherry picked from commit 73d57e5930191ccf8149498e0f93ac4c711b904b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb444bd9667f3575b66d4d08bdb2bdf81bf00752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb444bd9667f3575b66d4d08bdb2bdf81bf00752"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
