)]}'
{"id":"openstack%2Fkayobe~950512","triplet_id":"openstack%2Fkayobe~master~I79ba209ad2d9b444360cac34da3d0ed5db45bbd0","project":"openstack/kayobe","branch":"master","topic":"kayobe-release-2025.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-05-27 13:28:35.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-27 13:28:35.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-05-27 13:28:35.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-21 12:04:09.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2025-05-27 13:28:35.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-05-27 07:23:12.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-27 13:28:35.000000000","reason":"Change was submitted"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2025-05-27 13:28:35.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-05-21 10:38:59.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I79ba209ad2d9b444360cac34da3d0ed5db45bbd0","subject":"Revert \"[release] Use OpenStack 2025.1 release\"","status":"MERGED","created":"2025-05-21 10:14:06.000000000","updated":"2025-05-27 13:30:48.000000000","submitted":"2025-05-27 13:28:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"revert_of":949499,"submission_id":"950512-kayobe-release-2025.1","meta_rev_id":"a9df7318c3e20f1b355e150280a0d538d87b5ef4","_number":950512,"virtual_id_number":950512,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-27 13:28:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-05-26 16:41:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-27 07:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2025-05-27 11:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2025-05-27 11:02:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-21 10:14:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-05-21 10:14:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-05-21 10:14:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-05-21 10:14:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-05-21 10:14:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-21 10:14:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-05-21 10:14:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-05-21 10:14:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-05-21 10:14:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75d25b8082ad71f400b79b15a257cb1951411e9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-21 10:14:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41a5983ba607062e928e70d4cb74ecc550574280","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-21 10:38:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b1fc8fae1ecebcbdb1fca8b0039fb0fa1036a0d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-21 12:04:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8876a1c9b67d84decce2dec09c471b5df8789a3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 12:27:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22ee71fa29ce45c4abf5ae0027ca6214\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17fc5c31f87b44b0b48c346b92fb179f : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/523cc386732d404c809dc4552647c76e : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1aa20c56145b4e1b9a12d4f37871d769 : SUCCESS in 2m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7891c18f90647b183d3996166ae814f : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bee83b633624249ac6dc70c955fa13b : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65f8e0f37954e4cbb81d174e63acf06 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ca7472481f14121a9cb10362e0663f4 : SUCCESS in 2m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75cd9456140e451587d2fe19567b1562 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5e23db0d1a90493bb378697d58748be1 : SUCCESS in 9m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/65589e781e31421fba694fb0c58c86ec : SUCCESS in 10m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9d5fde7f7d7e41619ab0c57cc7c714fa : SUCCESS in 57m 10s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/00df8b76ee6e42069f9558d6852b329c : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a805a79029c4dd9a16d298f0dc08b6e : SUCCESS in 56m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2948f02591cf45f3b392698a62c8a1f5 : SUCCESS in 49m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3ab1b83f664c469bba677fe64af55e92 : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/975f3892edb84bb6b4fdef5fc300945e : SUCCESS in 16m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/961d3076e70f4d9c995bdcfdabfab61c : SUCCESS in 15m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1e5b484536ca4506b56d37e809454073 : FAILURE in 8m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4adcde9f915d47bdada73f13fb213c8d : SUCCESS in 2h 00m 34s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0f8c22576a4842bb8a8fef93769c13f8 : SUCCESS in 2h 09m 54s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fee1bb0fe4b42a7a99175f8703d671a : SUCCESS in 2h 08m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5156765dd8d54586839ea7a3869d0237 : SUCCESS in 15m 49s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8a35b4f9078442fb83f52667bd63685a : SUCCESS in 16m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29a4309cee1b4c7d80240d026adb5bf8 : SUCCESS in 15m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fcd5b1dc46044a698b8c3d7dff4303a0 : SUCCESS in 14m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a95066ac648044e1acacc1dcf0348903 : SUCCESS in 39m 31s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/75e3f7dd02fd4c56841d51f59ba2b44f : SUCCESS in 32m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bca06ecd11f4844a8d296026b3bf0a8 : SUCCESS in 27m 25s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1edca22363b548e08e687ac6123b4534 : SUCCESS in 33m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/704dc24ed3c346208422c0789ad0f51b : SUCCESS in 14m 59s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3b1be565d0374cd6bc51087bcead73b6 : SUCCESS in 13m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48fa3a479978442d8cb3cb7d2cc808a7 : SUCCESS in 11m 21s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/518c2c3e572d49e39e0012ddce02470d : SUCCESS in 11m 04s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1a0e4d8f29c84c108fb61aa5d574f45b : SUCCESS in 13m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/574ad8d49e4b48e08a61efa66c22d224 : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b7a1b67a210bb098aa8c9d81f4ca8b43efaf8664","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-21 13:33:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84747227fe31045eccf9fa12c0638892c04e7975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 15:54:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/273c618fad8c4d94981dd6b8e22d33f7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57438e42066446db8f7258386a527231 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58b648a2a43748c8bbe47b02efdfb248 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c2265869ad84fceb9a9e06e7ac0e24c : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68fc9744a5674b2e970654fe8d96a6eb : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b72506cb34d64ce38f7e131ddb9d0a7b : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c5876c242144dd9c708d0eee43907a : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef4a577ae6b64d588f7ffa3e985c9923 : SUCCESS in 2m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea27ed32b6844ceca320429778ef05c3 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a39693cdbac943bf9dd30739e665303c : SUCCESS in 5m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e5555411d704dfda438998b6d17bcef : SUCCESS in 9m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/547cc0cee4d94dba83cb5249d14ef952 : FAILURE in 12m 47s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/95aa9b424aa0418e97976b64485b87f0 : FAILURE in 12m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/180b9efdc67e41518f26935fcec7ebfc : FAILURE in 11m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a63893ddf47743b08496256057d171b1 : FAILURE in 6m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/04047ea3459c49189a06556aa4069fc3 : FAILURE in 16m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/746899cb725f46d2b137d09a5735329b : SUCCESS in 19m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a3eb6a5bfb04f3182a2323035d26f18 : SUCCESS in 12m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8df3a051077042de8ce08cc571f9e9ae : SUCCESS in 2h 15m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c481cd649101433288e6beade689230d : SUCCESS in 2h 12m 40s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/59c08db743624fd7b6c746c6ff9f71eb : SUCCESS in 2h 14m 49s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b65e61714bdb45f2aa54f84d62271dc4 : SUCCESS in 2h 06m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4d5642ec818c4dac9617d5a25861f795 : FAILURE in 11m 51s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6c7199110ace4aaabf962d66a40d43b3 : FAILURE in 11m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf10ab2f5a0e49aa9611d9d14d356b27 : FAILURE in 4m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a3f42fce45664103a329c50d694a59a6 : FAILURE in 10m 58s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4a90425851874369af14c41cf1cc0cf8 : FAILURE in 11m 27s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7e5efb4b51704675986262f127a7171b : SUCCESS in 31m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b7dff09b62c4082b8fd8b2b1818d878 : SUCCESS in 31m 37s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0e70a51256344c35a8f4a9f61b1753df : SUCCESS in 36m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/42480047237642238b8cb638a25f78dd : SUCCESS in 12m 15s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/866a788db09d4754a10e7c81c0714122 : SUCCESS in 13m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ce8c16cdb0349ffb271b44a493a48f4 : SUCCESS in 11m 34s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/5a52bbf20dfe42c39d08a016c0b9f898 : SUCCESS in 8m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/150c784ea0df412e855b2c67b6e7cc5c : SUCCESS in 12m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bb42a990be34e4bb2f2b2f9193fa1a6 : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9c0bfa0e03f70ae3656ebb51ba92b1c53142ac28","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-22 07:39:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c63f02f1d25c0ba72043610dff7d1b018bc47a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 10:09:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fe9009c62f24e1c9f08f63d6baaefdb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cf2c85bf41649f08d36d93b70c766a1 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e442bd87c334875ba6020b2a9c77cbf : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30e4337ae1344041a0aacc4ab2ddda7e : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/53601314ce974fceb507a321f1f85c9a : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26c5b9f8776647c0bfcde26b0b448c51 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97353a4fc8e84d2396a748766674719b : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca43a20fdcb54aca95aa67573199fc2c : SUCCESS in 2m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5dadc00d015342ee8a8c9a6d9007a738 : SUCCESS in 2m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2a99fce677314eb3a1b58590294780bb : SUCCESS in 10m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4f7482c9d66446688c2647c1f09e9a9e : SUCCESS in 9m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/984891ebff1d443295e6d77e1555231b : FAILURE in 12m 53s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/aa7870cf563b43d49224a3b9a16e79f1 : FAILURE in 13m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1283fac24df416c8973a5503fcb6417 : FAILURE in 13m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cf3ebbbb6e3c410cb7728ebe7593c7ce : FAILURE in 11m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/46d0d2cc9ea14980a1340d4466958eed : FAILURE in 15m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/de2aae1fd242486baa0eb41b15acc7df : SUCCESS in 15m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7137c804b95d4e7db9c830f6875595ac : SUCCESS in 12m 28s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6de283f92b2745b4bfc12bdfbe560f3e : SUCCESS in 2h 12m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/555da93569cf446b85ec00f3ed957dac : SUCCESS in 2h 27m 17s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/eb2687f89f244ff4978fe07421043341 : SUCCESS in 2h 05m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b55ecb6caf6401e8c231f000e482666 : SUCCESS in 2h 23m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/16525966c8634d81badc7159d4b89823 : FAILURE in 12m 32s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5da28a21016049ab9a7a3148ecb54fa2 : FAILURE in 9m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5335acd852f14057a1b9574f8d2ca096 : FAILURE in 8m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e1ce5be38d8f409a899efa479db10a3e : FAILURE in 8m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/68786530105b420ca857858167c79e2f : FAILURE in 11m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/515cc59c09954620a5a4d71e955761d5 : SUCCESS in 39m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5db4019441614f3bbedf8ce27256cc80 : SUCCESS in 37m 25s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d22a06bc73dc43e88fc9d07cfe63c5b2 : SUCCESS in 34m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0674a64507eb4830bb69eba74e11e341 : SUCCESS in 15m 12s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/9275597697844de38cc162bdaf89cf07 : SUCCESS in 10m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/038b4066100d4e5abd393b2acd96247d : SUCCESS in 10m 33s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/02e5d549189944149d230d08503c3518 : SUCCESS in 10m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/75a49a5fbe2741e483de12f9f8f1e685 : SUCCESS in 13m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ea425867c084f92a0e739b367f325cd : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b2deaac32a0cc30f6031a4987bc729a05b830fe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-26 08:17:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f4e25a459651016ebc6da31dec2f26815faa82b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 10:42:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b703241c947f45b4b784c5ce78f6f16f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2177619b19fd483094c2d335a5bab2e4 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15af0d0208e247aebb981b568b7ee52b : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66c4e6d110734ea49e6033cd6f7b8f9e : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/866165342914431f8fae364810a2898b : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/943e998c88db431ebda7c3e2688e9c42 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80cc5e8811b54d88927afcf76fe8bb62 : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dfffcba213342558b18ed5b22a73e9d : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d9dee43aef634ac29b93f7bc9cee7b43 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e87d4535e1a14cd08f0fac61818df7ed : SUCCESS in 12m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d309fb6e3b064e46900220ddcefc6d9a : SUCCESS in 9m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d7b7398964ca443c9e97d6de2ba1e97a : SUCCESS in 56m 50s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/ddf65b27c79247af8a700ae3b90c9956 : SUCCESS in 1h 01m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35e734e859bf48bb85d9ca38edd048f0 : SUCCESS in 1h 07m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9fe29944814644dea238be07655450e6 : SUCCESS in 1h 05m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7ce355dac5484634a057ec2655d28b08 : SUCCESS in 1h 12m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e33c66b9326047189624716cba5bdbe8 : SUCCESS in 17m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/911dfe0ca76544b09b8b443820753a04 : SUCCESS in 12m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9286dddd39d548cb8aaf585c610a3953 : SUCCESS in 2h 09m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/407cb671fdd246d1845998ba3db62474 : SUCCESS in 2h 04m 31s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1d6362267b964a69b14d14c308ec22a2 : SUCCESS in 2h 20m 33s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a69d54ef04b49d9a21d4a1c29df2ea8 : SUCCESS in 2h 20m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6fa4f20bd9c84a0b9be02cc4a4c25d4a : SUCCESS in 15m 29s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/40047d813a5e463e919967b04eb8fb89 : SUCCESS in 19m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/502c87fe8b214818b06068374d7fcf5c : SUCCESS in 16m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a3180421e21044738dabdb6dc4a31902 : SUCCESS in 13m 21s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/245a132622a64e09981cee39efaf2fd5 : SUCCESS in 34m 29s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ed4c0b37d8e044468c158f56aef984e5 : SUCCESS in 33m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efb91544a05a475ea0e68c0a096a0693 : SUCCESS in 25m 00s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/07cd32a4dab34e549507ce0d63fcacef : SUCCESS in 38m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a44647ae84e642f29f642481e321515b : SUCCESS in 14m 03s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/d722302655914689b3b7ce1728c7a47c : SUCCESS in 13m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68293723bfae4a4ba2e53e169229862b : SUCCESS in 12m 43s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/83e25444f4734b458831cafd303d72cb : SUCCESS in 10m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d7d4d64a2bf040a89465f9d64e3aa624 : SUCCESS in 11m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b8172a7d5764b49a3337ee9dff02fd6 : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":2},{"id":"2801a4797aedef644d026e06251ebf2a59e6078c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-26 16:41:18.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"18c6f8b71fd9278e894b62c12a8bca5c3d317927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 19:21:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dcd15f9005841b59b5f3f9d0ba3d758\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce079a55405c4377affb0a427101ee39 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c155d74f119a4079ba330f02685c71e8 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95a88efe04f84aed914e26e4db8f5bcf : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b2f47ae01934bc2bc2c64efe8ab1f5e : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69c14e22ef12478f93980d1e5d4dc2af : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db3a51489e104e378867ee267eb3d31b : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67d0a7663d6348c18d1a471e64193ef3 : SUCCESS in 2m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/749e63def8f2406aae1e4baf752bda7d : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/625f964578804bd8b0fef650250f3d20 : SUCCESS in 9m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/40cf624a9fab4f0b90b8cf4535fe1442 : SUCCESS in 8m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2943490a17974fed923cda013f5c9b9c : SUCCESS in 1h 02m 03s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/51c21cbbcca348a09f73215f07f8e8db : SUCCESS in 49m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/912c3a4219e14c6b92efdf2d8fa604de : SUCCESS in 58m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/61e064d7333f45279df24b59ad9ed623 : SUCCESS in 1h 00m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/16beac4b5e784d2eb83fde422f4babc5 : SUCCESS in 52m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b0f6bd360ec0404b9b4f76df1cfeab8d : SUCCESS in 19m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40262da6ab5d41bbac4acb0aa6f20d96 : SUCCESS in 13m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0af8c985b02d41e68e18e8b43b465c3a : SUCCESS in 1h 53m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e47f61bf3a04471a5e349374cc7704e : SUCCESS in 1h 54m 22s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ca30773f74224a07bb1197c075c4d9fd : SUCCESS in 2h 29m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce4697e0c46e41f288bf32916c73a826 : SUCCESS in 1h 48m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3dd140f32f954c03b5b9bfb3d1e3421a : SUCCESS in 19m 16s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/e9921a407b97477e8f8fdb451517a523 : SUCCESS in 15m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c93e1645d47543d98f97c6d46fb3239b : SUCCESS in 14m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7184f46bc6334e0d9cfc1b90099bbca8 : SUCCESS in 13m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4811a963b1ce437591459903f33e579e : SUCCESS in 45m 36s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/962f6892141f40fa92796047b64bf9d6 : SUCCESS in 32m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b121f1ca66e4090b5886f2db91debbc : SUCCESS in 29m 24s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7d8fd6e243ed4b228e72ebe0eb15c186 : SUCCESS in 37m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2aeede1f246d4cef80f13b4bdbae5702 : SUCCESS in 14m 49s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b8d47d6536e4464ba654cb7a4f8ba56c : SUCCESS in 12m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/092616cc350840f581144bc3ddc6d256 : SUCCESS in 10m 37s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/506efd0a890e416e87719a262bba5bca : SUCCESS in 10m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c4b47dc6dd704aa5b352c75706ce119d : SUCCESS in 11m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d01c5a12d52141a0868904eb97a7b5a2 : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":3},{"id":"29b6dc31c2bb159e9762336a15b6cc60946c4bf6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-05-27 07:23:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b60c9296745e1297a129205ef33564b3e2754634","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-27 11:02:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2977421972d4bef9a054d3444bc2145570c5c5f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 11:03:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d9663593276146a71bab8d7dc8382b80266635e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:28:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51d2389283b8406093719c3663af84d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c4e964034ff4ce394a787b83cb050e8 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59d84e44a9d8483cb56ae54a9adb011b : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8b87f5ea15f434c80e3a280dd9a326e : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30e45dfdbf984612a8b9a99e10c2eaa5 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/117749e943c14df7922023cee1e4a0f2 : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9312fb6aff344ce3b65224d705c100f1 : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ddb1a3fca3864aa192ee7f9833c2f472 : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f786279619f94c0dae0485aa3e4e69c7 : SUCCESS in 9m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9bec3adec6c04b539f2566a0fef49119 : SUCCESS in 51m 52s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/ed899e2ed1354c60af271467063f0373 : SUCCESS in 54m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2476394a3b8241058a90eff179ed35a0 : SUCCESS in 51m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3d30eeeda63945318bf55f6ac2f4b412 : SUCCESS in 47m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/df15b56e7ecf4ced8ce984c9ba1da0b6 : SUCCESS in 1h 05m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/edfc3563d2374a01bf781538c2878340 : SUCCESS in 18m 02s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd853158f3f74e33ba99b68ce00aff48 : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1d5db47d215f43008ad606b7b374fd34 : SUCCESS in 2h 15m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a52ba5599c74f0e8c9f5743e03685ab : SUCCESS in 2h 17m 56s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e0b6243ac15d4717b0cf3830039e9314 : SUCCESS in 1h 56m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2866aa31b4d54c7fb0b2145c942c86b3 : SUCCESS in 2h 06m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/60deda04f3fd4401b3b10a05dbcacbc5 : SUCCESS in 15m 53s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/41d2acebf4b34f98b2a17c5815a7b094 : SUCCESS in 18m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcd9647f886545b9aefdbfa5d4ee03ab : SUCCESS in 19m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2c26e64be98148af9b370d5fce6f167a : SUCCESS in 14m 01s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b438c5c338664edc95840924f7b6bfd0 : SUCCESS in 34m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd90ced821e6446a97a07ed7f10262ff : SUCCESS in 37m 20s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c81e5996ca54402e864936fa6e88c839 : SUCCESS in 34m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/99e3d59041844c3091b41291d9ffead1 : SUCCESS in 13m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b01decf0ea634c328bfed463454fd68c : SUCCESS in 11m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea7500f59fa34b35a75164d712616348 : SUCCESS in 14m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66d602a04f1e43ed823fd4d52e9e874e : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":3},{"id":"302f72a38e61d351bca4332b4f5c0965b81fa048","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:28:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a9df7318c3e20f1b355e150280a0d538d87b5ef4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:30:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ab85a31bd2b4640b5eb8db4b17da794\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3e1c83d3bd497db0d4a4dd3f527e7c : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02e8cf4a9e6140578caa132ab8bdaf07 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5b2317418f7ae207d9ebab686f94bbb693d6b5d3","revisions":{"f4f6a7ad811943ba4a8f55ae9e648bfa53239946":{"kind":"REWORK","_number":1,"created":"2025-05-21 10:14:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/12/950512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/950512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/950512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/950512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/950512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/950512/1"}}},"commit":{"parents":[{"commit":"3c916b3e95b403f6aef12dda32a055935c9be477","subject":"[release] Use OpenStack 2025.1 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c916b3e95b403f6aef12dda32a055935c9be477"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-21 10:14:06.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-05-21 10:14:06.000000000","tz":0},"subject":"Revert \"[release] Use OpenStack 2025.1 release\"","message":"Revert \"[release] Use OpenStack 2025.1 release\"\n\nThis reverts commit 3c916b3e95b403f6aef12dda32a055935c9be477.\n\nReason for revert: 2025.1 has been branched and released\n\nChange-Id: I79ba209ad2d9b444360cac34da3d0ed5db45bbd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4f6a7ad811943ba4a8f55ae9e648bfa53239946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4f6a7ad811943ba4a8f55ae9e648bfa53239946"}]},"branch":"refs/heads/master"},"764dea5933bd806f630d42fb986f58faddfee03b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-26 08:17:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/12/950512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/950512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/950512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/950512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/950512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/950512/2"}}},"commit":{"parents":[{"commit":"4e2220bb6cd2b0e6d6e8a2106113ec25db55c3a3","subject":"Fix compatibility with Kolla virtualenv configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e2220bb6cd2b0e6d6e8a2106113ec25db55c3a3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-21 10:14:06.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-26 08:17:46.000000000","tz":0},"subject":"Revert \"[release] Use OpenStack 2025.1 release\"","message":"Revert \"[release] Use OpenStack 2025.1 release\"\n\nThis reverts commit 3c916b3e95b403f6aef12dda32a055935c9be477.\n\nReason for revert: 2025.1 has been branched and released\n\nChange-Id: I79ba209ad2d9b444360cac34da3d0ed5db45bbd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/764dea5933bd806f630d42fb986f58faddfee03b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/764dea5933bd806f630d42fb986f58faddfee03b"}]},"branch":"refs/heads/master","description":"Rebase"},"5b2317418f7ae207d9ebab686f94bbb693d6b5d3":{"kind":"NO_CHANGE","_number":3,"created":"2025-05-26 16:41:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/12/950512/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/950512/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/950512/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/950512/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/950512/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/950512/3"}}},"commit":{"parents":[{"commit":"4e2220bb6cd2b0e6d6e8a2106113ec25db55c3a3","subject":"Fix compatibility with Kolla virtualenv configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e2220bb6cd2b0e6d6e8a2106113ec25db55c3a3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-21 10:14:06.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-21 10:14:06.000000000","tz":0},"subject":"Revert \"[release] Use OpenStack 2025.1 release\"","message":"Revert \"[release] Use OpenStack 2025.1 release\"\n\nThis reverts commit 3c916b3e95b403f6aef12dda32a055935c9be477.\n\nReason for revert: 2025.1 has been branched and released\n\nChange-Id: I79ba209ad2d9b444360cac34da3d0ed5db45bbd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b2317418f7ae207d9ebab686f94bbb693d6b5d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b2317418f7ae207d9ebab686f94bbb693d6b5d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
