)]}'
{"id":"openstack%2Fkayobe~950675","triplet_id":"openstack%2Fkayobe~master~I6e165f7c521eea7e3179142104c14fd07c1ee201","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-27 16:33:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6e165f7c521eea7e3179142104c14fd07c1ee201","subject":"CI: Add ansible-lint job (non-voting)","status":"ABANDONED","created":"2025-05-22 16:54:32.000000000","updated":"2025-05-27 16:33:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d96ec9da005c7aa465c35a22831c195cb23d9d5c","_number":950675,"virtual_id_number":950675,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-22 19:24:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-22 19:24:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fa022e3b4f5a41433c8e316031d756284811c54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-22 16:54:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c5133d3ef99a14b7b09589dc40e95eb575886e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 19:24:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/53260a9d87bb403ebe880685086a6ea3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88a7fea28be74882a809cf3e6008ba99 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93e42db51fa84022bd1ce8280ffef03e : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/baabe4e44ccf434dbba069a8c4045e94 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6129bf5deabd41f68731ee9555558023 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3d7c9ed63934b2ba3ed212edc87dfd2 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff4a8a7baab4b1dabffa6f2257f6289 : SUCCESS in 10m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/06e72a07293542c5b3301884146cd892 : SUCCESS in 11m 07s\n- kayobe-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/5b7d225b69d24a7d92e84272db787ad6 : FAILURE in 6m 23s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5bb4d8b18054471695243d3b54e4abfd : SUCCESS in 5m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a20aeb9fda6d49faab73bc8f5f4f39ee : SUCCESS in 9m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/86eed918c2114ef3bdc89f8593c3fb24 : FAILURE in 12m 09s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/b8d40d73cded4ba5a12938120a7eb900 : FAILURE in 12m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c004008f904b4907978df9e913030c3d : FAILURE in 10m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9689430c5159434baba70e0a552e83f0 : FAILURE in 10m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/44d8ee1d63b943efb5f8e9119d1a384e : FAILURE in 16m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f45e81d1cd1546028c888c50963230cf : SUCCESS in 15m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f07943a8a5574e1e8b73ef0dcb3fe545 : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/41ce6e1ef7514faf90d0b7881a1b6f5c : SUCCESS in 2h 03m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b23019677c2498b87f949ff76c70169 : SUCCESS in 2h 00m 23s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/33c98e7a375248f5b5e03d1844299ac5 : SUCCESS in 2h 14m 15s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e88f36f0dae4df0a4524b7aff966fa5 : SUCCESS in 2h 11m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d27adce9b22c49b2a65e44f3b28a3add : FAILURE in 14m 02s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/492fd7de315448d0a13788f74f4268b9 : FAILURE in 13m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ab2734000944809acafe48c62116763 : FAILURE in 9m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9012d316723b49ad835be9b073100661 : FAILURE in 9m 51s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0ce6122f913741f58a4d2e05755b153f : FAILURE in 11m 12s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/053ed73f7cc74533b8029bf1eff5b9c7 : SUCCESS in 34m 16s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a1d48b552464167b12f14679cd2bc94 : SUCCESS in 29m 49s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3443e540bacb489a85935650928f4172 : SUCCESS in 30m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8f34f1c5089c4edea98278abc6ce249c : SUCCESS in 12m 37s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1045571ffc4b4f009d7d0b69eb14a6d0 : SUCCESS in 12m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3f3b30b06774700ab1a492c11fc53e6 : SUCCESS in 13m 42s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/f482bc0492504cca94538dfe2a35f047 : SUCCESS in 10m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/32c0d87e2cfe4669af6e1751921bce51 : SUCCESS in 7m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad584180ab8b4349be397085742cdcce : SUCCESS in 10m 44s\n\nWarning:\n  Comments left for invalid file ansible/apparmor-libvirt.yml\n  Comments left for invalid file ansible/apt.yml\n  Comments left for invalid file ansible/baremetal-compute-inspect.yml\n  Comments left for invalid file ansible/baremetal-compute-introspection-data-save.yml\n  Comments left for invalid file ansible/baremetal-compute-register.yml\n  Comments left for invalid file ansible/baremetal-compute-rename.yml\n  Comments left for invalid file ansible/baremetal-compute-serial-console.yml\n  Comments left for invalid file ansible/compute-libvirt-host.yml\n  Comments left for invalid file ansible/compute-node-discovery.yml\n  Comments left for invalid file ansible/container-engine.yml\n  Comments left for invalid file ansible/container-image-build.yml\n  Comments left for invalid file ansible/container-image-builders-check.yml\n  Comments left for invalid file ansible/dell-compute-node-inventory.yml\n  Comments left for invalid file ansible/dnf.yml\n  Comments left for invalid file ansible/drac-bios.yml\n  Comments left for invalid file ansible/drac-boot-order.yml\n  Comments left for invalid file ansible/drac-facts.yml\n  Comments left for invalid file ansible/etc-hosts.yml\n  Comments left for invalid file ansible/firewall.yml\n  Comments left for invalid file ansible/host-command-run.yml\n  Comments left for invalid file ansible/host-package-update.yml\n  Comments left for invalid file ansible/infra-vm-deprovision.yml\n  Comments left for invalid file ansible/infra-vm-provision.yml\n  Comments left for invalid file ansible/ip-allocation.yml\n  Comments left for invalid file ansible/kayobe-ansible-user.yml\n  Comments left for invalid file ansible/kayobe-target-venv.yml\n  Comments left for invalid file ansible/kolla-ansible-user.yml\n  Comments left for invalid file ansible/kolla-ansible.yml\n  Comments left for invalid file ansible/kolla-bifrost-hostvars.yml\n  Comments left for invalid file ansible/kolla-openstack.yml\n  Comments left for invalid file ansible/kolla-packages.yml\n  Comments left for invalid file ansible/kolla-pip.yml\n  Comments left for invalid file ansible/kolla-target-venv.yml\n  Comments left for invalid file ansible/logging.yml\n  Comments left for invalid file ansible/luks.yml\n  Comments left for invalid file ansible/lvm.yml\n  Comments left for invalid file ansible/network-connectivity.yml\n  Comments left for invalid file ansible/network.yml\n  Comments left for invalid file ansible/overcloud-bios-raid.yml\n  Comments left for invalid file ansible/overcloud-deprovision.yml\n  Comments left for invalid file ansible/overcloud-etc-hosts-fixup.yml\n  Comments left for invalid file ansible/overcloud-facts-gather.yml\n  Comments left for invalid file ansible/overcloud-hardware-inspect.yml\n  Comments left for invalid file ansible/overcloud-host-image-build.yml\n  Comments left for invalid file ansible/overcloud-introspection-data-save.yml\n  Comments left for invalid file ansible/overcloud-introspection-rules-dell-lldp-workaround.yml\n  Comments left for invalid file ansible/overcloud-introspection-rules.yml\n  Comments left for invalid file ansible/overcloud-inventory-discover.yml\n  Comments left for invalid file ansible/overcloud-ipa-build.yml\n  Comments left for invalid file ansible/overcloud-ipa-images.yml\n  Comments left for invalid file ansible/overcloud-provision.yml\n  Comments left for invalid file ansible/overcloud-service-config-save.yml\n  Comments left for invalid file ansible/physical-network.yml\n  Comments left for invalid file ansible/provision-net.yml\n  Comments left for invalid file ansible/proxy.yml\n  Comments left for invalid file ansible/roles/apt/handlers/main.yml\n  Comments left for invalid file ansible/roles/apt/tasks/config.yml\n  Comments left for invalid file ansible/roles/apt/tasks/keys.yml\n  Comments left for invalid file ansible/roles/apt/tasks/main.yml\n  Comments left for invalid file ansible/roles/apt/tasks/preferences.yml\n  Comments left for invalid file ansible/roles/apt/tasks/proxy.yml\n  Comments left for invalid file ansible/roles/apt/tasks/repos.yml\n  Comments left for invalid file ansible/roles/arista-switch/tasks/main.yml\n  Comments left for invalid file ansible/roles/bootstrap/tasks/main.yml\n  Comments left for invalid file ansible/roles/console-allocation/tasks/main.yml\n  Comments left for invalid file ansible/roles/dell-switch-bmp/handlers/main.yml\n  Comments left for invalid file ansible/roles/dell-switch-bmp/tasks/main.yml\n  Comments left for invalid file ansible/roles/dev-tools/tasks/main.yml\n  Comments left for invalid file ansible/roles/disable-cloud-init/handlers/main.yml\n  Comments left for invalid file ansible/roles/disable-cloud-init/tasks/main.yml\n  Comments left for invalid file ansible/roles/disable-glean/handlers/main.yml\n  Comments left for invalid file ansible/roles/disable-glean/tasks/main.yml\n  Comments left for invalid file ansible/roles/dnf-automatic/tasks/main.yml\n  Comments left for invalid file ansible/roles/dnf/tasks/custom-repo.yml\n  Comments left for invalid file ansible/roles/dnf/tasks/local-mirror.yml\n  Comments left for invalid file ansible/roles/dnf/tasks/main.yml\n  Comments left for invalid file ansible/roles/docker-registry/tasks/config.yml\n  Comments left for invalid file ansible/roles/docker-registry/tasks/deploy.yml\n  Comments left for invalid file ansible/roles/docker-registry/tasks/main.yml\n  Comments left for invalid file ansible/roles/docker-registry/tasks/upgrade.yml\n  Comments left for invalid file ansible/roles/docker/tasks/main.yml\n  Comments left for invalid file ansible/roles/drac-boot-mode/tasks/main.yml\n  Comments left for invalid file ansible/roles/drac-pxe/tasks/main.yml\n  Comments left for invalid file ansible/roles/etc-hosts/tasks/etc-hosts.yml\n  Comments left for invalid file ansible/roles/etc-hosts/tasks/main.yml\n  Comments left for invalid file ansible/roles/firewalld/handlers/main.yml\n  Comments left for invalid file ansible/roles/firewalld/tasks/disabled.yml\n  Comments left for invalid file ansible/roles/firewalld/tasks/enabled.yml\n  Comments left for invalid file ansible/roles/firewalld/tasks/main.yml\n  Comments left for invalid file ansible/roles/gather-facts-delegated/defaults/main.yml\n  Comments left for invalid file ansible/roles/gather-facts-delegated/tasks/gather-facts-delegated.yml\n  Comments left for invalid file ansible/roles/gather-facts-delegated/tasks/main.yml\n  Comments left for invalid file ansible/roles/infra-vms/tasks/deploy.yml\n  Comments left for invalid file ansible/roles/infra-vms/tasks/destroy.yml\n  Comments left for invalid file ansible/roles/infra-vms/tasks/main.yml\n  Comments left for invalid file ansible/roles/infra-vms/tasks/prerequisites.yml\n  Comments left for invalid file ansible/roles/inspection-store/handlers/main.yml\n  Comments left for invalid file ansible/roles/inspection-store/tasks/config.yml\n  Comments left for invalid file ansible/roles/inspection-store/tasks/deploy.yml\n  Comments left for invalid file ansible/roles/inspection-store/tasks/main.yml\n  Comments left for invalid file ansible/roles/inspection-store/tasks/reconfigure.yml\n  Comments left for invalid file ansible/roles/inspection-store/tasks/upgrade.yml\n  Comments left for invalid file ansible/roles/ip-routing/tasks/main.yml\n  Comments left for invalid file ansible/roles/ipa-images/tasks/main.yml\n  Comments left for invalid file ansible/roles/ipa-images/tasks/set-driver-info.yml\n  Comments left for invalid file ansible/roles/kolla-ansible-host-vars/tasks/main.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/config.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/install.yml\n  Comments left for invalid file ansible/roles/kolla-ansible/tasks/main.yml\n  Comments left for invalid file ansible/roles/kolla-bifrost/tasks/main.yml\n  Comments left for invalid file ansible/roles/kolla-build/tasks/main.yml\n  Comments left for invalid file ansible/roles/kolla-openstack/defaults/main.yml\n  Comments left for invalid file ansible/roles/kolla-openstack/tasks/config.yml\n  Comments left for invalid file ansible/roles/kolla-openstack/tasks/main.yml\n  Comments left for invalid file ansible/roles/kolla/tasks/config.yml\n  Comments left for invalid file ansible/roles/kolla/tasks/install.yml\n  Comments left for invalid file ansible/roles/kolla/tasks/main.yml\n  Comments left for invalid file ansible/roles/manage-containers/tasks/deploy-container.yml\n  Comments left for invalid file ansible/roles/manage-containers/tasks/deploy.yml\n  Comments left for invalid file ansible/roles/manage-containers/tasks/destroy-container.yml\n  Comments left for invalid file ansible/roles/manage-containers/tasks/destroy.yml\n  Comments left for invalid file ansible/roles/manage-containers/tasks/main.yml\n  Comments left for invalid file ansible/roles/nclu-switch/tasks/main.yml\n  Comments left for invalid file ansible/roles/ntp/tasks/deploy.yml\n  Comments left for invalid file ansible/roles/ntp/tasks/main.yml\n  Comments left for invalid file ansible/roles/ntp/tasks/prepare.yml\n  Comments left for invalid file ansible/roles/ntp/tasks/sync.yml\n  Comments left for invalid file ansible/roles/ntp/tasks/validate.yml\n  Comments left for invalid file ansible/roles/opensm/tasks/main.yml\n  Comments left for invalid file ansible/roles/opensm/tasks/upgrade.yml\n  Comments left for invalid file ansible/roles/pip/tasks/main.yml\n  Comments left for invalid file ansible/roles/pip/tasks/pip_conf.yml\n  Comments left for invalid file ansible/roles/public-openrc/tasks/main.yml\n  Comments left for invalid file ansible/roles/selinux/tasks/main.yml\n  Comments left for invalid file ansible/roles/snat/tasks/main.yml\n  Comments left for invalid file ansible/roles/ssh-known-host/tasks/main.yml\n  Comments left for invalid file ansible/roles/swap/tasks/main.yml\n  Comments left for invalid file ansible/roles/swift-block-devices/tasks/main.yml\n  Comments left for invalid file ansible/roles/swift-rings/tasks/main.yml\n  Comments left for invalid file ansible/roles/sysctl/tasks/main.yml\n  Comments left for invalid file ansible/roles/wipe-disks/tasks/main.yml\n  Comments left for invalid file ansible/seed-credentials.yml\n  Comments left for invalid file ansible/seed-ipa-build.yml\n  Comments left for invalid file ansible/seed-service-upgrade-prep.yml\n  Comments left for invalid file ansible/seed-vm-provision.yml\n  Comments left for invalid file ansible/swap.yml\n  Comments left for invalid file ansible/swift-rings.yml\n  Comments left for invalid file ansible/time.yml\n  Comments left for invalid file ansible/vgpu.yml\n  Comments left for invalid file ansible/wipe-disks.yml","accounts_in_message":[],"_revision_number":1},{"id":"d96ec9da005c7aa465c35a22831c195cb23d9d5c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-27 16:33:19.000000000","message":"Abandoned\n\nReplaced by https://review.opendev.org/c/openstack/kayobe/+/951011","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5e31257d367950e9c1c096b65ebcfa882c032e1","revisions":{"b5e31257d367950e9c1c096b65ebcfa882c032e1":{"kind":"REWORK","_number":1,"created":"2025-05-22 16:54:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/75/950675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/950675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/950675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/950675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/950675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/950675/1"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-22 16:54:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-22 16:54:08.000000000","tz":120},"subject":"CI: Add ansible-lint job (non-voting)","message":"CI: Add ansible-lint job (non-voting)\n\nChange-Id: I6e165f7c521eea7e3179142104c14fd07c1ee201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5e31257d367950e9c1c096b65ebcfa882c032e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5e31257d367950e9c1c096b65ebcfa882c032e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
