)]}'
{"id":"openstack%2Fkayobe~950704","triplet_id":"openstack%2Fkayobe~master~I7cfb600eec268749ea1315112fe3a9429ab3ab8e","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-05-23 16:04:58.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"35010":{"account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"last_update":"2025-05-23 15:50:48.000000000","reason":"\u003cGERRIT_ACCOUNT_35010\u003e replied on the change","reason_account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-23 19:47:23.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-05-23 20:37:52.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-23 20:37:52.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-05-23 20:37:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7cfb600eec268749ea1315112fe3a9429ab3ab8e","subject":"Fix compatibility with Kolla virtualenv configuration","status":"MERGED","created":"2025-05-22 20:23:34.000000000","updated":"2025-05-23 20:38:42.000000000","submitted":"2025-05-23 20:37:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950704","meta_rev_id":"7486188196135a938ee15df256558d1c14554705","_number":950704,"virtual_id_number":950704,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-23 20:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-05-23 16:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-05-23 15:50:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":1,"date":"2025-05-23 19:47:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-23 15:45:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-05-23 16:04:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-05-23 16:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-23 15:45:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 01:36:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-23 04:33:03.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-05-23 06:34:24.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-05-23 06:34:24.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-05-23 10:21:20.000000000","updated_by":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"reviewer":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"state":"REVIEWER"},{"updated":"2025-05-23 10:31:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"c2e223aa11513070911f2ce9401a291bea45f547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-22 20:23:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82acf5e2478b1e4e1a01ba8c9d79596075e8b672","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-22 20:34:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b380e56a612a143eae6775bb5a14fde7e661cdd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-22 20:51:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0e2f0b3aacb47f55556fe3dbd4ca2b31465b68d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-22 21:03:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"20e56473947f21d11c098bda4ac7495e352672a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-22 21:32:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f9b0d9f767d38918782161cef93f6fc7d00e3e52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-22 21:45:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bf90d67ce92620284d3c95151dd53b9f9c26dbba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-22 23:11:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"36d7400d8b996122a54adf6f1d87e87bf915b8ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 01:36:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eaca6938bb04c208fbd570a61c1f16d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38fdc82a288e4b5d9687fcd5dd845f0a : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f654f8e61ad24c069e7907335ed1b63f : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6b32c9e0cc846ccb9e415599ec6f01c : SUCCESS in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e6bd624e0ae4daaa619d8db7a8df922 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/449850a917794f6db81d068b10f8f2b2 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada29f81f14c429882bd25f158e62e63 : SUCCESS in 6m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ede9b1683b294204bc446af95d320039 : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48091b11a5e04f008d67f2b553aea894 : SUCCESS in 7m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/127a073e803847c1937a54fa7a2e6eae : SUCCESS in 8m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e3d649caaf964cbca4a973b9ade7c5bc : SUCCESS in 43m 07s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/03614758a46a4f4cab51a052ff7377d9 : SUCCESS in 31m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2b7b66ba93e4a5eafccd63ee7fc8b79 : SUCCESS in 43m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5f21397c5c2945e8ba30d1dc6d465c01 : SUCCESS in 57m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cf51fdbe392a492bacfd5cf10f65ae67 : SUCCESS in 32m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b9cbad42623e4b52b9180b2cd1fb42b8 : SUCCESS in 16m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51abf8985616442fa3e7a36d483f3fbb : SUCCESS in 10m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/82b3261671ce410a8fa3e6244b36c547 : SUCCESS in 2h 13m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec847e3db6c0438297ba8e41fd48301d : SUCCESS in 1h 12m 14s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/36f74aa920cf4b20919856c07cb74499 : SUCCESS in 2h 20m 14s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cc5ef2012cc4a0c87b8f8979062c3c7 : SUCCESS in 2h 06m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e4b629cec4da4539b15986ee55af39f1 : SUCCESS in 8m 41s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/f759b014ea0d434bba30b892ab9285c6 : SUCCESS in 14m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fbb4b16187f4a93a944ed31021fdcb8 : SUCCESS in 12m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a65e4cf3d9cc4b2eb0b09c78d1f1f9a2 : SUCCESS in 16m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f45c268214c1494bb12e55536faa6df6 : SUCCESS in 36m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1506663fe50e481193f724bb0e929637 : SUCCESS in 25m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/903048183df44e2f8558adb1b28b4f50 : SUCCESS in 35m 52s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e1e56c4ecb434c41b3d6a96633fcfca2 : SUCCESS in 27m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/84375b8116b74eb6aabf2c31419fae8d : SUCCESS in 12m 46s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/47224b7478b947eabcddd861551dc054 : SUCCESS in 6m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0467617104f04c7982d36c714eff3e47 : SUCCESS in 11m 20s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/fed2ee88507b4f2b842c8a3e09306226 : SUCCESS in 5m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f38f3159c6d745aa9c8760970459eab6 : SUCCESS in 13m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38841a47a8d54c4ea1281e00f796eea7 : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":7},{"id":"1a93f86dbdf7e9933aab404a457049050e0ac0a1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-23 05:02:14.000000000","message":"Topic test4 removed","accounts_in_message":[],"_revision_number":7},{"id":"0ba882429ec41e2461d47bf26028d69d502cd6dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-23 06:33:39.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"5d52a93a2cddd52d5a5bdb43cd28fc057e1f3562","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-23 08:47:43.000000000","message":"Patch Set 8: Code-Review+1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"03ff987926661b810fd09d091db6f7ccc1c8fcc2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-23 08:57:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a4f387d95a08441dc86ef5a892de7c026509026f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 09:09:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a5e0e4a1ddf4e4cb07455a84a9fcb36\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1eaa7397c8194e13adb8f5034c81d7bf : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b357c455cb243c2b763b948bf9e8428 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6493dc1700f4e9da453c791d7379d98 : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7671c3a2825b4e98bca1fbb2a16228e8 : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a19f096e3d6451fb06fdf333d12cb6e : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f59762455a5d4c309b05c4d3ed98cf89 : SUCCESS in 8m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/55126386ca714256b23e9f4fa04bc64c : SUCCESS in 2m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f50ff812d074b60ae82ece56caa18a0 : SUCCESS in 9m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e32c402a1444a8bb2c0401019e59b97 : SUCCESS in 8m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2ac4a918523e4b4099715df1f1bab12c : SUCCESS in 50m 54s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/01e1c8b912454b9ea86ef397a744d756 : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33681e49d0c84b2ab21dab8976fd6aa8 : SUCCESS in 56m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/46bf5958f78443b8a8ef1d9d5b93e8d2 : SUCCESS in 57m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/38ac6fe77ea24f1798cbf2e6ec72d425 : SUCCESS in 1h 11m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/373359d899d94c6fb2081f8c3f884896 : SUCCESS in 18m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a32f4a3bc9c242bda613f403f88d2a4a : SUCCESS in 12m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/525e6c2286e24f68be78d60371afc48d : SUCCESS in 2h 19m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92d019f6ad984818b3d92f625ba0625c : SUCCESS in 2h 23m 05s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/260cb9bdb9a841db87221cec26d6b200 : SUCCESS in 2h 29m 30s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2e021dd45664152b84405c5c9c28b24 : SUCCESS in 1h 11m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2ff9d74a47184993bd99ec09c54fea99 : SUCCESS in 16m 07s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/136b044ea2434dc391b875a022640ca2 : SUCCESS in 15m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4caa288fdc2d4aad8d4eee40c9048799 : SUCCESS in 12m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ec0267300f24654bc58947f730c3d81 : SUCCESS in 13m 16s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ab8b91f68d8c4d0fbc623b73519379d6 : SUCCESS in 33m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/27c336a33a184cb896d6a98148ace8a9 : SUCCESS in 29m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/445bb97574d245f2bd69e8f7d74f0ea1 : SUCCESS in 25m 59s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/4ca9922438174007b8d199d7e924cd42 : SUCCESS in 32m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1d6ab46d6379413aa1c0d52ff1a24030 : SUCCESS in 12m 04s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/ffe15f69e36d43ae83ce02e499a62f52 : SUCCESS in 10m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e132f91af9f74c4d9f94a87e137083b8 : SUCCESS in 10m 18s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/c24e9a36f8ce42789dac8ffe3fd9304d : SUCCESS in 10m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2d01b5e38ab74570823bb819985a8e43 : SUCCESS in 11m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28282be0943a4549a4c58aea70572243 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":8},{"id":"c7f380d030517d652d1f60021e0fccfa4ed01fa4","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-05-23 10:21:20.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"fb55228e6faccef6f9e94728ec2d04fbaf047f79","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-23 10:31:31.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d01d12d79aa1f2c35b3556c40058f70022309afb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-23 11:44:24.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"76b0b864bf4c603bc0a70ba61487df527904eaa0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-23 13:23:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"21ba45b206dcd352691ea370a2ef1cb034fe3c53","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-23 13:32:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bdce3f48d6abf6fcdd2016b39315b12d4e1eb366","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-23 13:38:49.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c1072610a67ad1b6d374d875437ef8fa0a01fa9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 13:55:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a49a3c705cde45708e6102990a87b3b5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7badb8410f0b40d3bcd6580441219c7b : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92d00aa5158a427aa378b0999a8e5d11 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/104948407c924096b98bcaf9f75e8c9e : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc799599cbfc45b09edbd60f245191b9 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d041ce8b3ef492b8ba474961539ea8a : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c8594fb1824455927e8deea30424f7 : SUCCESS in 9m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f8146c09b04b4fe5b4835b6ee3441a1c : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ffe63072ee3b44e29920854e24e50d6b : SUCCESS in 12m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d75e0a00d9014ff69eb1cbfe88f108d1 : SUCCESS in 10m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4ec761bfc11944399a6c84e7c0261415 : SUCCESS in 56m 56s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f77cf78b2996470eb62c48d804c179c6 : SUCCESS in 49m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74a32e230cb14bb28f0b4bedc223af26 : SUCCESS in 48m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e70fae44aa874addba601fd126255bc5 : SUCCESS in 56m 26s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5285ffe028e84ead89595433eeed5d97 : SUCCESS in 32m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/decc53d0d03a495e8075c24989b87eac : SUCCESS in 14m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df903d0e488d4b20a7a4d0e20f16d144 : SUCCESS in 6m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bf3d888e9d52479fa9a49bf8fcbabb7b : SUCCESS in 1h 47m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f92c00e027c84a5eba3ebb637d32dc26 : SUCCESS in 2h 05m 28s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/387347f2306b40919557736bfe8433e4 : SUCCESS in 1h 49m 23s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97a1b48c0ca148358e24f57c01308477 : SUCCESS in 1h 13m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5443cbf12f6049b2a29de2d3349db321 : SUCCESS in 8m 16s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/2237e7b7b6144fcda4677e2713960ccd : SUCCESS in 8m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66cc0bfb51fe41bba664a25263016764 : SUCCESS in 16m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cb943b2c2222468cbe1ccf36f697bb04 : SUCCESS in 16m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a0efd3fd55c94918be2a7e9c5099635d : SUCCESS in 38m 58s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/83f2804da05747f29b56a27e2e67058e : SUCCESS in 25m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68b1dd818f574ab181dd94efdaaf4ff8 : SUCCESS in 29m 28s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/74a93261a733481e87e000c16e5101db : SUCCESS in 29m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/83f81daf7f6e4f4e86bb63373f87c0e6 : SUCCESS in 11m 36s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4417a24927a14f2bb75391d1ae940ccb : SUCCESS in 11m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d917bb8a51004f56a8baacf215af37ca : SUCCESS in 11m 07s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/fbb87dab3ad247b784825b81aca3095c : SUCCESS in 10m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ab2d00f6bdf04a028dbc2f02054b9207 : SUCCESS in 11m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19accab5b14e468a84b65fa245ae6d77 : SUCCESS in 10m 30s","accounts_in_message":[],"_revision_number":9},{"id":"9c5e6ad5a99a130703b69817cac4a7db05bbdfcb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-23 13:59:41.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"7d60022a6e88dc8c82944688f204b4e4f3831c0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-23 14:24:09.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"c391c029f9b9dfca720eebbc8a75c8932d29cada","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-23 15:16:58.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"31b6ff3710aa3bb946773156b231006746fc8c65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-23 15:44:42.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":11},{"id":"f4ee8094a4b581c98ef250b0816bf77f39789b03","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-23 15:45:45.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":11},{"id":"fab03e94c5bf8e4582bcb8bb0933add986568901","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-05-23 15:50:48.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"03764719c9a794ba530bfde25d22aaa380c8422b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-23 16:04:58.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"269b5764a5f4e7acf33ad38a7005f10bb971adb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 18:07:56.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1779c396c00b4ed5bacb9a2c85aa138b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/090161bca2be4b4ea39bec2a7f1441a6 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9df5135bca5c450a8381b9416c498eca : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef5bf9fbec1d43c09684cdbb9c373431 : SUCCESS in 5m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e173a312e35441789ea252b592887a6d : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cc9f57146eee43ed950eabe15165f2f6 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eaed6ce8048422ba2518f2482d8e62d : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb05cd95535f49a5b49ec1600d200fa1 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bde7869e8e0f4134a1b2aa92af249c06 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1fbc4a1003a44cc966dac958b35b907 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c93ce02dd0c4b6096d5fd7d7bf3d551 : SUCCESS in 12m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/09caca4a89aa4920ad42c38aee6fea7a : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0743da2e5dbd4d6c8fa9f093d6c55ea0 : SUCCESS in 1h 04m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53b153f823074881a8d9a7c4edca965f : SUCCESS in 54m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6a982b65842e482a8081a63250ac98e5 : SUCCESS in 59m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/97b9bd0a336d42d48f85e4f34dcb0549 : SUCCESS in 1h 02m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/25dba8b57dd845a6b928333968e16493 : SUCCESS in 18m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20e33684a40445f9ac8e26854ccb0e72 : SUCCESS in 7m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ef22022c693b45cc86787e2b10353a91 : SUCCESS in 2h 06m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/741caf08e97445ceac70a6016bc726b9 : SUCCESS in 2h 11m 57s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/22263ec1f02b4c2eb77b6f94fa4235fb : SUCCESS in 2h 13m 44s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/425ca693661f43859b4ee7b0c21ed00c : SUCCESS in 2h 12m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/325e4b007ad94b7aba20fe5e58e1bd09 : SUCCESS in 16m 28s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/9cb5bd97d88e441ca52a633726bd512e : SUCCESS in 18m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63191c009b504aa6b628ecc73efc0a57 : SUCCESS in 13m 53s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cd1e22effe6749fba660317b6ad7b9dc : SUCCESS in 13m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4128b58407164592832c4c14b0709372 : SUCCESS in 36m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/59714fedbbb34571a8dde310fc872240 : SUCCESS in 32m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b373851548a549d5a0eb270683a70666 : SUCCESS in 31m 19s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/613995db0d5f4b438e7d98dc60d0c953 : SUCCESS in 38m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/178225039ebb46f98c0e44d7d64433e3 : SUCCESS in 12m 06s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/136799a99a2e45e5b6da1154c6017d63 : SUCCESS in 15m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19a6ca575a5f492ab27c54e58cdd544a : SUCCESS in 10m 36s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/0bdf746c587d44cbb0aa196bc41fdf1b : SUCCESS in 15m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9e5ec3231dd3479f9ebc6e9847f6ef4c : SUCCESS in 12m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c463f83d8d34415bbf776c5bf5e61324 : SUCCESS in 11m 08s","accounts_in_message":[],"_revision_number":11},{"id":"15f8572c6378378ad090d0305f229627fc98ac9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 18:08:10.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"315ce5185b32d9114f9ea19a7103555b9598d31e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-23 19:47:23.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"fcf4cd278524c25ddbdfdeaf7d9d58f4c85322a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 20:37:52.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46e726b4a1fc495cb653008410a7a6db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c39be7530c9f4a40a1a223ebc314a5a0 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ce0e2b7abf9402f9279b56eaec18d2e : SUCCESS in 2m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f44fb66e9a84f5bb9443ed102d9917a : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7bd1236a4648639ce54f75779911a0 : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/705a4e62f38c46fd9f469ecbbc0cc023 : SUCCESS in 1m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/66c817094b5b476a97ba7f091a8f6ce8 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/66e033394bd6497a9e74588e2a665720 : SUCCESS in 8m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9373b00f5cb6448d9dde79067838b9e3 : SUCCESS in 12m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0984b940874d41e0a2f90f6b02a79404 : SUCCESS in 52m 57s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/d79b3f4df87349a0bb30aee751e07f4d : SUCCESS in 56m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4ee93e5d802406d8abbd0d1c1ce7744 : SUCCESS in 56m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1b4c39fdc9504701a093c186142210cb : SUCCESS in 55m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4b47ab2d13f042058a2efbb97e2f1b0f : SUCCESS in 33m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/24cc87fe089a49ea9ace6a7ac1d0b0a3 : SUCCESS in 15m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bde8157456045d79907534ca523f6e1 : SUCCESS in 11m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/85a5ae45dbd244d38c139aed30787cb0 : SUCCESS in 1h 58m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d183f49b520c4a2da5d83d3fe17a44bf : SUCCESS in 2h 12m 18s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f2ebc220f86b4272af9d7889cc78af33 : SUCCESS in 2h 25m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9affd4a6920c4fbaafe541db355e7b96 : SUCCESS in 2h 06m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/18045638c3054d99ae24255c718cd360 : SUCCESS in 15m 55s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/eff751c983994983a2b4aeb6f4cec23e : SUCCESS in 17m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9d24c0a4ac4497499680a32deb06c27 : SUCCESS in 15m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bbb7fce4638b4a08a419a8cf38c09d9b : SUCCESS in 13m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1f1cc9d30cbf4d8e87ef11815c846003 : SUCCESS in 33m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f00eb03d88e46dbb18a671ea5568747 : SUCCESS in 31m 02s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ad22768bfd9448bc9d28139c5488c2a9 : SUCCESS in 32m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f910d2bb75b04472bc134c573e316856 : SUCCESS in 11m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c37300dd40542178e329d506edf4516 : SUCCESS in 10m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6949e4a55e00409d8f5aa63d1d36a33b : SUCCESS in 15m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6947a15ac8dd46f895cd33067d8a8def : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":11},{"id":"8e0822d06ba1d00b9529c16a34e3344f84a68b8d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 20:37:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"7486188196135a938ee15df256558d1c14554705","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 20:38:42.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebc94e0e3c2b462e949b52fe767fd049\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e052cffaf7b54122ab2c30ddc85dac47 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a74bc529144c49ebb871a1b375ea3cef : SUCCESS in 38s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"4e2220bb6cd2b0e6d6e8a2106113ec25db55c3a3","revisions":{"867762ea8ee1da5741f7c177597719705eb1ce2c":{"kind":"REWORK","_number":1,"created":"2025-05-22 20:23:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/04/950704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/950704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/950704/1"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"subject":"[DNM] Test 4","message":"[DNM] Test 4\n\nChange-Id: I7cfb600eec268749ea1315112fe3a9429ab3ab8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/867762ea8ee1da5741f7c177597719705eb1ce2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/867762ea8ee1da5741f7c177597719705eb1ce2c"}]},"branch":"refs/heads/master"},"abe3b960bc57ffcc0b76a39feda5690aa4b0046f":{"kind":"REWORK","_number":2,"created":"2025-05-22 20:34:57.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/04/950704/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/950704/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/950704/2"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:34:21.000000000","tz":120},"subject":"[DNM] Test 4","message":"[DNM] Test 4\n\nChange-Id: I7cfb600eec268749ea1315112fe3a9429ab3ab8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abe3b960bc57ffcc0b76a39feda5690aa4b0046f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abe3b960bc57ffcc0b76a39feda5690aa4b0046f"}]},"branch":"refs/heads/master"},"6bde2b3a21ad48a418c93914a514437eb985ba4d":{"kind":"REWORK","_number":3,"created":"2025-05-22 20:51:36.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/04/950704/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/950704/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/950704/3"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:51:27.000000000","tz":120},"subject":"[DNM] Test 4","message":"[DNM] Test 4\n\nChange-Id: I7cfb600eec268749ea1315112fe3a9429ab3ab8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6bde2b3a21ad48a418c93914a514437eb985ba4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6bde2b3a21ad48a418c93914a514437eb985ba4d"}]},"branch":"refs/heads/master"},"1937050264b3c94cea05077d71866b55fcf49e52":{"kind":"REWORK","_number":4,"created":"2025-05-22 21:03:05.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/04/950704/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/950704/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/950704/4"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 21:02:57.000000000","tz":120},"subject":"[DNM] Test 4","message":"[DNM] Test 4\n\nChange-Id: I7cfb600eec268749ea1315112fe3a9429ab3ab8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1937050264b3c94cea05077d71866b55fcf49e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1937050264b3c94cea05077d71866b55fcf49e52"}]},"branch":"refs/heads/master"},"de91acfd97615d20d9eaf8860955c584a9283382":{"kind":"REWORK","_number":5,"created":"2025-05-22 21:32:50.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/04/950704/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/950704/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/950704/5"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 21:32:42.000000000","tz":120},"subject":"[DNM] Test 4","message":"[DNM] Test 4\n\nChange-Id: I7cfb600eec268749ea1315112fe3a9429ab3ab8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de91acfd97615d20d9eaf8860955c584a9283382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de91acfd97615d20d9eaf8860955c584a9283382"}]},"branch":"refs/heads/master"},"17ae91b9689b32eeb8fd8869579684e4ffa6f25a":{"kind":"REWORK","_number":6,"created":"2025-05-22 21:45:03.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/04/950704/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/950704/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/950704/6"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 21:44:56.000000000","tz":120},"subject":"[DNM] Test 4","message":"[DNM] Test 4\n\nChange-Id: I7cfb600eec268749ea1315112fe3a9429ab3ab8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17ae91b9689b32eeb8fd8869579684e4ffa6f25a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17ae91b9689b32eeb8fd8869579684e4ffa6f25a"}]},"branch":"refs/heads/master"},"a53d3949ee5160d8207902e8caaed31a01e8f005":{"kind":"REWORK","_number":7,"created":"2025-05-22 23:11:27.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/04/950704/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/950704/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/950704/7"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 23:10:54.000000000","tz":120},"subject":"Fix virtualenv and interpreter","message":"Fix virtualenv and interpreter\n\nChange-Id: I7cfb600eec268749ea1315112fe3a9429ab3ab8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a53d3949ee5160d8207902e8caaed31a01e8f005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a53d3949ee5160d8207902e8caaed31a01e8f005"}]},"branch":"refs/heads/master"},"9df8006649dbffa8e306738a2c68374f970d1e38":{"kind":"REWORK","_number":8,"created":"2025-05-23 06:33:39.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/04/950704/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/950704/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/950704/8"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-23 06:33:07.000000000","tz":120},"subject":"Fix compatibility with Kolla virtualenv configuration","message":"Fix compatibility with Kolla virtualenv configuration\n\nThis change updates how Kayobe handles Python virtual environments\nto ensure compatibility with the virtualenv handling used in\nkolla-ansible [1] and ansible-collection-kolla [2].\n\nSummary:\n- Removes hardcoded \u0027ansible_python_interpreter\u0027 from templates.\n- Introduces support for the \u0027virtualenv\u0027 variable used by SDK roles.\n- Enables cleaner and more consistent interpreter selection.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/949812\n[2] https://review.opendev.org/c/openstack/ansible-collection-kolla/+/949767\n\nChange-Id: I7cfb600eec268749ea1315112fe3a9429ab3ab8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9df8006649dbffa8e306738a2c68374f970d1e38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9df8006649dbffa8e306738a2c68374f970d1e38"}]},"branch":"refs/heads/master"},"a8adc019170f913c3445c19311fa335d7b80991b":{"kind":"REWORK","_number":9,"created":"2025-05-23 11:44:24.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/04/950704/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/950704/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/950704/9"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-23 11:42:37.000000000","tz":120},"subject":"Fix compatibility with Kolla virtualenv configuration","message":"Fix compatibility with Kolla virtualenv configuration\n\nThis change updates how Kayobe handles Python virtual environments\nto ensure compatibility with the virtualenv handling used in\nkolla-ansible [1] and ansible-collection-kolla [2].\n\nSummary:\n- Removes hardcoded \u0027ansible_python_interpreter\u0027 from templates.\n- Introduces support for the \u0027virtualenv\u0027 variable used by SDK roles.\n- Enables cleaner and more consistent interpreter selection.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/949812\n[2] https://review.opendev.org/c/openstack/ansible-collection-kolla/+/949767\n\nChange-Id: I7cfb600eec268749ea1315112fe3a9429ab3ab8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8adc019170f913c3445c19311fa335d7b80991b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8adc019170f913c3445c19311fa335d7b80991b"}]},"branch":"refs/heads/master"},"022b739131f66a3f44bde4883bf9c9c5fdefa490":{"kind":"REWORK","_number":10,"created":"2025-05-23 14:24:09.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/04/950704/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/950704/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/950704/10"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-23 14:24:01.000000000","tz":120},"subject":"Fix compatibility with Kolla virtualenv configuration","message":"Fix compatibility with Kolla virtualenv configuration\n\nThis change updates how Kayobe handles Python virtual environments\nto ensure compatibility with the virtualenv handling used in\nkolla-ansible [1] and ansible-collection-kolla [2].\n\nSummary:\n- Removes hardcoded \u0027ansible_python_interpreter\u0027 from templates.\n- Introduces support for the \u0027virtualenv\u0027 variable used by SDK roles.\n- Enables cleaner and more consistent interpreter selection.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/949812\n[2] https://review.opendev.org/c/openstack/ansible-collection-kolla/+/949767\n\nCloses-Bug: #2111594\n\nChange-Id: I7cfb600eec268749ea1315112fe3a9429ab3ab8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/022b739131f66a3f44bde4883bf9c9c5fdefa490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/022b739131f66a3f44bde4883bf9c9c5fdefa490"}]},"branch":"refs/heads/master"},"4e2220bb6cd2b0e6d6e8a2106113ec25db55c3a3":{"kind":"REWORK","_number":11,"created":"2025-05-23 15:44:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/950704/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/950704/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/950704/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/950704/11"}}},"commit":{"parents":[{"commit":"cdba0957b8544ff163df904c7cb6b07d6a3e0f79","subject":"CI: Update command to match allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdba0957b8544ff163df904c7cb6b07d6a3e0f79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-22 20:23:24.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-23 15:44:35.000000000","tz":120},"subject":"Fix compatibility with Kolla virtualenv configuration","message":"Fix compatibility with Kolla virtualenv configuration\n\nThis change updates how Kayobe handles Python virtual environments\nto ensure compatibility with the virtualenv handling used in\nkolla-ansible [1] and ansible-collection-kolla [2].\n\nSummary:\n- Removes hardcoded \u0027ansible_python_interpreter\u0027 from templates.\n- Introduces support for the \u0027virtualenv\u0027 variable used by SDK roles.\n- Enables cleaner and more consistent interpreter selection.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/949812\n[2] https://review.opendev.org/c/openstack/ansible-collection-kolla/+/949767\n\nCloses-Bug: #2111594\n\nChange-Id: I7cfb600eec268749ea1315112fe3a9429ab3ab8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e2220bb6cd2b0e6d6e8a2106113ec25db55c3a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e2220bb6cd2b0e6d6e8a2106113ec25db55c3a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
