)]}'
{"id":"openstack%2Fkayobe~950903","triplet_id":"openstack%2Fkayobe~master~I79cb938017721588012d820b86740b7219be5cab","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-27 07:15:23.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-27 13:28:40.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-05-27 11:02:07.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I79cb938017721588012d820b86740b7219be5cab","subject":"CI: Set previous_release for 2025.2 Flamingo","status":"MERGED","created":"2025-05-26 16:41:18.000000000","updated":"2025-05-27 13:31:42.000000000","submitted":"2025-05-27 13:28:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950903","meta_rev_id":"c426676d8efaf8100f3a0ac756c2ea895af94aa5","_number":950903,"virtual_id_number":950903,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-27 13:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-05-27 07:15:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-27 07:23:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2025-05-27 11:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2025-05-27 11:02:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-26 17:55:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-26 18:37:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-27 07:23:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-05-27 09:35:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"4df2924b3793f6f4c372f76e45a11f4b70683450","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-26 16:41:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9152805f72616972da4da03cbcafffc612e03764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 17:55:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66af7af9f34b478c918985ac5bd66701\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2ed4d33abfd4a84b4a5dabf62d8c102 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea969eb0676d4adea7cc6d623feb2082 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dafae4ac0f3f4467bf6e11e2c2ab9098 : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f87dedd93ae54e46a533260587f667a0 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b38ebbd126345c1a527906000f05d00 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95d650feeadf449a934b21540bd39acc : SUCCESS in 9m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a9964ed61ee349ebbd1570c5aa958de5 : SUCCESS in 56m 46s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1ee4f5ce018b4ad5b95dfdf3c63dfd2d : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c45745fcf814e8dbc6122c0fc2e7bfb : SUCCESS in 56m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/be101906bbc9430fbc9eccc4ef25ec21 : SUCCESS in 1h 07m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0482638df6b542f48339a698db1130cf : SUCCESS in 58m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/17b7f753f7fb4ca0abaf6e1dfc3f9039 : SUCCESS in 17m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bef514a670d4cf2b8c81d3bbaf8957b : SUCCESS in 14m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e1cf8c1924ed4db1b6dad27dabd9c9d9 : FAILURE in 59m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e4ce2beef064cceb6363eecbf7d1230 : FAILURE in 53m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/15342135548043ae858a51b898348c27 : SUCCESS in 16m 32s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/9e090fbac6f1473f867b1f19a8b7e450 : SUCCESS in 17m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c03be6975c1047b4b6857c298df08e72 : SUCCESS in 14m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8193ed0afb0145878b0c47e47597e59f : SUCCESS in 13m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d3e96c1b807548f4b13a8342f93917bd : SUCCESS in 38m 44s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/41d7db6a327b4f0295951843b16c9bab : SUCCESS in 28m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddc40a6d443542fe9689863ac217d705 : SUCCESS in 28m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/98e48124c0204aa289b35c72bcd825f3 : SUCCESS in 11m 33s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/34c21519e7da40a2b5c0eb68563e572d : SUCCESS in 13m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe7aea457d1c43dbb55049af8217507a : SUCCESS in 10m 31s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/564d927b212b44348fae0d8ab6434870 : SUCCESS in 10m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/93fac380edb44a639613109d8ee4dc59 : SUCCESS in 10m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75383a93bb5d4c619ed78cae4fcfca7d : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6a6f29fe848fd744be971c6318eb33479acb4982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-26 20:41:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7afefc0f98dec5bd82aa11d38cdffdbdb776ba79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 22:28:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebbbfa2e395745909cf5cfbaee2b2cb4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7ffa6a02d16471aa6d9062bccc07812 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d610a9d285ff4208a188d2f296790d37 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38d4cf06455f4f2191e0485d247a9e0e : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fecf421ceb574cf2a6ac0e55110caa94 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d3c68ba02f2442b9657aaf827774b26 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b02effae9549da8b3a771123dc5d93 : SUCCESS in 8m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ffcdf651bba94c1d8a84bfbbc3e04568 : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6579b38bf6684ead809568a8d5672b93 : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8090f62b68a04e47bf3cddb1b76cb4e1 : SUCCESS in 9m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e1c99911491c4013944fd0d0a3a5fd34 : SUCCESS in 58m 18s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/691e3055b75f47f9bce90496df893ad8 : SUCCESS in 49m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9013c7bbc9e54c16aa592ff5aaae97ef : SUCCESS in 58m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f1def9ad1e5444c88ac1e8f95b3b244 : SUCCESS in 57m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b69a43c130614a8b8ac1a5793794d5e5 : SUCCESS in 1h 03m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3b8da2bd12754b2d96c1974796eee608 : SUCCESS in 18m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1a2fde326874daab24883438b435df5 : SUCCESS in 17m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ed1c12c4f5da467482fabe2cbd9f73ad : SUCCESS in 1h 31m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/beb48b6311804f83b9a7421ff020f979 : SUCCESS in 1h 41m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/46127178b8144440bae6636343a742f1 : SUCCESS in 17m 08s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/43c8e3ee85534a67a0530f3ae69c3485 : SUCCESS in 15m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44babb5f33be4d1f9b68ab6e25ce6a43 : SUCCESS in 11m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/70270399358f46b99820470961d34332 : SUCCESS in 13m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4bd9a8c3556e4d8583d73eb9f931078c : SUCCESS in 31m 32s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f19f2ce6d0b9497a9eef0c40d261afed : SUCCESS in 28m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3cc5a8ad34d40eaa8b0afeba2c74eec : SUCCESS in 22m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/05334547c9ce46bb8d9ed213e618e965 : SUCCESS in 10m 16s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/a8da60d77bf243d1ba6223d8dfbfe555 : SUCCESS in 12m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e09fd4120ebd49e2945d18842643f5c2 : SUCCESS in 12m 12s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/fb9f2304dfb24dab92883f045ec414e0 : SUCCESS in 12m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e9d98a48dd324d7e838644c5ab93fc7f : SUCCESS in 12m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adbfa87b96444339926264788e7c3678 : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":2},{"id":"d580932d9fba744385325d0a4d81786de6633857","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-26 23:22:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"122cd53363a510683cf41e11e1f9305f478b7a3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-27 07:11:58.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"512f3f595f4a754a89fec5c7115539dd42e4e14d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-27 07:15:23.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"41830704864b268fcd1b1a27e3672a7806e1b99a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-05-27 07:23:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3d571cdc2d0db5ba842f5be3ba621c017aa66067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 09:44:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5eab9cc58884319b86ba82e9549cbe4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53df94a1c2534b0898a9c1b1782b5b1d : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7920044e2a5a4ddb930f14c46f1110ed : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a118adb0d4c46d5aee3b3441781ed3a : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7bbcbb4e37184be4bb59b8319a1dda5e : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f24f3a80c486485cb3cf7ecdfe2c17f9 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6994fb5b87d8438c885f423d5ccf762d : SUCCESS in 9m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b16f6c3491a42619b929cb0ecec556d : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/af9dab7888694b63893cd0ba1209d6ee : SUCCESS in 10m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c5e81680867245feb42d1c9f61fb2077 : SUCCESS in 11m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2f87d04f4e384dbc821e3e08a601a50e : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7f85a8c14f334275906c7b7f0c1e68d1 : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aadbe809c5c8410e90f5b7678250a3e5 : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/651ae587a4a345b28a490a0b8a3a4ec3 : SUCCESS in 1h 08m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/90fd9748e5ee4004a323bdac9047dfa2 : SUCCESS in 1h 14m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/42d1fa271fb94c46ab53907be37f43cf : SUCCESS in 18m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42239e852c984527b8f9c90d281eaca8 : SUCCESS in 13m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/23e66b99697b448f87c1e118b9256c24 : SUCCESS in 2h 01m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3667e182654432685525d1890252cfd : SUCCESS in 1h 54m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/183e176dd7e942a9b2c73a896e589aa3 : SUCCESS in 17m 45s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/114ee1d9271a4a99bf6ee091035789a0 : SUCCESS in 16m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04815043833e4de0a10ad4e2d1edd456 : SUCCESS in 15m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5e5e027bf41d4b97940fe55800efa6ae : SUCCESS in 14m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4e5a8bd8431448738b5e0eea6dfb01a2 : SUCCESS in 43m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e3ebf66733ea4580acf19893107757a3 : SUCCESS in 33m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6b468d1e85f48499ceb201cc4892765 : SUCCESS in 26m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4f0699141ed647f9afee4e5621ad1e0d : SUCCESS in 12m 15s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/6e80009063ad4323a5657bba48c4120a : SUCCESS in 13m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24d8c7d2b0c64af6b3fdce428c899f90 : SUCCESS in 11m 34s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/401c5866811a46a0bad969bd62d8e189 : SUCCESS in 11m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c540b801779c47a7a2127c010570bfe5 : SUCCESS in 12m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6115617c532a45a4be93f3c872578eef : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":3},{"id":"808003e0fdeb70c188af0036abfbb45f1613d46c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-27 11:02:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"90969c1b6edeec5341656c4d88576b4d09c62d6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 11:03:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a1dd2550c071b6089f0900073655cba25fad249","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:28:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9861985ada924c2790bbd11339dc628c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffaf97ed50ac440a9904a9204fafe827 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/405e08b4368c495e81033318d06e4162 : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/96fc324bbe0c4acda6a086de233f4d67 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd9dd687978041fdbc51147382a5d2d9 : SUCCESS in 9m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/64bea17e99c0434d95ec841515d80106 : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c7db03cd9f14aaab3a7ff2dd7cdc6f5 : SUCCESS in 9m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ce4d26bad8334f6da0be4c6741084ccb : SUCCESS in 9m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c6fb417626994fa0bc73ebc4b14a2cd5 : SUCCESS in 1h 02m 14s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/ea3c096dc8ec48e1b4044c648d027405 : SUCCESS in 1h 06m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f5acefb28d547f7a27585e3af3c95af : SUCCESS in 58m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d783ff6a86aa496c8cd266f0bd5f1405 : SUCCESS in 54m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/75fbe60084b0448f804d954f2ab5b1b3 : SUCCESS in 55m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f3ea71ffbc234d2ea54a4c3bc751017b : SUCCESS in 19m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47ea0ad3393048c5997bcc283a34918a : SUCCESS in 13m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/223e37d7e12845168e907e7bf52e4a14 : SUCCESS in 1h 34m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e573e62f9ffa42e9a85bd9f3a5b24b11 : SUCCESS in 1h 58m 45s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7fd94b52a67f4a5c8772983e47491e7c : SUCCESS in 19m 40s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/be6cbfc947974ff9b5055c19234bfa72 : SUCCESS in 16m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f60f9aa5182d4c4390f563eec17bc6dc : SUCCESS in 12m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b603a4acf01f4fc6bbbfd5f1fd2fe41a : SUCCESS in 16m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b6670be32fcf4af1bb8df7f354bdd1c9 : SUCCESS in 26m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/537fe4c57c364a228986c7aa19ffa172 : SUCCESS in 23m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f1d145844e8540e5823edb45611d5aef : SUCCESS in 12m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1696d3a5fa54215aa0410295013ad7d : SUCCESS in 11m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c8db735a916645a09c1e212d33d966b8 : SUCCESS in 14m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80d10605fed34d049eceaa95d3dd7b00 : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c4a0b768ad7874c732f3d37235763005e8df17a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:28:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c426676d8efaf8100f3a0ac756c2ea895af94aa5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 13:31:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f3bcb1ddc8546cfb006869d08bb0606\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72ae089392248eeb7cf0cd3819ba4e5 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2f6aee4e8785be58cc6dd68e7b27e997af91361f","revisions":{"e4e338c083328098075ebfcc4b6861a06554b1b7":{"kind":"REWORK","_number":1,"created":"2025-05-26 16:41:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/950903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/950903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/950903/1"}}},"commit":{"parents":[{"commit":"e3cb2e1149eb9649beeee335caa3289587dab22f","subject":"CI: remove SLURP jobs for F cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3cb2e1149eb9649beeee335caa3289587dab22f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-26 16:40:55.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-26 16:40:55.000000000","tz":120},"subject":"CI: Set previous_release for 2025.2 Flamingo","message":"CI: Set previous_release for 2025.2 Flamingo\n\nChange-Id: I79cb938017721588012d820b86740b7219be5cab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4e338c083328098075ebfcc4b6861a06554b1b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4e338c083328098075ebfcc4b6861a06554b1b7"}]},"branch":"refs/heads/master"},"a05776053e69240dd5987acd63fc9aedb7fdb707":{"kind":"REWORK","_number":2,"created":"2025-05-26 20:41:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/950903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/950903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/950903/2"}}},"commit":{"parents":[{"commit":"e3cb2e1149eb9649beeee335caa3289587dab22f","subject":"CI: remove SLURP jobs for F cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3cb2e1149eb9649beeee335caa3289587dab22f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-26 20:41:14.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-26 20:41:14.000000000","tz":120},"subject":"CI: Set previous_release for 2025.2 Flamingo","message":"CI: Set previous_release for 2025.2 Flamingo\n\nWe also to remove the RabbitMQ queue migration from the upgrade jobs.\nThe queues are already in the correct state when deployed on 2025.1\nEpoxy and the script is absent from this release.\n\nChange-Id: I79cb938017721588012d820b86740b7219be5cab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a05776053e69240dd5987acd63fc9aedb7fdb707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a05776053e69240dd5987acd63fc9aedb7fdb707"}]},"branch":"refs/heads/master"},"2f6aee4e8785be58cc6dd68e7b27e997af91361f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-05-27 07:11:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/950903/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/950903/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950903/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950903/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/950903/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/950903/3"}}},"commit":{"parents":[{"commit":"e3cb2e1149eb9649beeee335caa3289587dab22f","subject":"CI: remove SLURP jobs for F cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3cb2e1149eb9649beeee335caa3289587dab22f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-27 07:11:43.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-27 07:11:43.000000000","tz":120},"subject":"CI: Set previous_release for 2025.2 Flamingo","message":"CI: Set previous_release for 2025.2 Flamingo\n\nWe also need to remove the RabbitMQ queue migration from the upgrade\njobs. The queues are already in the correct state when deployed on\n2025.1 Epoxy and the migration script is absent from this release.\n\nChange-Id: I79cb938017721588012d820b86740b7219be5cab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f6aee4e8785be58cc6dd68e7b27e997af91361f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f6aee4e8785be58cc6dd68e7b27e997af91361f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
