)]}'
{"id":"openstack%2Fkayobe~951122","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~I852e945ae2e6261f8b9a549fbba75adc450459c6","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-30 11:37:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I852e945ae2e6261f8b9a549fbba75adc450459c6","subject":"Fix syntax error in drac-pxe role","status":"MERGED","created":"2025-05-28 11:10:46.000000000","updated":"2025-05-30 11:38:00.000000000","submitted":"2025-05-30 11:37:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951122","cherry_pick_of_change":951001,"cherry_pick_of_patch_set":1,"meta_rev_id":"6bc962f0aa5bf42e7aa9c7f4c583b2478bd7f960","_number":951122,"virtual_id_number":951122,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-30 11:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-05-30 09:40:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-05-28 12:14:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-28 14:20:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-05-30 09:40:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-28 12:14:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-28 13:31:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-28 14:20:13.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-05-30 09:40:51.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"b279f2f1bac525d1573632115f37a0e256bcaeb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-28 11:10:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"382bc6877be91bc8775cb570b3d96ba312ca6e92","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-28 12:14:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"28b29bb97aa9a92d785455b31d72a86fcbc54052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 13:31:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9901e442b0c34c65802a0510221b6e72\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2984590bfdd4494b14a5778a07ccdca : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd30ba5f3f7f4fb9b914cc39c9016f8d : SUCCESS in 7m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8628b7bff33843e38e7b00104689580a : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db0681baebe043e3aa45d945325ca94f : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91cbd101300641c8b7f12ecc27b134de : SUCCESS in 9m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/00759f3ac9a14c10bbc71a0c652be4b7 : SUCCESS in 1h 01m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/24b59275bb424a1db478d81a5d6ec77c : SUCCESS in 1h 03m 15s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d4fa2631e9e440fadcb102a27ec9c09 : SUCCESS in 1h 02m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cdc0978fefb452d9ba5864b799e9c83 : SUCCESS in 53m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/812a4498489b479fb52ea59f19b62719 : SUCCESS in 1h 17m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/004d0bb9bd6547ee9476ca4056af43d9 : SUCCESS in 16m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d745bb1648b244baa2c5043f90d19c4b : SUCCESS in 17m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/072375894a5a4bb9b6419131344c91c2 : SUCCESS in 14m 46s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77330c2a0e8d4a19861fb2d2846a1264 : SUCCESS in 12m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/920242528f214c7d98bb0bdaa361e540 : SUCCESS in 1h 40m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c736983efce44bffa7676b643f0066b4 : SUCCESS in 1h 33m 21s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/524b45cf9c864ad19975ff928dec3887 : SUCCESS in 2h 09m 48s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cc2c0acc2b34e4184405aa78f7d40ea : SUCCESS in 1h 42m 28s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8e02fae2135d49bfbffe2b020be1d844 : SUCCESS in 19m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecc3cbb7be1f4a6093e6b4ad7eb8d86a : SUCCESS in 17m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3b5698a596d49bdb0ae7ad730ed125c : SUCCESS in 17m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bc672073823d42a5864a6757319f5c5a : SUCCESS in 36m 34s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ca238ca0878c4d0fa3dd6559444cb8a1 : SUCCESS in 27m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8560a386ec543048d84854e948fdee9 : SUCCESS in 21m 27s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/53d82bdcd45c4e4f9811681218ae9d6d : SUCCESS in 30m 39s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96c1bf6065b64d389abae23f9e0b79ef : SUCCESS in 23m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/12276d133647447aaeca1f70269621a5 : SUCCESS in 17m 38s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/fcaf6141d57c4885921988aac680d211 : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd0d389f66fa4ae89ff9128c4dadbb1b : SUCCESS in 15m 44s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/a5cd414fafea4bada3768b31bcd8e315 : SUCCESS in 13m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9731d9631cc74158af80fe164d6b9bc6 : SUCCESS in 13m 30s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/3b8d8d9ab58e402d9e84babc052cba0d : SUCCESS in 9m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e3f1ce311166409bb62d256a2efc7b36 : SUCCESS in 18m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e108e8bb5b71402fb804507e4b24c0c4 : SUCCESS in 13m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bab6a4992601406095cfe47dd612545f : SUCCESS in 14m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff2bd46cfb16427ebb410122ff3c043a : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa78c152aa254284b8b2df2269cf3b79 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0d80a52e5f642bdb8ea9bdbb2b3a362 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef5074cf3c7c4c049f87d20d97f1a9c8 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a48c16abe8f6b280985bbf8c55c77d91e63f5e5b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-05-28 14:20:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"328af788ec636378f3b1f67d341561d24985c4d5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-30 09:40:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1ea6f7436de8b530a7993cfeeaa7311e36f9dd83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 09:41:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6555c6c10cbfc847a17f5cae4d268c727b69da3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 11:37:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83ebd1b71f924673bfccc3ee0fe049ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd558a2899d41378e14f038b5ef021b : SUCCESS in 6m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2cb8cf814c7441b8aa5d5239e280ae25 : SUCCESS in 3m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7a98c1a992ef43699a29c4fa8ab73c4d : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/126fac794d294cfcb0ed4b23c0e5d1a9 : SUCCESS in 9m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f294d7c0c54e4747a6e082b72516c78b : SUCCESS in 1h 01m 52s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a87381baf064c998621ae2b5ed66894 : SUCCESS in 52m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/096f9fa20c534411a154351eda5b09ac : SUCCESS in 56m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4b025b4a2bc94334be309b24746752d2 : SUCCESS in 1h 08m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/db635abd2a9d4d29939ee297d6d13983 : SUCCESS in 17m 00s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1250e216c5364f839316a4b744afcba3 : SUCCESS in 13m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dba3f4a8f4474f73bdb80b5e7f3fa269 : SUCCESS in 13m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0d12ea2af76b46f2baf3d3a7447f0135 : SUCCESS in 1h 47m 57s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c5a48eb94b04b4494a74ec8c006c566 : SUCCESS in 1h 49m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5902b4a49d9045db802c7eaced36c1d2 : SUCCESS in 20m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6605a322f6eb4fb1a744674a299221a6 : SUCCESS in 17m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8138b0349f624ccdb24075957f536f4a : SUCCESS in 22m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b36094dd7eb346b78cdb101e5b783537 : SUCCESS in 26m 15s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca0e48ab8e0d419f98c0a3460d86f130 : SUCCESS in 28m 28s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/27bc989fce264fa795fcac215674e2a5 : SUCCESS in 24m 20s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c97c7d6242754dcf8368f5f962608641 : SUCCESS in 30m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2a866346c61049f19561173daec45be3 : SUCCESS in 14m 52s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a9714a9cb6a4885af88a66955609c36 : SUCCESS in 15m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cff063976d3443b0a242da8e95888e4b : SUCCESS in 9m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d8bb8572ee984c589fdbe6698ee3a747 : SUCCESS in 15m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc5564685f5b4cae8b73a6e0d6034389 : SUCCESS in 11m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fc737f7c11940da80db3e6cf6c578de : SUCCESS in 13m 22s","accounts_in_message":[],"_revision_number":1},{"id":"759e9455b89d9ffb5cb035a29e8faedec3fc6022","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 11:37:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6bc962f0aa5bf42e7aa9c7f4c583b2478bd7f960","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 11:38:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d65744a66eb4a6eb33bb05d913abfe0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1779cbdab3b24b5ba713b7c1242616af : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2c3d2b1a746991165e4732e4b0a7b8e6423024d","revisions":{"d2c3d2b1a746991165e4732e4b0a7b8e6423024d":{"kind":"REWORK","_number":1,"created":"2025-05-28 11:10:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/951122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/951122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/951122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/951122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/951122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/951122/1"}}},"commit":{"parents":[{"commit":"23fe82dd62b8b7cc8544d9a2f4e8fd41f6540229","subject":"Fix authentication failures due to missing system scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23fe82dd62b8b7cc8544d9a2f4e8fd41f6540229"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-27 13:57:18.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-28 11:10:46.000000000","tz":0},"subject":"Fix syntax error in drac-pxe role","message":"Fix syntax error in drac-pxe role\n\nProbably the result of a copy-paste from an earlier line in the file.\n\nChange-Id: I852e945ae2e6261f8b9a549fbba75adc450459c6\n(cherry picked from commit 9b805ad1902ad9c805d210cca0dde8a5481389c5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2c3d2b1a746991165e4732e4b0a7b8e6423024d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2c3d2b1a746991165e4732e4b0a7b8e6423024d"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
