)]}'
{"id":"openstack%2Fkayobe~951425","triplet_id":"openstack%2Fkayobe~master~I2409428e7b929bec32c2b0034e82b4d03b343567","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2025-06-02 18:49:10.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-06-02 18:49:10.000000000","reason":"Change was submitted"},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2025-06-02 18:49:10.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-06-02 14:40:30.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I2409428e7b929bec32c2b0034e82b4d03b343567","subject":"Make custom Gather facts run with default python for kayobe-target-venv","status":"MERGED","created":"2025-05-30 16:13:24.000000000","updated":"2025-06-02 18:50:03.000000000","submitted":"2025-06-02 18:49:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951425","meta_rev_id":"6a01b82f3835221ec52a40713bb61ba510b28662","_number":951425,"virtual_id_number":951425,"owner":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-02 18:49:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-06-02 14:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":2,"date":"2025-06-02 14:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-06-02 14:59:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":1,"date":"2025-06-02 14:34:26.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-02 14:34:26.000000000","_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 18:18:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-31 14:32:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-02 08:32:19.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"REVIEWER"},{"updated":"2025-06-02 08:35:46.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-06-02 14:59:13.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"4d0e61ceb3c69eb8f0a76cef7cb4d7ef899f3789","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-05-30 16:13:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa00fecf77ef7d91cd11ce3367891928137e9b19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-05-30 16:14:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0f67ba58a7246cccf8153c85a89886aa537e8e96","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-05-30 16:15:36.000000000","message":"Patch Set 2: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"544d46370dc1565c4d5b05996a6de7f5cf9d222f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 18:18:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3a949b23c4544c5b7522fd88997b11f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ff4d519f26f4cd1adc1111fa67e82ca : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d96052e0bbb14894883ec6f5233ef2d4 : SUCCESS in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13220b14b5ca4536a57bc8f335ce4ad8 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b53d94b4762b40d9b07240b0ae713086 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e2f862e7dab4716b47cf4a3f7a9f35e : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc64063f03f74ee6bb7fccb8e132c2a2 : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/557c2f5f772f45b3abb92e5d96e459cc : SUCCESS in 8m 12s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8027da391c9f4971bdfe00b27abe5f17 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ece990746b9b41439a7d66f84f20bf8e : SUCCESS in 13m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/220ab5175e714201a2a62a67271ea137 : SUCCESS in 10m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4925e9bbb12842cf8149b49a63422e31 : SUCCESS in 56m 32s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a2a9428bd66c4ea28d792279eb5c9015 : SUCCESS in 56m 08s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/697c8043353f4e35a9e109f882bac90c : SUCCESS in 50m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1fb4dac737aa4c6d8eea6afd130e2e5f : SUCCESS in 56m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a29cf6ac33db4484a536afd89a5bc2f2 : SUCCESS in 1h 09m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/aee5e5d44e0c4b0dae6b9246445d922e : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/192050438ae0428b94695b5c1546162e : SUCCESS in 12m 42s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ef3c2148af194a0b8a1656b90203f755 : SUCCESS in 1h 47m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dafd200e0064485bbf5ae0e87331dee3 : SUCCESS in 1h 40m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b0aa7dfbd11a43948faad7818907f812 : SUCCESS in 15m 09s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/4c23b630b1594158bdaa4ee4abb8494e : SUCCESS in 16m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40f9eaac066f42caa96837b3894b991c : SUCCESS in 14m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/34a0832fb6214ce9bce90fda2c473069 : SUCCESS in 13m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4235e8e04d314cf780302fd54705edcf : SUCCESS in 31m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2e5aed2b4fa844e58cedabee641458d1 : SUCCESS in 29m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18619ea5b59441b19ea41c6efd498b51 : SUCCESS in 28m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ee05c058f9314aaeb5fc9643ea9cd9c3 : SUCCESS in 12m 30s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/7889ce2bd63a447aa240609721f6db01 : SUCCESS in 16m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b9c364611fe4203aee2438ca91098cd : SUCCESS in 14m 22s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/1f8a400a4f6c4ac3b5121d36ecc569bf : SUCCESS in 12m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a7bfe3b63bda45bfa2980e27640f731e : SUCCESS in 11m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3af53721b68344fd8365a3b0c73d2d07 : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":2},{"id":"15cc80dff1524fb1d26f6a04ca96b5fca0d63828","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-31 14:32:28.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab3fe300ce4ba1d547d17537416d21ae0b4012bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-06-02 08:03:31.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bdc648c6b642aa6cf1bf415343c42d8faf260855","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-06-02 08:03:44.000000000","message":"Patch Set 3: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"313586ac0bbc0c4fa5ee99dc097785a2056b2f19","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-02 08:32:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a9e0ff21f5289c575f297fdb901c89393a4c55fb","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-02 08:35:46.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"52a1f1331e92fb20ba048408510f5dda6da820f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 10:28:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/897db3cf9d2244f6a97d318f6bfbd834\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf48df0627784191ab1b6551d2a9b731 : SUCCESS in 8m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f3101d988e904a13a79648406e1d0616 : SUCCESS in 5m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df69c253e712410caa72764c4f879ed5 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e63b7b199454ddc9206d823b0aa6882 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45ff9787a79244dca03d6d47f5aa08e6 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f2580ce5f9d409c923ccaca9d8a3c24 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/994fbf7a9c304507b1ed7932c10485a0 : SUCCESS in 5m 28s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/326656ab920e4443bf9cba55badd55b3 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f7a35c24122d425f883e8420f42c425d : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c8063280d4648c38bfebd60e77fe23d : SUCCESS in 8m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0183a6a3bb6f47fa9e6d4bad5c1784fb : SUCCESS in 53m 48s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9957501399ce4a7faea38f12dc76103b : SUCCESS in 1h 02m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f201fa018c74bfa81b80e2a4025d649 : SUCCESS in 54m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8b8b13882f8e4112b650ce0339bac54e : SUCCESS in 56m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/eb98db21eda14eac877a6ab1206c4758 : SUCCESS in 1h 10m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a9eb9d174ea8422384e9015af25b02af : SUCCESS in 16m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb5a44b6d39646f1b32219e820d8185f : SUCCESS in 16m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/663f4d71b75045f98114403fba9cc01c : SUCCESS in 1h 51m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0228e6358faa47f59aeb31348c89eaa1 : SUCCESS in 1h 49m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bce87b162e0b4f5ea1f972a0ef3fb44e : SUCCESS in 17m 53s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/0c4c3c7cb57847e086fe856d7a93c369 : SUCCESS in 16m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/726b57cd8a31486f9c32c7db39cea5c0 : SUCCESS in 18m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6eb2276987a14d95bad50d4f5815d63b : SUCCESS in 15m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/26eac4d2697a4afa900465325acd6657 : SUCCESS in 41m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9e882571c15a46799385967cd9462e38 : SUCCESS in 30m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fcf58f83b0243c5b83e60c180082ae0 : SUCCESS in 26m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/381b137fa2f34a3f85659c75e878d6b0 : SUCCESS in 11m 51s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/623539b9f0b747428ead33e0301d6654 : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77fbb0c1d0d6498e9adde7c4fd464d8d : SUCCESS in 13m 23s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/6b1bc03fa6ac431c8146f70cab2c1678 : SUCCESS in 11m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1fd2985742654d8c9deb101f3f81a9ee : SUCCESS in 14m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21b1ad9f9f1c475abd42d3a860856955 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":3},{"id":"851c8aa24766ebad1e798e3f52ac8467674c1e9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-06-02 14:34:26.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"9f7a42a0004ac4304145c02927b9b9fb1b9472a6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-02 14:40:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3a65a1ff82ba667232a1ebc97fa7ecd33928a2fd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-06-02 14:59:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8cd6756cecbd9daf3482f0525b6ed927ab86c307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 17:00:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee2f0814d38d4de1885dbc47a8851663\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58a5da2f460d4e198540a358d021e20f : SUCCESS in 7m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a1bf2a5e1de247109dbe35aa8e33117f : SUCCESS in 5m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de1485a6929d42f9b2a15dd431ee8940 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3c811570c647d484df375e306f7f78 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a69e453b51ee4694a00d2777d2b676c1 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6292ef4e123344f08909673125a114a4 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95057fea0e20460598e2751f3d933596 : SUCCESS in 6m 25s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/038df46f4e9e40bf822204b9c3c0919a : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b50c76bef3454762a2b1446b6d1e93c9 : SUCCESS in 10m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/480d3876d9ff4551bd53474dcbabfa45 : SUCCESS in 9m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4c4073a51d484afc83f322f03c29756c : SUCCESS in 58m 46s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f716f41f52df44e696a41ee3e4264818 : SUCCESS in 1h 01m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b906f9b213d3497a95ab34b05cf0de1e : SUCCESS in 57m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6bd25776ceec4366bf1c6c7f27967457 : SUCCESS in 57m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/29905906e2304d389d89b3661b84017d : SUCCESS in 1h 10m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e02c6007d5a74a9cb9c0e77d511e1f9a : SUCCESS in 18m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4452db9c6944e6ebd3f38f4f44c2706 : SUCCESS in 13m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b910c752d4a648c1a20e350e7ecd27e2 : SUCCESS in 1h 45m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99a3c145327241ab8c43bbc201857109 : SUCCESS in 2h 02m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/be393a2f9ec849bcb6512e30e49f2232 : SUCCESS in 14m 49s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/62cbdff3383e43e0837411af40944e70 : SUCCESS in 16m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b0d449548644988ae4cde28677e9c9c : SUCCESS in 14m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fe65b20d6f974d1d9a75f6a752b1a08b : SUCCESS in 16m 48s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f9262b89d00944758bf3aff13799a899 : SUCCESS in 36m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9a5443be01c44bbb937496ac30d4949a : SUCCESS in 33m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10b34c505fe2405da12bcc7c6dc425e3 : SUCCESS in 28m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3d4ec7002c634fdd935e08e9fac70dd8 : SUCCESS in 14m 48s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3c3a567d26744b769416d34a70f576f5 : SUCCESS in 13m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe6de21770414544aa1ca84b8d4017b5 : SUCCESS in 10m 13s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/e9c68f5e6df649d8b5403a5190874de5 : SUCCESS in 10m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bcb3e7c5ba2344f78f482eac6df46038 : SUCCESS in 11m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f37cb74dbe6549258376dc87bfb1856e : SUCCESS in 11m 37s","accounts_in_message":[],"_revision_number":4},{"id":"43c51c08845b991fc7cacc8cd636b761c042fa5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 17:01:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9395fbbd2c8a28164764213252cdcd7fc233e33b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 18:49:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b3681861e7545e5a1854f20e20e6c6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a155d219e0484466a8aeb4af2b6227fd : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/460e23e86e2249ef86922e8b8f81fe8c : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1aed4b421004b8ab6608d924ed8d542 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30ba289a549243c59433e35221d329ef : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0cb9a33d016a44278454c8702ed0dffd : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/88c04ca2d46d4c62b6798decc8d12c55 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3dc4c611302d426d95c778b17e09d792 : SUCCESS in 12m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/39f08d7c5f5244bcbfe897fd773b6ee4 : SUCCESS in 9m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dccf689a684c410f93f89d4fe34284f2 : SUCCESS in 58m 11s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1b44f9b4289e443b8f7754ecc3110667 : SUCCESS in 1h 01m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd88851e89c94926a958aa7128a18fc2 : SUCCESS in 54m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/045d36437640453f96f9f09cbd2ff43b : SUCCESS in 1h 06m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/202a83a914334689a0e4cb5fc046c5e5 : SUCCESS in 1h 11m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b3590814735544dc9c91b99e42a252a4 : SUCCESS in 16m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b07338746f74d6096a173f9cc8acd04 : SUCCESS in 13m 59s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6308253bde6e4dc5ada2541dae1ac85c : SUCCESS in 1h 38m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1e23166f4d664fe5b7d92ddda9f330f1 : SUCCESS in 1h 43m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4d16ec2f8caa489185a2c9c0ebfb0e82 : SUCCESS in 16m 03s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/296b32bca9e1479ab3b6cc54a8173b44 : SUCCESS in 15m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85625a53abc040398454245c8bd8b672 : SUCCESS in 13m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ad6312489c7f4056be98d8c843867863 : SUCCESS in 16m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/64df5111e47f4c8580d15fc9cbd5fd1d : SUCCESS in 27m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d405f02fcf5c4c88b03519c07c268601 : SUCCESS in 32m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/240f0c71dbd6485f958731799166f024 : SUCCESS in 11m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce21610f0bd349af8301ba2ba6a895f1 : SUCCESS in 13m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/09c8780f07194ba599d968d5d3b9972e : SUCCESS in 15m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33dae07a47a545d8afe1eef0f792cf42 : SUCCESS in 14m 15s","accounts_in_message":[],"_revision_number":4},{"id":"6dad120fe72cfd39ac56ee6f485d61481294dcb2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 18:49:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6a01b82f3835221ec52a40713bb61ba510b28662","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 18:50:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dd826f7495149609b757bdbf976cf8d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/297646b43b464760b7b94e17e00480ae : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9554ddd2d339e27b68f360425beb11830b370357","revisions":{"e38796cacb090f620febf0056b5a1a79f3fd5ca2":{"kind":"REWORK","_number":1,"created":"2025-05-30 16:13:24.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/25/951425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/951425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/951425/1"}}},"commit":{"parents":[{"commit":"2305e75c9420b1a78f913a7ed33f25d4cdc636a1","subject":"CI: Bump ansible-lint and prune skip list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2305e75c9420b1a78f913a7ed33f25d4cdc636a1"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-05-30 16:03:44.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-05-30 16:07:19.000000000","tz":60},"subject":"Make custom Gather facts run with default python for kayobe-target-venv","message":"Make custom Gather facts run with default python for kayobe-target-venv\n\nThis change make custom \u0027Gather facts\u0027 task for kayobe-target-venv run\nwith ``ansible_python_interpreter \u003d /usr/bin/python3``.\nThis is to ensure custom \u0027Gather facts\u0027 can be run even when there was\nno ansible facts gathered before.\n\nCloses-Bug: #2110957\nChange-Id: I2409428e7b929bec32c2b0034e82b4d03b343567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e38796cacb090f620febf0056b5a1a79f3fd5ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e38796cacb090f620febf0056b5a1a79f3fd5ca2"}]},"branch":"refs/heads/master"},"cd323805b20853860e2f0e0a07ae18d5200f0581":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-30 16:14:10.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/25/951425/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/951425/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/951425/2"}}},"commit":{"parents":[{"commit":"2305e75c9420b1a78f913a7ed33f25d4cdc636a1","subject":"CI: Bump ansible-lint and prune skip list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2305e75c9420b1a78f913a7ed33f25d4cdc636a1"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-05-30 16:03:44.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-05-30 16:13:57.000000000","tz":60},"subject":"Make custom Gather facts run with default python for kayobe-target-venv","message":"Make custom Gather facts run with default python for kayobe-target-venv\n\nThis change make custom \u0027Gather facts\u0027 task for kayobe-target-venv run\nwith ``ansible_python_interpreter \u003d /usr/bin/python3``.\nThis is to ensure custom \u0027Gather facts\u0027 can be run even when there was\nno ansible facts gathered before.\n\nCloses-Bug: #2112201\nChange-Id: I2409428e7b929bec32c2b0034e82b4d03b343567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd323805b20853860e2f0e0a07ae18d5200f0581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd323805b20853860e2f0e0a07ae18d5200f0581"}]},"branch":"refs/heads/master"},"0a6a5e6584911fdc89b39b20b30eb79e62120f60":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-06-02 08:03:31.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/25/951425/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/951425/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/951425/3"}}},"commit":{"parents":[{"commit":"2305e75c9420b1a78f913a7ed33f25d4cdc636a1","subject":"CI: Bump ansible-lint and prune skip list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2305e75c9420b1a78f913a7ed33f25d4cdc636a1"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-05-30 16:03:44.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-06-02 08:03:08.000000000","tz":60},"subject":"Make custom Gather facts run with default python for kayobe-target-venv","message":"Make custom Gather facts run with default python for kayobe-target-venv\n\nFixes an issue introduced in I7cfb600eec268749ea1315112fe3a9429ab3ab8e.\n\nThis change make custom \u0027Gather facts\u0027 task for kayobe-target-venv run\nwith ``ansible_python_interpreter \u003d /usr/bin/python3``.\nThis is to ensure custom \u0027Gather facts\u0027 can be run even when there was\nno ansible facts gathered before.\n\nCloses-Bug: #2112201\nChange-Id: I2409428e7b929bec32c2b0034e82b4d03b343567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a6a5e6584911fdc89b39b20b30eb79e62120f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a6a5e6584911fdc89b39b20b30eb79e62120f60"}]},"branch":"refs/heads/master"},"9554ddd2d339e27b68f360425beb11830b370357":{"kind":"REWORK","_number":4,"created":"2025-06-02 14:34:26.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/25/951425/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/951425/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/951425/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/951425/4"}}},"commit":{"parents":[{"commit":"2305e75c9420b1a78f913a7ed33f25d4cdc636a1","subject":"CI: Bump ansible-lint and prune skip list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2305e75c9420b1a78f913a7ed33f25d4cdc636a1"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-05-30 16:03:44.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-06-02 14:33:53.000000000","tz":60},"subject":"Make custom Gather facts run with default python for kayobe-target-venv","message":"Make custom Gather facts run with default python for kayobe-target-venv\n\nFixes an issue introduced in I7cfb600eec268749ea1315112fe3a9429ab3ab8e.\n\nThis change make custom \u0027Gather facts\u0027 task for kayobe-target-venv run\nwith ``ansible_python_interpreter \u003d /usr/bin/python3``.\nThis is to ensure custom \u0027Gather facts\u0027 can be run even when there was\nno ansible facts gathered before.\n\nCloses-Bug: #2112201\nChange-Id: I2409428e7b929bec32c2b0034e82b4d03b343567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9554ddd2d339e27b68f360425beb11830b370357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9554ddd2d339e27b68f360425beb11830b370357"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
