)]}'
{"id":"openstack%2Fkayobe~951720","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I2409428e7b929bec32c2b0034e82b4d03b343567","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-06-11 14:21:16.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-11 14:21:16.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-06-11 14:21:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2409428e7b929bec32c2b0034e82b4d03b343567","subject":"Make custom Gather facts run with default python for kayobe-target-venv","status":"MERGED","created":"2025-06-03 21:11:45.000000000","updated":"2025-06-11 14:22:10.000000000","submitted":"2025-06-11 14:21:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951720","cherry_pick_of_change":951425,"cherry_pick_of_patch_set":4,"meta_rev_id":"caf2e454a1e1a2d4ec020204673dc02625e52ffe","_number":951720,"virtual_id_number":951720,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-11 14:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-06-03 22:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-06-10 14:30:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-03 21:30:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-06-04 08:05:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-06-03 22:38:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-03 21:11:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"CC"},{"updated":"2025-06-03 21:31:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-06-03 21:31:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-06-03 22:38:57.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-06-03 23:47:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-04 08:05:01.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2025-06-10 14:30:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"5fa413e822355f49c27a56f99d00de3009373dc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-03 21:11:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8159e3bd4acf25319789739cf43602229dd4f8df","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-03 21:30:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"87b3dc7ad622137c93016c6f892bcc4214f46011","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-03 22:38:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8de07a94fcef40baf3e15ab5a01c6181fa96c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 23:47:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b0f227584fb430291954fe591384ce0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92d90b2581c44a53b687a9504f47e8b4 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/591f542ac09046bc83e8c34f6e1c0580 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6fd77684f2924056affee1ad036e1d57 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fec929a8ab649ee81ab6eb91ac2c777 : SUCCESS in 7m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/167453024015423bb0f689cd3b0e9647 : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/388bc57f598e4a16b2e6bf062974610f : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c6737f3266148f683acac9cd08cf56b : SUCCESS in 8m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5ae27f18145b4e20a701cd1302b4137f : SUCCESS in 46m 21s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7330a267d9144f1cbb746993de3aeb4e : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fbbba5539cc4868bf1c05f06650a8c6 : SUCCESS in 1h 08m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3c468abc5b524031abfb10c259ba887f : SUCCESS in 56m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b515c0de8ff54d1f893e6b5cb27eb95b : SUCCESS in 49m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e5d4dbbfdb6e4eca87622d5809503907 : SUCCESS in 20m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bdcdea32f7f43c9be399ae630f03717 : SUCCESS in 12m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c2f9427678934591b18814f2090e3f8f : SUCCESS in 1h 46m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c05780f6abad449a978c8194da5fae50 : SUCCESS in 2h 28m 46s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5a8c279be68347ad97aded4288415c47 : SUCCESS in 1h 51m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2dab11af34cc4c8e8870a8c30410e8a3 : SUCCESS in 2h 05m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d6a2614bfac84a00a54a790a2227273d : SUCCESS in 17m 49s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/345c2c8c741b45468f2078217e0c51e6 : SUCCESS in 14m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4c452dd698b4adb8855e23cdbfcf7e4 : SUCCESS in 12m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/79466d3eba6c407cac1f1a45d27ebbdd : SUCCESS in 11m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8cf0a68a19544e7f89a962b8be003725 : SUCCESS in 30m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e49ab529e71149e38b09219821b50bb8 : SUCCESS in 26m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df43f40edd2245388d08b5b3f0fa1780 : SUCCESS in 31m 44s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/11a87d6881b440eaa56353f74013aa3a : SUCCESS in 28m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cd935802c9914810a9c7a5a7c89becb0 : SUCCESS in 11m 38s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/310c7f0c3ee24459a763ea6e34f93365 : SUCCESS in 12m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a316660d7bb249939a5b762544ced102 : SUCCESS in 9m 09s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/46231f70d1bb4b41bdb82d08a58aa844 : SUCCESS in 8m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fe9c1a74a18548348f1ad3c16516cedd : SUCCESS in 12m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2386e53edc34e2c88db2d5e3577eaeb : SUCCESS in 10m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2400e23fae184c5c23da2984483fa5de3c05eee1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 23:47:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"271e3d04a8a5fcdcb917e45e28fc0475161f21f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 02:46:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab43269d2bfb473baa61c08ddc1d03b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a03ae2cfe4bf4becb92cc3b5324d5b1b : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/506568dbe03d4ca6901dc329651127c8 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6292bb76c0a94960801e4646dac0472a : SUCCESS in 9m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/30ddeb48bfd046f28dcd0450ed91b40f : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3d141b5e4bb941efbcbb424d032f497c : SUCCESS in 12m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71bcdaed7ab941e9b8160873e74805ce : SUCCESS in 9m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2b68df944cea4a2d99c46ba433e9da8b : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/4a31400dacb3470dab78bd04dc626a4b : SUCCESS in 48m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f61feb238ef945be8e69d94bd1747a8a : SUCCESS in 47m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3f54a1089471472d9496160d75987f47 : SUCCESS in 52m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c6ef840bed8d4d92bd46e48fec245ea8 : SUCCESS in 56m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a2e488324050466e91e7f4ddaa390df4 : SUCCESS in 16m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdf21c233cf04e818f0a69ef80aa8dad : SUCCESS in 12m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2f1dec5b331145998a5d5643e1ae9539 : SUCCESS in 1h 47m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/797d15f4249c4d66a656042a0199d608 : SUCCESS in 2h 19m 53s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ddf823fa458e47a3958ef364e1597795 : FAILURE in 2h 06m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/482533f7e42842698bc369b043814324 : SUCCESS in 2h 26m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6d860b1338e541bfab450d684119f610 : SUCCESS in 14m 25s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/704c6b71cb0045ec81fe2d010796f646 : SUCCESS in 17m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/573a9cd72d9e4eb48dfdd52b641b11af : SUCCESS in 13m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1176a6d0d7a94c468c40ca76ef5638a3 : SUCCESS in 15m 07s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/85883c65d2504508bd9229814f3bbb02 : SUCCESS in 37m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0a964dcafe54d7da8d22ae3f25eb8ea : SUCCESS in 37m 01s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1885b2a063f64da183a67c02c3e74816 : SUCCESS in 31m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3f0401cc096142c99603c3721943e857 : SUCCESS in 12m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea66cf9395fa4dcd95286c855cfc5f2f : SUCCESS in 13m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/29e3ac866a134e73af63dd08ad3f6934 : SUCCESS in 14m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/179adf60e0c6411c933de94e1eae8893 : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1b9806248d394dd20c0c26e0869acb6b7e6a8d73","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-04 05:27:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"579f477b05317413efc6d16e610f26148abac86c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 07:56:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e891d368eb543e1b23295b06a5e9eab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b9121bc4a744513917a3e2cd0359501 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b5458c656b44c3946fc4cedce7836c : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/946f88d063a44adebc2a35daf49ceba9 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ae24e0b33741be8146e9152b24ce18 : SUCCESS in 7m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/520159b9afcc415db12e3a13e10c7d4c : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fe6b3d86ea1341e69f4067758f1b6bc4 : SUCCESS in 8m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d2691521c27d4712a7b1d59506375af7 : SUCCESS in 9m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d07e618092f94f948a76e44bb3b0351b : SUCCESS in 55m 12s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/38b6cfd1d3de46fe81841f95808279b8 : SUCCESS in 1h 01m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1fdc23c066ce442dba090972eeb10d1b : SUCCESS in 55m 32s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8e6955e03d2045c198709fedac06ddb7 : SUCCESS in 58m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/66d75beb941646399a8c350618d80329 : SUCCESS in 1h 16m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7bb8a54860fa4a099320ba0148efb1fa : SUCCESS in 14m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/108ebee4d4aa4f48ba9a2fd108ff0fb8 : SUCCESS in 18m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8e642b71928e4f5d8f68b01b77ae6384 : SUCCESS in 2h 05m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0bf467f1ad774da798fcc15a8e1d6b91 : SUCCESS in 2h 23m 14s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5b30097f3ade43728e9f8bdea97ae383 : SUCCESS in 2h 06m 58s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d97ea626bfd49818cb4c301ffad2684 : SUCCESS in 2h 24m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/40d8c36f1f38442b8ed327ba5140528b : SUCCESS in 23m 28s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fd602df280674456978099e470fbe45d : SUCCESS in 16m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7995397632fc454b9c2d413efdb45719 : SUCCESS in 13m 35s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/721858c3dffc495589b8506906bebc3c : SUCCESS in 13m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6b73391c02b149f79cc6d8a4fc7ee036 : SUCCESS in 34m 34s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2660d0cd76204b38aaa9d7e7b69c1071 : SUCCESS in 34m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86d3ab40788e43d1a824957f9f913f0c : SUCCESS in 27m 30s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f8c979028ce24b8da957a667542d52c6 : SUCCESS in 29m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1850ed93a87c4f18aa0c8409e4b3218e : SUCCESS in 11m 23s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/88959cc0e12a4143a24182afa2a05824 : SUCCESS in 12m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/537fac71a34643fe9c510985377ab337 : SUCCESS in 10m 53s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d733d1da718d49c9a3b1a2a09bba502c : SUCCESS in 9m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d094cac58c3d4ce981eb5d70c46f6a82 : SUCCESS in 10m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2d6f8fcd0914fc0b98562915517007d : SUCCESS in 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e10f230f15ea3f725a9b8fb065c5f0563a560dcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 07:56:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"783698b803d5813c6a34e2803baeb32b0ae9d668","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-06-04 08:05:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e36357b433e325ab4ca7b07a232dd33831095116","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 10:31:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3cfe8b190424999b5069e5f024d6796\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b10ba190e38a4b6989a57854de071a9d : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1687707222e4477a9e50f7b33a8baae6 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32cd593050ac4d438eeef8b5994e9857 : SUCCESS in 9m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe801b11b8df49599c122e4ccd7e39bf : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/70603516c6cc4d5ba0360c06594c84e4 : SUCCESS in 11m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b081d186e3e4e899c4a2d7a0e21c305 : SUCCESS in 9m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/58d3a7ef51694dcb8d1beb5278ad0c6a : FAILURE in 8m 43s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/9f895792fa4c4fa7b91ea43da1a14955 : SUCCESS in 1h 08m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a53a1569c87e4242b6b62b3c3eacfd26 : SUCCESS in 1h 07m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f06c0b8c28c64a1dbe4d9c1af6c3fdf1 : SUCCESS in 1h 04m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/948e52ff37f144d1abc970b4a5cb2572 : FAILURE in 14m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0274c2b67b104d1ab05a9fd4f46d1916 : SUCCESS in 18m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a23faf5d3b9408e9e3242b30d6ddc55 : SUCCESS in 16m 48s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b578c457953a4fb0b8c0bcdd467f96dc : SUCCESS in 2h 10m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e20dd005e4743c399103646d6c9445f : SUCCESS in 2h 11m 53s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/718688eeb65a4171b0552cef22e62477 : SUCCESS in 2h 26m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65a9da2cb7774521841ad91285ae2e5a : SUCCESS in 2h 15m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/733fc3d71cf441a29302761e63c5031c : FAILURE in 11m 02s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/482a5c1d9761498e9034cf896734dcee : FAILURE in 9m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a31cf302e4e4f2a95120925898f6d5a : SUCCESS in 18m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6fa9f75c128a4f1e9eee3eb5f8d9ac84 : SUCCESS in 14m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a7c4b0cb440a4183adad556fd0c16258 : SUCCESS in 34m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee9310bd517040fab90eba62437b9c84 : SUCCESS in 34m 33s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3d142ce60ba44ce3a4aaac47fd273014 : SUCCESS in 28m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/179ee6f53d3d40cb9e43fdb843e36cc7 : SUCCESS in 15m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80c5426f62004520820b597b019c8146 : SUCCESS in 14m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7fc8222828d34efaa787cb7ceb0b8a99 : FAILURE in 9m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7da00de52a3a4c67820d914986d4af6b : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e9374b419ac2594ba6f6f4912a2ddad44f74947c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-04 11:11:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b274fdd53a5a0f398be729753535ba0a72decfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 13:19:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/817c3baf46d047598f8888db23b0315b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b77722f57694b20a97c147a54dd08cc : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d253aae9fb2c406cb23d9e5f4ab6d6e1 : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3b7ca5ad7254736a3e1469028d1b6fc : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/887016a035b544bc8454f4966edb3515 : SUCCESS in 10m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/267edc7ab60a46c28d469dec583f9ec8 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/82463356947e4358b3cb040408c69efd : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd9abe0c0e2b4a2b854d11694138571a : SUCCESS in 11m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b8f35014f0484df8a64e8dd7de3b0f83 : SUCCESS in 1h 00m 28s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/21d055fad05845848a675585cb9d6e87 : SUCCESS in 1h 00m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b92419204a149439dac9fc3b7ee7ede : SUCCESS in 50m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8c144c0caaab49bc86144efa5015ec1f : SUCCESS in 56m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4a708df4609b4264a4c9d7767a2391ce : SUCCESS in 1h 08m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8e59f9b738ed412fb40829cd6a98bbfd : SUCCESS in 16m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/091b469480aa49e0b36b14f2864f2bab : SUCCESS in 13m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2ca0fcd9600c4d7d86144317bd794bc2 : SUCCESS in 2h 02m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb78c8710380402a83f54ded1c16ebba : SUCCESS in 2h 07m 39s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9bdae9f627f443688f2d8c62e8601fac : SUCCESS in 2h 06m 05s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2b45a657c2a4a1090dcc2a4be28abc1 : SUCCESS in 1h 59m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d9c6c17459a949549442877e2753e27c : SUCCESS in 14m 46s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/16e7d3120cf644e1ac8d5b5805c13cb3 : SUCCESS in 13m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0c49a0b1cbd4575ba0e3f9e64310874 : SUCCESS in 13m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/103076672dce4038be2622d22fc0cb10 : SUCCESS in 13m 49s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fdd02d880edd4a20866fcda93676d839 : SUCCESS in 37m 02s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/397b39663d6746b1827fdb6e113abc78 : SUCCESS in 40m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2c16c3f7e81455e8fb8ea5acace0fa8 : SUCCESS in 28m 57s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d24f888fbe8b4321ae9f2365a003e297 : SUCCESS in 33m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fc1a3b6d1b864269bb04d7a4cf507202 : SUCCESS in 15m 36s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/0ff98017281d469d90558ecb369d6124 : SUCCESS in 13m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/500147db66794f1bbeb556c3e0fe0406 : SUCCESS in 13m 59s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/b5e24710cf0d4d79b582e229b607e1f8 : SUCCESS in 13m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/717c4a5d9af24b7eaf967dd555b3a653 : SUCCESS in 11m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cbf733432a845acaca8a3212837d9f4 : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5eef4f96e902cc8b2864bc6cfca63c7f33518b56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 13:19:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a46393a7e0eed9e9c2552f433c3b64b1fb9cfeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 17:18:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7775999c8e34dc2a8991e2120133255\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b903afdc385424ea17231d45507a7ed : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7305d8a86d246949a751994b2bc2044 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a6d44716564d5d840264524668a983 : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a7887310ba134151aa30a395e6564ae7 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c19f8075643142f3a7a92e205f743117 : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8b5f811991d444fa8413a751a1e00e77 : SUCCESS in 8m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/58c7cea8b346434b9ef6b27acb7e1d96 : SUCCESS in 1h 00m 17s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1f5fdec929a24037aefdf32a9bdab598 : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10ae466feeea43d6a57befb85f5a931d : SUCCESS in 56m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9347d1bfb86649c1b254a1cb9460cfd3 : SUCCESS in 55m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fd674d4d329741ee9519344d8bed8c2e : SUCCESS in 1h 07m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3ab17b2cfa39451ca29e92e0f64fa0da : SUCCESS in 15m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e321fee8ae17450fa1acf6e47d3a62a3 : SUCCESS in 12m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d7cbbda41fa247b9ade7e14c86c1c933 : SUCCESS in 2h 17m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b40e637defab4230bcbfd66d817d54d3 : TIMED_OUT in 3h 02m 30s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/79ea80c1ced044f9a5f03ddb44985c9e : SUCCESS in 2h 13m 42s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1832ca87ac946d19f378c93e887067a : SUCCESS in 2h 11m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7880a6ae835942e58c39db713508299c : SUCCESS in 18m 49s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/f0fd13d6bc1542cc973e97a5e32581af : SUCCESS in 14m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/854729eba8f940d58a1a0a777cadb89f : SUCCESS in 15m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/146f77c3130d4570829bb1485ad6c3bd : SUCCESS in 14m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/34a96c5ed416448183f74a0fc3dc53da : SUCCESS in 32m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23c8b15998704e89a54ebeb2efc7340d : SUCCESS in 33m 48s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a71f21c6d01b4429a53fd4b8134d7e3b : POST_FAILURE in 33m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f046831fd12241dba602934780e71e23 : SUCCESS in 13m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e6c5cd1e1ca4f91a43437479f374e2a : SUCCESS in 12m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9c57f45dde694eb6af1a6ac523df0259 : SUCCESS in 12m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/215eacd7012c4681a3c2e41a924017ba : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d8e04dea2fde1a94f08ea31d466aa371f698fc37","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-05 08:51:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca4f3b960d5d3c4b2bab06949bd477c77942fdb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 11:21:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74a6418cf5004d0d9218ea12b147a732\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/853de521448b483cb5ef7e2e3a25f3eb : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7333e4eb30214104aeeacad63a68b0cf : SUCCESS in 2m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e7e09a0762743969f2ce679ec893355 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb767933a2b54ef7b4add97adf140a52 : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e8ac6f04e9f24ca383800d80815c4959 : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2baf9d325a3e4de09e240566e5198b6b : SUCCESS in 8m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef5daab2520549cbb1dc2a5f5b873d30 : SUCCESS in 8m 40s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7b1bc8457cec46fe9629a017b6ad6573 : SUCCESS in 29m 55s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/403e77a67fdf4aa6b98cdd587d5d887c : SUCCESS in 1h 01m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf9d46affa5741d58efa27cb05614027 : SUCCESS in 49m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/38b58a2a85c74a94960c4933e978efd6 : SUCCESS in 56m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/33fdd272d9e14587aeef46d44d23002f : SUCCESS in 1h 04m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/95b04912493341beb7e1e8fdd9968f8c : SUCCESS in 20m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd2c0000ccc645108178cdc2c45dcef2 : SUCCESS in 12m 11s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a03ca94f6f7445ed92d895cf6e4860c4 : SUCCESS in 2h 08m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c5db48303c6420ebd6ece4359123421 : SUCCESS in 2h 24m 20s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a24ca605aa5a4dc79349c8baa1688824 : SUCCESS in 2h 16m 14s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa63f83135cb4eb694ea5b74b83aa81f : SUCCESS in 2h 27m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2add07a2d6604bc999ea9332a293ab1c : SUCCESS in 21m 39s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5fea9394f40943af919561ceef08187c : SUCCESS in 20m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44499fac9e93435f89c7a38562b1e28b : SUCCESS in 17m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/50b32b4d3316460bb4b37112dc13eeb4 : SUCCESS in 16m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/80b6f66a5e4e4e4cafb4b48377159065 : SUCCESS in 45m 38s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/019f823bf2d34e8c917fc956c1c18c12 : SUCCESS in 22m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87e3d038fefe4abbbc7a68da130087e0 : SUCCESS in 25m 22s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/30f0c88408e94a16a90e0fde9c075e57 : SUCCESS in 35m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/adb1f05b89154a01b44d840bb8acfd65 : FAILURE in 13m 11s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8ce50518736244619c59e7a938771a99 : FAILURE in 12m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec15a97468cb450699246125de703333 : SUCCESS in 5m 45s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/e25c83b1fd7f40dca50d0bc50223a034 : SUCCESS in 15m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9f8ba6cdaf2f47029007f7dc75ddd09c : FAILURE in 12m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e3b75bcf08c42d0bf9881207f64893d : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":1},{"id":"06b0a8376d650f36be6a81bd1a2c63877c49b90b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-10 14:30:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c411a9146a49bc03d93dab27c309a8a269d1e4aa","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-06-11 09:09:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e827ccd516799b0569de7576d4d698cd4e61368f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 11:36:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/595ef7db6c554c0f8e6392c81346018b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b29c07369a394796b202654ac0d17cb1 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a32291d6142c419896aabb2cc934d0aa : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b27128eeb9064f40a5677bc163dc20f3 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaccdfc00e6a48fcbd743d9e28f3a485 : SUCCESS in 10m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d43c4a89b3f493181071269ec0f55fb : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/33b6171b634c4aec8ca8ebab8ee82b32 : SUCCESS in 12m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/41ef79a00b174c188f83c7034ae929c3 : SUCCESS in 10m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e760060e899e4e7fb3b647be67745b1e : SUCCESS in 59m 07s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/7aeeae22b5c9453dae02cdb08a4eb804 : SUCCESS in 1h 08m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb78276f5775463bb1a8bd3a075b5a2c : SUCCESS in 37m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d50325da87c2495e91c713e4c98a5772 : SUCCESS in 54m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/923e60e2823247a38be7a3d93fdcaaea : SUCCESS in 41m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3bbbdb7e21fc49a78d26f98b623345ca : SUCCESS in 11m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e52779c4de794fc88cf5ef94f32c0677 : SUCCESS in 13m 43s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/096ad81c16ea4774b3890bb54ff6dc95 : SUCCESS in 2h 05m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab654384fcd14a54bf160b5d01391665 : SUCCESS in 2h 20m 08s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d578cc6d102a45fca12b27dd4f2285f3 : SUCCESS in 2h 05m 04s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50806f1d9786418ab750108000ffefe7 : SUCCESS in 2h 04m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9cfd991eb1534f81b928e07ed8afbd6c : SUCCESS in 14m 52s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/32a9c9c493314b5aac34f60da896685a : SUCCESS in 16m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/61df2d306346476aaf6e3550df0aa651 : SUCCESS in 17m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f3f5835f8c1f4fc4931fc2d4ed87f360 : SUCCESS in 13m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8211c611e9154692a5d00b8daa3abeed : SUCCESS in 39m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/84335b3aa2244271bb59cfc4e93913e3 : SUCCESS in 33m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36009f156cb644b7b9155d374aa9d7f4 : SUCCESS in 27m 26s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/89f3d4a3d4d6460083b0161041a90788 : SUCCESS in 30m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f26be41f58454a18a2931c6362cf2f0c : SUCCESS in 14m 14s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1fcc1dfc276a4a5887a1344d1dbbd50f : SUCCESS in 13m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80e7f75fa9654544bb1b35873453dc48 : SUCCESS in 9m 45s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/6efe69bf28414cb0b05c6de0a88db47b : SUCCESS in 7m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f549e8710c4b4098a06cb6871d63bd09 : SUCCESS in 12m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/801c6c501433422fbc8e8c07ce9c8c26 : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8e4583040a11929ac9e4f22a5dffc7eb3e7ac365","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 11:36:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4010263ff86fb67378193c012eded8bf5a6fe4eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 14:21:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8878b7b5aa34748843477aad1f8b1a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60558325355f4186b53bd79560a04717 : SUCCESS in 2m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06108cdddcff433790079b11e4912d2d : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a1ab7684c0e466ea862358f99376adc : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f2006041f03461cb202b765d9dd8606 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa5e8aa38cc948078818ac4eeb536100 : SUCCESS in 8m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dcf95308ecd84bb68eb3539660f7739b : SUCCESS in 9m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f8473357fef14cc1bd106f9b93edaf21 : SUCCESS in 1h 01m 12s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/c9b7f71cb8c149c2979869e71cee4558 : SUCCESS in 36m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0699bb247266464b82112c90b4111b5b : SUCCESS in 58m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/354fe3e9ce4b4617879e5db153780f94 : SUCCESS in 1h 08m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7b488ff8c1e94085a4b9770819dc8eef : SUCCESS in 1h 21m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/23d321c24b28466da73e9eef1365b482 : SUCCESS in 22m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0bc684cc02042eaa89d631c6ca23fbe : SUCCESS in 17m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9476706cc5fc4c4bb1b90a0b66cdbb41 : SUCCESS in 2h 32m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fb78d26d14c4b1f97e6ab1888a1fccc : SUCCESS in 2h 32m 08s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/26a106af09b54ff598bfb4cb2c0e1a6e : SUCCESS in 2h 37m 46s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4d7187df1a84a7f9ec1cbecd27d7e41 : SUCCESS in 2h 28m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ebeb6394a1284e3a96032a4cc7814fcb : SUCCESS in 19m 02s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/444ed30bfb374b5a8618a7c36f60b11d : SUCCESS in 19m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/115f2242980743008853aead6121571c : SUCCESS in 7m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ed9b834f1c344f4b9a0922573836edb8 : SUCCESS in 17m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/005b989e1b714bebb29103f64f40223a : SUCCESS in 19m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9be79b6bca64419b8fbe1fe4c3a905cd : SUCCESS in 37m 32s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0249173689034503abe411c90ec8a56d : SUCCESS in 32m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7570af72dc9248408306ee3dfdcbf86a : SUCCESS in 15m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c795f165cef4e209e1b67314c535f08 : SUCCESS in 6m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/29e3140b5eae434587618e1a0004e035 : SUCCESS in 14m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dee0e133bfc491184d8aed71790ed8a : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2854b746d77058bc82a0a7cb35cbb83f34d697dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 14:21:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"caf2e454a1e1a2d4ec020204673dc02625e52ffe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 14:22:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/006f0256972549f2b022fc0cd5e76886\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f3e0c481c004727a481bbad32bfeeee : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55e896da62b4000ed00506e46ae4f0e3d37dac7b","revisions":{"55e896da62b4000ed00506e46ae4f0e3d37dac7b":{"kind":"REWORK","_number":1,"created":"2025-06-03 21:11:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/20/951720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/951720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/951720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/951720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/951720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/951720/1"}}},"commit":{"parents":[{"commit":"67cc6af6e4f8222098bcf96c669d1d46b5e5fdf4","subject":"Fix syntax error in drac-pxe role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67cc6af6e4f8222098bcf96c669d1d46b5e5fdf4"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-05-30 16:03:44.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-03 21:11:45.000000000","tz":0},"subject":"Make custom Gather facts run with default python for kayobe-target-venv","message":"Make custom Gather facts run with default python for kayobe-target-venv\n\nFixes an issue introduced in I7cfb600eec268749ea1315112fe3a9429ab3ab8e.\n\nThis change make custom \u0027Gather facts\u0027 task for kayobe-target-venv run\nwith ``ansible_python_interpreter \u003d /usr/bin/python3``.\nThis is to ensure custom \u0027Gather facts\u0027 can be run even when there was\nno ansible facts gathered before.\n\nCloses-Bug: #2112201\nChange-Id: I2409428e7b929bec32c2b0034e82b4d03b343567\n(cherry picked from commit 9554ddd2d339e27b68f360425beb11830b370357)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55e896da62b4000ed00506e46ae4f0e3d37dac7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55e896da62b4000ed00506e46ae4f0e3d37dac7b"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
