)]}'
{"id":"openstack%2Fkayobe~951898","triplet_id":"openstack%2Fkayobe~master~If07b7b5aeff61c4a1181d4e4a6cc256456125c0a","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-06-11 17:58:40.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-12 10:48:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If07b7b5aeff61c4a1181d4e4a6cc256456125c0a","subject":"Use latest CentOS Stream 9 GenericCloud image","status":"MERGED","created":"2025-06-05 18:44:31.000000000","updated":"2025-06-12 10:49:32.000000000","submitted":"2025-06-12 10:48:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951898","meta_rev_id":"bfa204e12e49c9adc1dbcaa7753dcabc115db88e","_number":951898,"virtual_id_number":951898,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-12 10:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-06-12 08:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-06-11 17:58:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-06-12 07:50:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-06-12 08:43:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-12 08:36:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-05 20:31:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-10 19:08:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-11 20:34:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-06-11 20:34:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-06-11 20:34:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-06-11 20:34:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-06-11 20:34:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-06-11 20:34:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-06-11 20:34:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"1b31d61d413d05125cf0a2cd72287d95863ec392","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-05 18:44:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f524f50508668687abd0051800376e418830fcd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 20:31:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a8fece8db8a4a229027b9edc5f279a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a94c90cd974b440c92849568d5279ba8 : SUCCESS in 5m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/511e8b1e9d154e4683734dddb4cb244e : SUCCESS in 3m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee0eac039d9b4e4d9d108cef3e0ebe15 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0ac7ef577444f1e9517aeae7dc12471 : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be2c0672a6f14e3898ba479ffdf1076a : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f74817e6a174299a9f28695a850d297 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/214493cb462e4578bf34a72d7591de9b : SUCCESS in 6m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52525193f4164c9fb19772c8ad0ae145 : SUCCESS in 2m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2602990a0a2c44fca8d26ecda027787a : SUCCESS in 2m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a7feeff33cdf4c5ea21d115e7ced2fc4 : SUCCESS in 6m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1f42443178f4df098a8be6df787be52 : SUCCESS in 5m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/20c499ce665948ddaf2bd5280d7d3df1 : FAILURE in 27m 40s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/b7c63c9844ff41c1a9775d690564a6bc : FAILURE in 27m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a44f5267a7d04146b86381edb5247e0a : SUCCESS in 56m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f6fe84aeceb843e98707668804455b9b : SUCCESS in 57m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9ef010bc869e4cb980d9fc0b2375f2b2 : FAILURE in 39m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1d0f7385728c42e8bb75667f8e0b012c : SUCCESS in 17m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70dd930e36c646ee9144d4829e909453 : SUCCESS in 12m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3d58f078a72a45bb8dea7372e392aef6 : FAILURE in 1h 15m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/024aecc2810a47ec82318cb9dd53e210 : SUCCESS in 1h 45m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a6c1479ac72c4ec29733a9f23695b241 : SUCCESS in 17m 21s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/118183cc2b034f9995760b5c9c7519bc : SUCCESS in 17m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e55684a36224e7699b586df08631f6f : SUCCESS in 24m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5541e45f7ad04d1d9ab679adde706635 : SUCCESS in 8m 29s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5023f82abde045b784cbf7910c85ab7b : SUCCESS in 19m 42s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/96c9163e28ca43e698c858c9278c7267 : SUCCESS in 19m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e51d3a972dd146d69b923cece9d514bb : SUCCESS in 27m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/007ffb6ec41c4550aede4f67dc09d26a : FAILURE in 14m 21s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/3c9eed89b49c4514a84680fb191d5a35 : FAILURE in 10m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3884f98062e4b07b255ae20ad98074b : SUCCESS in 6m 01s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/8b450666ce2d49db9b86b4de4c16e23a : SUCCESS in 13m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a50aa4fcba5a4b12a1b882ece383a898 : FAILURE in 12m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e39f5b9358f4cccafd36761234b4c7c : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b2be1323fd11566ae156eb1953540f2183b0a07e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-11 13:56:37.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dbd1f566730926fb261542a01a8547af167039ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 15:40:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16e97cece0244251979ccbd769b31db1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e89bb5d8a1084120a9706b3ba9b047ab : SUCCESS in 9m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c439727ce9b411bab52f134ce94acc8 : SUCCESS in 5m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdf63cefaaa7408c8fbff96eecd55eee : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1e9c3c2a5534a5bb4808a6491779c2b : SUCCESS in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/241157cfd3ca416581b0cc5ba667e172 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3da6c438839f4a259113feacb13f5296 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cf7633ae7e44c84bcd71507e2f75ca6 : SUCCESS in 6m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3ddbcb9dac64436b96e23456d88f526 : SUCCESS in 2m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7c13dd0ad9864aa2a4368efe6871e3fa : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b9ebf5ae78d64854a45c817467946e0e : SUCCESS in 9m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/908e5f5222144a8495311966d261e7b4 : SUCCESS in 7m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c8c0ca8abe64806b6e12c9a49c40278 : SUCCESS in 1h 08m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fdf8736f90804f5eafab676f8c850c23 : SUCCESS in 59m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d1112bdac4aa473ba94d749244bdc514 : SUCCESS in 10m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e9cf07ef8284337bea0bad4bb20d880 : SUCCESS in 15m 22s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/487b260c183f4cb99529e7dfd4f674c0 : SUCCESS in 1h 43m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fb115da87cad4bad8c3ff898a1e2a0af : SUCCESS in 16m 55s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6b1572af8d4e44e7a3ddd594ef9e1462 : SUCCESS in 18m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b8d0f4395cb49ccaefedb3f191e47eb : SUCCESS in 18m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/39581fd7ec224f2caaa4bf7b72a79873 : SUCCESS in 16m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ca73999a42694faf80161c4568002e75 : SUCCESS in 41m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/985c6164aaff4a9cb381a1ef0d42c183 : SUCCESS in 31m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c87b9b3bf1e4ec3ab5dcc4311a77ede : SUCCESS in 30m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7286417c465c4fe19d70835da80b22fe : SUCCESS in 15m 13s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8e899c7951074b6e898980ed6b06c5e3 : SUCCESS in 13m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f33c6a3948b045109f1c3443df6f755f : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/7cae5720d315407ab3b94b14b21e2c50 : SUCCESS in 11m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7eaa9541a85944a992af89f6d2461703 : SUCCESS in 12m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1d9c10e14e64cedbab4668c0717eb2d : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":2},{"id":"e3ad425380d546eeebab565dc227964f47ca2e78","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-11 17:58:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"02cc0bb22d2638b9368a64b13542c38d613d685c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-12 07:50:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e497be138ec51b19a12969e0a9cb91975015de9f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-12 07:53:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7ae78d5400caa644675c925cd9884337b4fbdba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-12 08:36:20.000000000","message":"Patch Set 2: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3934a28b9e00fb1e7fadd0bf3d8a0e4884904965","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-06-12 08:43:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"74dfdda4bfc91964af39322d42edd2dc713f7f2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 08:43:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8c0788070e9cbb55cb5b90cd8f497a15b3804ff5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 09:08:36.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fc32e0d051a47f48ed9796f7ca0d29d\n\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/706362a4164d4ddbb05979c5a9b230b6 : FAILURE in 26m 38s (non-voting)\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/477bcb205ccc443fbff7c2b59347004a : SUCCESS in 17m 46s (non-voting)\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1b0ffda737444ee7b6797d804b419ad2 : FAILURE in 18m 15s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f5fa174c84054c05ac6a36c2f0004dbd : FAILURE in 27m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/507f7ea07b384e80a05e3ef6302554a7 : FAILURE in 21m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/820dbb6a80d347668b45d285c7ddb14c : FAILURE in 1h 12m 54s\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/af16a0a981494e96866be86a82e31c31 : SUCCESS in 43m 33s (non-voting)\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/8e5e021da9364d67bffedef39dcf9520 : SUCCESS in 1h 10m 31s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a89446d574d144df86c0cd174a9e9990 : SUCCESS in 33m 43s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2ba49aa9cb424349b3b404e7f3dbc425 : SUCCESS in 51m 39s","accounts_in_message":[],"_revision_number":2},{"id":"f3764ca5207140e1576f08da0d8dc689bf8155cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 10:48:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af0e144a924c478ca2fe7787737749bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/677ab945501645b49fccafcc83078dda : SUCCESS in 9m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cac9a17a0d364713916ce38d94b42a01 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d23a31454fd47c9a9223e6382ac4784 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8532ca2960384dde96b3e9dd903cff2a : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/452de721029c4655895c634280454243 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96bbec5adf444b9884b8cd7dda0dd199 : SUCCESS in 3m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d153ba4842ad4074a88d779535385d18 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0cb75754f7b24f52a672fdd6faaa90d7 : SUCCESS in 11m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d0a2f1a48f949c18779f583f047b34b : SUCCESS in 11m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/301d8ab657f54fd88b52bdc9a400aea3 : SUCCESS in 1h 06m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e66bf0bcae4a4da8974e737b9e9bf1fa : SUCCESS in 55m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0a4546cc2e6243b49569304c02b1c181 : SUCCESS in 19m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86530bb7780543d7a8cf18ee14fc410d : SUCCESS in 11m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2500b66e4f0945a8bd27b37f19b0ddfb : SUCCESS in 2h 03m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8e62a301649940c6b14dd3ed616ad4ac : SUCCESS in 10m 23s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/39069d7cd0a4487bb35a0cc79cfff7d1 : SUCCESS in 17m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f725d05578e04b5c97065a8b89d1e7ba : SUCCESS in 18m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cd685d670bbf415b84aadd810de53ee1 : SUCCESS in 13m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/79dace8e32d04d26adb7e7f7071002e2 : SUCCESS in 25m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6d3bb765ec0401fa13beebde0cfed7f : SUCCESS in 17m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/96ad7125b6e244ee8bb4c864bec3d62c : SUCCESS in 8m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4feb59859c144609daa762ad922d47b : SUCCESS in 11m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6f9d98e701b74f4b920eee6fd7e48f11 : SUCCESS in 13m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fb0dbf1ceec4e07ba3fd2985706aba9 : SUCCESS in 12m 24s","accounts_in_message":[],"_revision_number":2},{"id":"6d09af730429a99e4f10ef6a6685cf3ca58895a7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 10:48:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bfa204e12e49c9adc1dbcaa7753dcabc115db88e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 10:49:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73d587d923c74309bba59f79624acb02\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df147bd2974c4bba98e54d06831ac54c : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b263786919b4075a4ff99005be7f8e5 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"46d28dc7996efafca5be3c9f8280fbd291806f48","revisions":{"41c2cc1c5979ec55e982eed662c3df011546ea66":{"kind":"REWORK","_number":1,"created":"2025-06-05 18:44:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/951898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/951898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/951898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/951898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/951898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/951898/1"}}},"commit":{"parents":[{"commit":"9554ddd2d339e27b68f360425beb11830b370357","subject":"Make custom Gather facts run with default python for kayobe-target-venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9554ddd2d339e27b68f360425beb11830b370357"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-05 18:34:56.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-05 18:34:56.000000000","tz":60},"subject":"Use latest CentOS Stream 9 GenericCloud image","message":"Use latest CentOS Stream 9 GenericCloud image\n\nChange-Id: If07b7b5aeff61c4a1181d4e4a6cc256456125c0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c2cc1c5979ec55e982eed662c3df011546ea66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c2cc1c5979ec55e982eed662c3df011546ea66"}]},"branch":"refs/heads/master"},"46d28dc7996efafca5be3c9f8280fbd291806f48":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-11 13:56:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/98/951898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/951898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/951898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/951898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/951898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/951898/2"}}},"commit":{"parents":[{"commit":"f6398a330812c0601b3476778492080d679afed4","subject":"Merge \"ansible-lint: Fix jinja[spacing]\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6398a330812c0601b3476778492080d679afed4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-11 13:56:24.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-11 13:56:24.000000000","tz":120},"subject":"Use latest CentOS Stream 9 GenericCloud image","message":"Use latest CentOS Stream 9 GenericCloud image\n\nChange-Id: If07b7b5aeff61c4a1181d4e4a6cc256456125c0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46d28dc7996efafca5be3c9f8280fbd291806f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46d28dc7996efafca5be3c9f8280fbd291806f48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
