)]}'
{"id":"openstack%2Fkayobe~951927","triplet_id":"openstack%2Fkayobe~master~I9e1f13221b36c0bdcc40c47762b51e26349d1dc1","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-24 14:35:57.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I9e1f13221b36c0bdcc40c47762b51e26349d1dc1","subject":"CI: Disable proxysql on centos/rocky","status":"ABANDONED","created":"2025-06-06 05:34:22.000000000","updated":"2025-08-19 19:16:02.000000000","total_comment_count":3,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"824cb8adef3cf23c24d82054b20fdbbcf07331b1","_number":951927,"virtual_id_number":951927,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-06 08:07:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"}]},"reviewer_updates":[{"updated":"2025-06-06 08:07:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-24 12:39:40.000000000","updated_by":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"reviewer":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"state":"CC"}],"messages":[{"id":"2f2a0d0721085f177fa5c1611342f40d9fcac9bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-06 05:34:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"263b159ad08f0bb57a6d1ed997c2961442478b37","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-06 06:30:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cac34dc81a25ee82ad1b864abe01cd0eb2889dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 08:07:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4cd4b4c6e884bf3bccd8af0f78423e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f6a95829044dabb41ac8a8e180c7fb : SUCCESS in 4m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e610557c0dd6483fbb5302e0f875042a : SUCCESS in 3m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8dc470bc17fe4b8fb36369edc861af3c : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e976948bb344ec39cf0195c56a8956d : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28e7c10e90d64a238b39c3d622b99338 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d7089394bd214a5ba83719e1a04a28c8 : SUCCESS in 2m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a437276fda9441059e125505370602da : SUCCESS in 4m 11s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2851ea45c1084723899d09c29596a07c : SUCCESS in 2m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b55510ab0b394de09a177b509e6902ea : SUCCESS in 6m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/239c604e4b4b432bb42be8831d39f6d8 : SUCCESS in 5m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/01747fdff195442587f973e278007898 : SUCCESS in 35m 57s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/94ee45a90c7f4c3e90ec99cb4e3f7787 : SUCCESS in 31m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed91b1c726814d50a61d11068dfdf5b0 : SUCCESS in 33m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cf95b31525ee4911858986c8e44e400d : SUCCESS in 30m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/789fd22b490e4831ae54693d37463019 : SUCCESS in 54m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/800133d11c7f4d689192a421af5113c0 : SUCCESS in 15m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9b4ca9a7ab1492b85e457b40783d74e : SUCCESS in 11m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/caea16439b93495db2c432bc0959b771 : FAILURE in 1h 11m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53f4f3f5e2ec4f85b6e291e32ab1854e : SUCCESS in 1h 36m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2f0d5acdf91f406e8a3f6c1d76b7b6a8 : SUCCESS in 13m 41s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/9d98f0734e974bffb8825e5027f07f1e : SUCCESS in 16m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e512386d42c744af8db377898cab20ce : SUCCESS in 11m 51s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d8c3604b1fe44fcc8f88557b9e51104e : SUCCESS in 14m 08s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/881fa6bbaa874d9392a814dadb11c4c9 : SUCCESS in 31m 56s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8c8af19d537a47a6acc32ca9bdca94d8 : SUCCESS in 30m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78f3e844fa9c4552a296833695f4c45e : SUCCESS in 21m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b78b5d5a35634b768ccc4af97f626517 : FAILURE in 10m 26s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/59191af6db7a4d3a85125289d28f5568 : FAILURE in 10m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97e7797c19254e98a233a7b3a00fdea2 : SUCCESS in 9m 49s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/ac441b84deae40038bf029440df8b019 : SUCCESS in 9m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0f834612e7c045889ae353aad3b90d1e : FAILURE in 8m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d928d9b4ed24b598a4b31322be4c75d : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"46072d87225b2dae2c39a367c6e23683e2a0b80d","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-06-24 12:39:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7adbaed304b04d9812d5b8bfc3d17b126634c360","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-06-24 13:02:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2536993ad5a17c5995659685c8e6141a1ef787a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-24 14:35:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"824cb8adef3cf23c24d82054b20fdbbcf07331b1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-19 19:16:02.000000000","message":"Abandoned\n\nWe will focus on Rocky Linux 10 instead.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b960d09963b9d5a173a929ac1874126ba16d4776","revisions":{"1e9452ef7d27ecb5b4020d1f7a4c84938dce87eb":{"kind":"REWORK","_number":1,"created":"2025-06-06 05:34:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/27/951927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/951927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/951927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/951927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/951927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/951927/1"}}},"commit":{"parents":[{"commit":"9554ddd2d339e27b68f360425beb11830b370357","subject":"Make custom Gather facts run with default python for kayobe-target-venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9554ddd2d339e27b68f360425beb11830b370357"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-06 05:34:07.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-06 05:34:07.000000000","tz":60},"subject":"CI: Disable proxysql on centos/rocky","message":"CI: Disable proxysql on centos/rocky\n\nDisable ProxySQL which now fails on centos/rocky because Kolla images\nare missing a recent change [1].\n\n[1] https://review.opendev.org/c/openstack/kolla/+/951741\n\nChange-Id: I9e1f13221b36c0bdcc40c47762b51e26349d1dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e9452ef7d27ecb5b4020d1f7a4c84938dce87eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e9452ef7d27ecb5b4020d1f7a4c84938dce87eb"}]},"branch":"refs/heads/master"},"b960d09963b9d5a173a929ac1874126ba16d4776":{"kind":"REWORK","_number":2,"created":"2025-06-06 06:30:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/27/951927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/951927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/951927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/951927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/951927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/951927/2"}}},"commit":{"parents":[{"commit":"9554ddd2d339e27b68f360425beb11830b370357","subject":"Make custom Gather facts run with default python for kayobe-target-venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9554ddd2d339e27b68f360425beb11830b370357"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-06 06:30:25.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-06 06:30:25.000000000","tz":60},"subject":"CI: Disable proxysql on centos/rocky","message":"CI: Disable proxysql on centos/rocky\n\nDisable ProxySQL which now fails on centos/rocky because Kolla images\nare missing a recent change [1].\n\n[1] https://review.opendev.org/c/openstack/kolla/+/951741\n\nChange-Id: I9e1f13221b36c0bdcc40c47762b51e26349d1dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b960d09963b9d5a173a929ac1874126ba16d4776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b960d09963b9d5a173a929ac1874126ba16d4776"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
