)]}'
{"id":"openstack%2Fkayobe~951973","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Ia3cd43b3cde2474eed7fbd1fcb63ad7405b9e2f6","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-09 18:20:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3cd43b3cde2474eed7fbd1fcb63ad7405b9e2f6","subject":"Bump stackhpc.libvirt-vm to v1.16.3","status":"MERGED","created":"2025-06-06 17:16:30.000000000","updated":"2025-06-09 18:21:31.000000000","submitted":"2025-06-09 18:20:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951973","cherry_pick_of_change":951915,"cherry_pick_of_patch_set":4,"meta_rev_id":"e3d3e40a3cd94e37d6ef79019273134fd23fae64","_number":951973,"virtual_id_number":951973,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-09 18:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-06-09 15:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-09 09:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-06-09 15:55:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-06 19:45:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-09 09:55:51.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-06-09 09:56:44.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-06-09 09:56:44.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-06-09 09:56:44.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-06-09 09:56:44.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-06-09 09:56:44.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-06-09 09:56:44.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"4743d09f46d06682720d13f45df61cdf6729e920","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-06 17:16:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"59a0a2f1634d55ef0bd2d60856ad433a54d75d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 19:45:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/682bb86797064da9b6d2ab7ab922604b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f17a6f4d4671445eb4ff04a12046dc1e : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fab47b6826a4b12ad3e6987d806de9b : SUCCESS in 5m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b577f987e59049078d61a23a28f34e0a : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d12786e7bc6f4f6a84b0af61890a7c45 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c812ef925dfd49fd8b9196e86f3f8502 : SUCCESS in 2m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fdaf94aef10f49668b34596085f6862d : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e09243c1b29f42c188f3e9a2abb5307c : SUCCESS in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4daac3256374aa8989a39a85990e1f5 : SUCCESS in 8m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/aac166c075ec4050b8f960783acacfeb : SUCCESS in 1h 11m 11s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/8dadf9d403ad441f906068c6d8d72f82 : SUCCESS in 33m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e40276c677e4bde91df6a5ac9bdc32c : SUCCESS in 53m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fdf63e9065c64be1b8806e88e92cd897 : SUCCESS in 59m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b7cd13575bce4014b3acc31cd4304950 : SUCCESS in 32m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a01ed69e654149e28cf1d4b0a6ca4c97 : SUCCESS in 15m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/227abd70a3494c158331a306f2493da3 : SUCCESS in 18m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5f1450e618d14a59b69fbf85f4e6f5c8 : FAILURE in 42m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3efb09e9c351401288070b7a78ba0672 : SUCCESS in 1h 57m 25s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bc1159ef0fd844c98030b4363da2332a : SUCCESS in 2h 23m 54s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d7802e5b10b4c0ea5b56a82ecf1bfa3 : SUCCESS in 2h 08m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/58fd2b745016494d9e0f50c4a0ea488c : SUCCESS in 11m 49s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8c1c77dd64664e67aca7650efd387a05 : SUCCESS in 8m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b482851f9b74dc582a5d32877eee05c : SUCCESS in 15m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b4a2990223d6432e85c6078618f2daae : SUCCESS in 14m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/50c58ca5b1da482ca7f84f21f991d53b : SUCCESS in 31m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6980af1318dc4a6cb7b7a331eab6889e : SUCCESS in 38m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e209bfea08a249a3a59945a46db25dec : SUCCESS in 26m 39s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c0f5b2abe48a4747a3701779ecb2cd07 : SUCCESS in 29m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d8a86f554312404892e2c4a360426dfa : SUCCESS in 14m 06s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/a5da06a2cfe446c78e177e2dcb8e9aad : SUCCESS in 11m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe38003b547349d2a5cc850fef994d48 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/fbfe94594e2a4bd3a6a5718e79359c9c : SUCCESS in 13m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d9460897332242218ad3e47ccecabe38 : SUCCESS in 11m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4d75f75facf4b2ba3ea7b2c68f29d65 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":1},{"id":"bbcdead84846b843ee7a8f7bd1ffca6d142b039b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-06 20:11:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b33164916bc0d3b8d1dc7ae20eab42786818ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 22:35:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db2384297e5444e2995554314db79f6b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3a1b0c79f6649bd8acba394ef4ef3a2 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fae3df063d3d45cd947ed6bcbbac8a86 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5de64cc8c53247edbcb932ff782a9c72 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f7b96b074264ff489d2fe1ae3892436 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3c3bb0f56c1469a89cc50eb6a5a5e26 : SUCCESS in 2m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/002ab32bcd8045c78fdca1fbea0c65e1 : SUCCESS in 2m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/76bcac4aeef94d13bbfa4a2bf37211ae : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0804f0608fc54e74a34cc00214a233b7 : SUCCESS in 8m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3392e14d26b140c9be7a19baf6a80316 : SUCCESS in 31m 16s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1261b3d40d2e458cb72df28e64e3006a : SUCCESS in 56m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/812cf5eeb101420d95e4f85fda00c2ea : SUCCESS in 29m 24s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ac56762eb9e147d68bcf4fc30232708a : SUCCESS in 57m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ffbe1053e56c46f59e6c2cf4e1ebc39f : SUCCESS in 33m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6291a521e0234b78b91b60e43f968bbb : SUCCESS in 14m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d8fe9b5f8ea4c7a81fc0fe72436a101 : SUCCESS in 11m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/19e14a8d79cd4c41bc02259b86a3b083 : SUCCESS in 1h 20m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/145d058ff73942779c1971c1151d2777 : SUCCESS in 1h 20m 36s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1955995bb85148a8a0143a1b78881e73 : SUCCESS in 2h 06m 39s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0132dc5e943e4793942935f5fe657cf6 : SUCCESS in 2h 23m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9f8a39c14e0a4f5ca7c78a1d8511aea6 : SUCCESS in 10m 12s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/802dbce0b62745288739092a0776b40b : SUCCESS in 8m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e08ef312ae944d2eae1d535371e67da6 : SUCCESS in 12m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2ba027cf2d9643babe6e68e1fa19f837 : SUCCESS in 13m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f0f3e0ca71b444a69e9bc5f8cb58bc14 : SUCCESS in 21m 13s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/571e02b450584044b3b84c3992fe6e31 : SUCCESS in 20m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/581b5de250ea4077a9bcaf879e43c85a : SUCCESS in 28m 01s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/426bb0d616904af0844018d5b6ff8548 : SUCCESS in 22m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/98a06102bd55416e9410ff7fa70d5382 : SUCCESS in 14m 27s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/bf1d7e7fcd19419eba4a4b512dd37df7 : SUCCESS in 6m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ffe05bf8086412595dae71eaf2182bd : SUCCESS in 7m 52s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/466f9a3701ce40738d6beb21b8192afb : SUCCESS in 11m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bf4665127d79493e83698efe0adbdfc4 : SUCCESS in 6m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e61f4685afe7483ea518cbec725f919c : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9a450626eef0eefb34e9126fb1ffcbda53254a09","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-09 09:55:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"daa32a2c73471b491920a742a73a891e1b509e9a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-06-09 15:55:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a163d61cb30a75f95098ca2017c3ceb41132e30d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 15:55:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29f56b7464d0eb9c7631c60f96b7be3f00f5af1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 18:20:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fe4a66d9ea44eba904732a662fd9ab0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/837dd21b6df84d9e8c2a9c58bf9283cc : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0ad07097f244e6eb32b27a71901b9fc : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b820a1ebb39a491fbba77ef77865b727 : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c4bcf2ec3b64c659b8601c8f8f6fdc3 : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29731a2da64e44dca3dcafb1d7ae6c0c : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4208d96984814fa2bb067e62c692fb07 : SUCCESS in 13m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac0d7d9643ca4a70957da98ca1ab5b79 : SUCCESS in 8m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1d0b0f1472b34c869c9ee4a0835aebea : SUCCESS in 58m 44s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/b45d2b8234d84286a3024858d884ed49 : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95e3c0e3258c4b59802513c5d0ce6c83 : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1f2b9505cfdd45c085569ddad63d50f7 : SUCCESS in 33m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/272b643594264f2f922ea993bca0dc63 : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7144aca6269e4e79acbfbddc3e2dd857 : SUCCESS in 17m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e0cdac160d448d5a1d293caf5bfceda : SUCCESS in 11m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/31661dd7587b4d409225af073e0aca07 : SUCCESS in 2h 17m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/649eb20830e148b7bd33042c63e74ea3 : SUCCESS in 1h 20m 57s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f48eaf5b592a49d0a566501639743773 : SUCCESS in 2h 15m 14s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a04d54c0da844f6ea218572fc21fda6e : SUCCESS in 2h 00m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/81bf7604aa694cd2a1846276ffd070b9 : SUCCESS in 10m 00s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/b6cef731c95b40708abea08dda694328 : SUCCESS in 18m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba3bd5b0cf474fd0a1856d504d3c8d57 : SUCCESS in 15m 51s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/166e8fb1f04341a18e9a8828afd6f770 : SUCCESS in 13m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fec91a6e9ccc4f268d948b1b336eb546 : SUCCESS in 35m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2087624a0e624286877347fbddd2ff73 : SUCCESS in 27m 42s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5f3b2339cf6a4f558e7fb87bb7317a3d : SUCCESS in 34m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/08e55155e47146ddb5a53d3251977342 : SUCCESS in 13m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4193b6ed5d39455389d9899826959950 : SUCCESS in 11m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/027ea040c61445ff9f9329705ac270da : SUCCESS in 13m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6159d42722124264bd8484fa7132780d : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f48cf595a9c83d20c8fa0847209827796b261049","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 18:20:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e3d3e40a3cd94e37d6ef79019273134fd23fae64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 18:21:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4905649e090c4a1abdd5235e66252022\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2df661393ba94ea199e45b48e4b84e6e : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51597336f036489aa797c8ef270b62a5 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"107e6a81c62dd985540ba8f22d329c2d1bcb6c01","revisions":{"107e6a81c62dd985540ba8f22d329c2d1bcb6c01":{"kind":"REWORK","_number":1,"created":"2025-06-06 17:16:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/951973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/951973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/951973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/951973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/951973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/951973/1"}}},"commit":{"parents":[{"commit":"67cc6af6e4f8222098bcf96c669d1d46b5e5fdf4","subject":"Fix syntax error in drac-pxe role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67cc6af6e4f8222098bcf96c669d1d46b5e5fdf4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-06 08:50:12.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-06 17:16:30.000000000","tz":0},"subject":"Bump stackhpc.libvirt-vm to v1.16.3","message":"Bump stackhpc.libvirt-vm to v1.16.3\n\nThis fixes seed/infra VM provisioning failures with libvirt 10.10.0\n(packaged in Rocky Linux 9.6).\n\nCloses-Bug: #2112628\nChange-Id: Ia3cd43b3cde2474eed7fbd1fcb63ad7405b9e2f6\n(cherry picked from commit 20195400150ae15b165316e8b0f66254f766ee1d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/107e6a81c62dd985540ba8f22d329c2d1bcb6c01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/107e6a81c62dd985540ba8f22d329c2d1bcb6c01"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
