)]}'
{"id":"openstack%2Fkayobe~952161","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I4548410d533731138de02780dc8b2013e71730d0","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-10 11:02:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4548410d533731138de02780dc8b2013e71730d0","subject":"Fix image registry conditional","status":"MERGED","created":"2025-06-09 19:47:02.000000000","updated":"2025-06-10 11:03:29.000000000","submitted":"2025-06-10 11:02:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952161","cherry_pick_of_change":951961,"cherry_pick_of_patch_set":1,"meta_rev_id":"19c1cdf5cdccb70ef3fcb955888a76945c3e3690","_number":952161,"virtual_id_number":952161,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-10 11:02:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-10 05:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-06-10 08:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-06-10 08:31:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-09 19:47:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2025-06-09 22:01:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-10 08:31:47.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"4c6d5abb67604d37a064dce2325c57c8ab20f2c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-09 19:47:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"35d095c75c0b7026819a126b182247ad89640934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 22:01:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/246abbf5967f4d5581f00aa5ee15eae0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/451d104fb17743bfa0af3849729ddeba : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1a51c6c6dd47709cfc7260f15854b7 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8db2f2b8266b41d2af0f596b76d54981 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dc26f0f35cc41c7a598afc47e5985a2 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/265467376ece4e0da567e58780e54fc2 : SUCCESS in 1m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/739780a7fceb445e8e29c92d46ecdc73 : SUCCESS in 2m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6147ed043c294a658cf652177676266c : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b4f4496f9aed4ef7a151e5f82cd481fe : SUCCESS in 5m 14s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bc3859e5dd09491d9fcd730f9021bd69 : SUCCESS in 48m 01s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/0b6dd6c4502b45be9764190606e076bf : SUCCESS in 32m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1781f8d11b414cb59afabbebc2e0eb7a : SUCCESS in 57m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4664d155169e4084bfd2faecef66d643 : SUCCESS in 31m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bf2c1d2b2b9c4d53819acdc867556715 : SUCCESS in 38m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9ac1c6c14b9a486384d3ded38480294d : SUCCESS in 17m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4931006fbf0d4aea9a6892800ee0b0fd : SUCCESS in 6m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5e9ae0892cdf4f59ac7a60fa339ef382 : SUCCESS in 1h 29m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1eb3c0178a3494d99be9f154431b810 : SUCCESS in 2h 05m 49s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8e192eac49b74ba5a6eeb42d7a47a777 : SUCCESS in 1h 16m 07s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1343fed36fe4a069d91e629a1b3c36c : SUCCESS in 2h 09m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3e761076b6d64493a5d8c8f74872b86f : SUCCESS in 13m 20s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ff6c0283da4e4ac1858190701f735971 : SUCCESS in 13m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c874e943c88c4fde96433d5551b1ec0b : SUCCESS in 11m 50s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b6bf74b42c944fa481d44981f8b08653 : SUCCESS in 11m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2fbcb35f42654252825c63903e8aa2ab : SUCCESS in 31m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b1c1a5ac062e4464b9146f657f1d861f : SUCCESS in 28m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e17487342613457f9e26540641f0bbb6 : SUCCESS in 21m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/29c0c9cab8994819ba2fce1f2e1a445e : SUCCESS in 24m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b1f433313d354cd8bacf07c3d935c3f5 : SUCCESS in 11m 06s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/2d41b671973b4c95827e581a54a2ca24 : SUCCESS in 12m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c16bdfba93154509b174cfc0ecf60068 : SUCCESS in 9m 46s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/a79d6321a5f64ce8a29b6d7430ef9df4 : SUCCESS in 9m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ce7b65f690a4256b77af277f52c3409 : SUCCESS in 10m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78961164852749bbb6755c0970311d97 : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5132567139623e4837a3291fadd033f3a09a5fc0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-10 05:24:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"becd269bb8aa227a921b14584f0f7b55a06616aa","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-10 08:31:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae50427409d35b9d552e11ac51c9a04d650aab4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 08:32:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e87169c753079b542bf1dcbdf7d76a3735538919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 11:02:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81f65e008e6f4c6490c7fee2994dab1c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84e27f5da6e542bca627ada1750fe3e0 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af670fc8382c4c1395f592b9fcfec150 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3473ab63e5ff4e83a9365242ce6095a4 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ddc7ace8274480685c40d1cfa66dcd7 : SUCCESS in 2m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a7221c9cf81945978a83385f7a932a7b : SUCCESS in 3m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35f69c502d28419e9b788c7e10d10a65 : SUCCESS in 7m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1870a0f68d724318b53b6d9fe5e74d24 : SUCCESS in 9m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a605caf08706439495f75657465c0a07 : SUCCESS in 30m 01s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/4ff836c2e0c24012a0d487f5e8e5849b : SUCCESS in 32m 44s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b375505ed8744896aa9157877a8d65fe : SUCCESS in 28m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/80835659bea5477494769d4740967896 : SUCCESS in 31m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cb1fc6e2ebce462081e61598b410e122 : SUCCESS in 1h 05m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5bc3c833485d479785ad516d49070103 : SUCCESS in 9m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57c3707055104f55b104ad7e32edd572 : SUCCESS in 7m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bb60b5dd63624d76b88aa887bec7784d : SUCCESS in 1h 15m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9f884b4ea284cc887f7233100eeeabd : SUCCESS in 1h 12m 36s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ae77bab6cb8f4fe896aecdbfe1823fd4 : SUCCESS in 2h 25m 37s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9dcc68ff4db4072ac53319790bb2329 : SUCCESS in 1h 09m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/122ad1c7fcc949a8bdfb0e579082e74c : SUCCESS in 14m 26s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6034c951a20f42e7a5ac092317189b6d : SUCCESS in 16m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8491483439224f23ad9dc44391960429 : SUCCESS in 15m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fe83635e7eb641bdbaf1fc25f4b26349 : SUCCESS in 13m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8c956332503d487eaf9d456fc2a04250 : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0205012688fa4d03bdd0c01705a03bc8 : SUCCESS in 29m 47s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/27db0d86f770495abd0f94425d5e3d69 : SUCCESS in 31m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a5df0fd5900f4a0e9df45c6b1db37980 : SUCCESS in 12m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc2e6ef9bf7f4f75b882b20df9a5918f : SUCCESS in 10m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/558adef618b34cf9946b6d894e0fcc56 : SUCCESS in 8m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/298c8c3cc1d841f9b90e6c165227d630 : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f399f20a211170ee0c252edbaaa99144435abf8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 11:02:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"19c1cdf5cdccb70ef3fcb955888a76945c3e3690","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 11:03:29.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb79e18e416049a1bd033704985385bd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11531d5734b94dd983a03bf30d9f824e : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff3c6e86af8e41ae9646216b99b0cdb7 : FAILURE in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7d81f49be10ebf4cd96afd8e36aa5ffe2c32cb5","revisions":{"b7d81f49be10ebf4cd96afd8e36aa5ffe2c32cb5":{"kind":"REWORK","_number":1,"created":"2025-06-09 19:47:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/61/952161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/952161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/952161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/952161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/952161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/952161/1"}}},"commit":{"parents":[{"commit":"107e6a81c62dd985540ba8f22d329c2d1bcb6c01","subject":"Bump stackhpc.libvirt-vm to v1.16.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/107e6a81c62dd985540ba8f22d329c2d1bcb6c01"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2025-06-06 13:59:28.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-09 19:47:02.000000000","tz":0},"subject":"Fix image registry conditional","message":"Fix image registry conditional\n\n`kolla_docker_registry` evaluates to an empty string if neither\na docker or podman registry are configured. This leads to broken\nbehaviour if an image build is attempted with a regex specified.\n\nCloses-bug: #2112646\n\nChange-Id: I4548410d533731138de02780dc8b2013e71730d0\n(cherry picked from commit 8f0c9a7ec718cd1cf5a2ffd230e0c99c36822af9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7d81f49be10ebf4cd96afd8e36aa5ffe2c32cb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7d81f49be10ebf4cd96afd8e36aa5ffe2c32cb5"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
