)]}'
{"id":"openstack%2Fkayobe~952163","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~I4548410d533731138de02780dc8b2013e71730d0","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-10 11:02:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4548410d533731138de02780dc8b2013e71730d0","subject":"Fix image registry conditional","status":"MERGED","created":"2025-06-09 19:51:55.000000000","updated":"2025-06-10 11:03:32.000000000","submitted":"2025-06-10 11:02:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952163","cherry_pick_of_change":952162,"cherry_pick_of_patch_set":1,"meta_rev_id":"88f09ee77c404b5b9462a1050515f7ffca3f2df6","_number":952163,"virtual_id_number":952163,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-10 11:02:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-10 08:32:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-10 08:32:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-09 19:51:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2025-06-09 21:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-10 08:32:06.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"90c58a74d695773e85f35ef0304217366475b4b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-09 19:51:55.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"3f956fa2f5c67e6ef35b16ec4ed56605acbb92a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 21:33:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de60b63df41f4b748e517b9921b02b80\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a60129e7c8c484fa50187f16e810132 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d31ba75c6ff4f5b8f46580138595cfb : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f9692b46ee946cc8c79f55151b9a42f : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/25100f59bda54fcf8b1037b7e8fe29c7 : SUCCESS in 3m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a3a455f30c24064989b705d19115dc3 : SUCCESS in 6m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/34c60b1ec2574899b668a0c714782520 : SUCCESS in 5m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/44f803a7f3804561a7266c73581997e6 : SUCCESS in 58m 42s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/820ce271281a4dd88794bc74c3733e11 : SUCCESS in 55m 06s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15384758e8b6463ebef3d67ce587d78b : SUCCESS in 49m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0c96b17e3c647d29f41cb1b657a6807 : SUCCESS in 34m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f5038f53f2854243a6b98116a580ee18 : SUCCESS in 51m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c753e1795d6345d59c1dc117a1fc5a12 : SUCCESS in 11m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/beb4456784244363b3d5d8f59eaf35c2 : SUCCESS in 18m 51s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97a86f420d2049ab99ebd8780a5ccc26 : SUCCESS in 11m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd17688af0464214891406c488bd90f2 : SUCCESS in 16m 36s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/f63f1f29e8074fef9e74029ff7430d1e : SUCCESS in 1h 40m 16s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e10e735156cc41a48983b050310f4c19 : SUCCESS in 1h 31m 23s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/58452d73e6e74fb3a8b8df2ed6ece02b : SUCCESS in 13m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe7e316922c04ae1ba38f5000569daa7 : SUCCESS in 14m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b201538736814426bc713f0db6cab2e3 : SUCCESS in 15m 48s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c5506e2ac9ef4657b157eee108b02a79 : SUCCESS in 32m 59s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/da9449e7f12a4cc49c934ea842d128ff : SUCCESS in 28m 25s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99ee90e5ac4a48bf9c27b87ac045bb7b : SUCCESS in 13m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0d96a1966dfc4c2998b209c361709d68 : SUCCESS in 16m 51s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/86ad5a91ca5949c1972cb3c88fa41ba6 : SUCCESS in 12m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/537b5789e4f44b878342769777907abd : SUCCESS in 12m 25s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/5848a01047124c2bb3b3f0ca5e3441bc : SUCCESS in 15m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d726a163a3d4409804484c65d8e6baf : SUCCESS in 15m 27s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/19d115ad84c84365bd4be3fdd9bc9ab8 : SUCCESS in 14m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b1255fb0e7744cfc87aac62a541aac9f : SUCCESS in 15m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f8772672235454c9a7c405709836423 : SUCCESS in 14m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9635edb8348468bb5f897ca188c2088 : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55f715db9ae54177a6953a2ebb956e83 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d98c45e2b50a49509e03a47cf8454206 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e56a7191830473489447ef3e43402a1 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b461b5c5b96e41deb22a86810d0bfd16 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9939cef3204eac9e70ef88de33b7b7fdfe43cd9f","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-10 08:32:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6d1bd2e07d8bc92bcd990bbe0a30a2161af49336","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 08:32:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c97faadf95010c0f389fb80184e244e773703715","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 11:02:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1fdf266390b4e18a1f547404c29f39c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66047500731747aba6ba0de808ea3cd9 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04c3f1ecf50144f7989d32a5f6e50374 : SUCCESS in 3m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0288219646204b7fa27ae52f28dc3e80 : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/01b7b871b99f4878acb4f56424547651 : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ba0f2c04ed048f59ea9058831b4f97a : SUCCESS in 10m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e6fb0eb0bee2406f9d06f1dcb2b87f74 : SUCCESS in 59m 22s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/734bef29b8ff4d579893154e5d621992 : SUCCESS in 54m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df9c9583df194e53b1079abef3251381 : SUCCESS in 52m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6291b66160f84f2a8b06c7e8a831279a : SUCCESS in 1h 06m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9a040fd04a914e6c81278d6a7bcfab30 : SUCCESS in 16m 54s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9d3397c18174af499a90bb703fa3f34 : SUCCESS in 12m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04d5a551b7b2417f9a9ee1b211389bf3 : SUCCESS in 16m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/67cde42590f24be483991ddc4d9e42bc : SUCCESS in 19m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/032e6af9b483493d8f2591851696d83c : SUCCESS in 20m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8468cc15b554fab86473a59535707e4 : SUCCESS in 21m 01s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3bde1c51318241e5889d52c3386c2c30 : SUCCESS in 27m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5462a3fe6a3a418cb595230cece62700 : SUCCESS in 24m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7b5b695860f24f2d9776725168905a8f : SUCCESS in 12m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a032f9646a1b4bb2b0952f2637ed8007 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1b5c9a68f644f348fb6b16bbf003229 : SUCCESS in 10m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7912a99144c04ddb977d48fad23439ce : SUCCESS in 15m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a1e74acd0274ac2a6c16b1ca27a176c : SUCCESS in 8m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/945a316c6d4b40eb8a3b4c8168ef30ab : SUCCESS in 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"00966ec2c10bae7178fb4f4e06c9a34e7f79681a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 11:02:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"88f09ee77c404b5b9462a1050515f7ffca3f2df6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 11:03:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a895d085779436083e64519a217425d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0994939edab34a459460d1e837d97db6 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed9e4e8e22fa4100a319b2f19c7a2862 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ddca6265f64fa8e9cab331cb571b8dff0a476804","revisions":{"ddca6265f64fa8e9cab331cb571b8dff0a476804":{"kind":"REWORK","_number":1,"created":"2025-06-09 19:51:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/952163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/952163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/952163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/952163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/952163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/952163/1"}}},"commit":{"parents":[{"commit":"afc8fa9d7a0e4eb4629c290a5f33f24ef67f9eca","subject":"Bump stackhpc.libvirt-vm to v1.16.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afc8fa9d7a0e4eb4629c290a5f33f24ef67f9eca"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2025-06-06 13:59:28.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-09 19:51:55.000000000","tz":0},"subject":"Fix image registry conditional","message":"Fix image registry conditional\n\n`kolla_docker_registry` evaluates to an empty string if neither\na docker or podman registry are configured. This leads to broken\nbehaviour if an image build is attempted with a regex specified.\n\nCloses-bug: #2112646\n\nChange-Id: I4548410d533731138de02780dc8b2013e71730d0\n(cherry picked from commit 8f0c9a7ec718cd1cf5a2ffd230e0c99c36822af9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddca6265f64fa8e9cab331cb571b8dff0a476804"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddca6265f64fa8e9cab331cb571b8dff0a476804"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
