)]}'
{"id":"openstack%2Fkayobe~952441","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~If07b7b5aeff61c4a1181d4e4a6cc256456125c0a","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-17 15:18:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If07b7b5aeff61c4a1181d4e4a6cc256456125c0a","subject":"Use latest CentOS Stream 9 GenericCloud image","status":"MERGED","created":"2025-06-12 11:04:19.000000000","updated":"2025-06-17 15:19:13.000000000","submitted":"2025-06-17 15:18:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952441","cherry_pick_of_change":951898,"cherry_pick_of_patch_set":2,"meta_rev_id":"71e051ae09a4603c040126b3f6b56eefb5e32d1e","_number":952441,"virtual_id_number":952441,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-17 15:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-06-12 11:50:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-06-17 12:26:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-12 12:10:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-06-17 12:26:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-12 12:10:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-12 11:50:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-12 12:10:49.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-06-12 13:27:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"ccd165649dbbbcc3823c4cba8e0e87acd2d52fb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-12 11:04:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b84f522e7bab6058fc89bf5aaae02a41448754c4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-12 11:50:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cfbee3db007222eb493f7a2ff5fb1a8c4c833f11","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-12 12:10:49.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"2a389efc9285ca374fcc4339f53a5be4fffd69d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 13:27:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a422d9e301a46d891d9cd689a624225\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8898ed7080dc4d78b45b60f64482f814 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c69bf28e384d0cbec9ef2013241aad : SUCCESS in 5m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8522534d3bcf4d179719ec90ad1ce43e : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e8cef82f1354976b57571cd1372f391 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75ec8a3f1c81424bb6ad4ac755987414 : SUCCESS in 2m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bd98f662f9c64ba6955c1ee74ee9c1f1 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2a445f4044074ee28c6d24fd9c7f91e6 : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/351a4f8429454c1e80cdf42efa38d348 : SUCCESS in 12m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/17dc4faaaf9544a68a383d3ad893c704 : SUCCESS in 52m 00s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1873545472614afb84c2946db73c3605 : SUCCESS in 48m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ca001cd69904c47b749f5acce5a50aa : SUCCESS in 50m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ff47374e1192488f81804886b1d5b4a3 : SUCCESS in 47m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/acc34f675dc2458c98e7d451cb2409d1 : SUCCESS in 39m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1213e93973094615bf97dc1a63fa81db : SUCCESS in 18m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/448c8e908c04450583eed0361f8fab60 : SUCCESS in 17m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dd06b8375412442abcd98f1670d60e34 : SUCCESS in 2h 18m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47f8a211e8404bce84981794942b3a7e : SUCCESS in 2h 01m 19s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7175afab071b49578629d047309987dd : SUCCESS in 1h 28m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e395eaf5f8148d4a6a5b9a537c71010 : SUCCESS in 1h 45m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d8971dbcf4044f90b2ec08ff77ae2a9c : SUCCESS in 13m 31s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/f539c9488c194c2589624413af513545 : SUCCESS in 17m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10f33cf1489b491bad15e9c89edf9348 : SUCCESS in 9m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/efe0986e6293448db83cd495b58e166f : SUCCESS in 13m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6f961a3c46fe4370ad62909f67e9ef52 : SUCCESS in 22m 21s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0d23c7f53e3a48a5bf9f37732726689e : SUCCESS in 33m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78330844dbdc484eaa907635f66bc102 : SUCCESS in 30m 44s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d3fabbb094474ea889be6e1df3a12a05 : SUCCESS in 32m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4fb44d8c14914d589ffd82922d481291 : SUCCESS in 13m 16s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/8f59a038fec74330b96011f805e988e3 : SUCCESS in 14m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6cb9da660d24fafb073df70bf7858b3 : SUCCESS in 10m 47s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/d66558f9663f4314834178822c15274b : SUCCESS in 10m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8721ec3d50844c5bb0b2a469f0a105c1 : SUCCESS in 12m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/febde37ec68e46b88d54411182bed10a : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ebeefae3c64ced18da2080edc3781af00d546a65","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-12 13:59:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76c783926db0fca19a520c81acff00c69359f546","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 15:18:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6904f0585d045c6895d6ded0a6dbd6a\n\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7e94d5912a174d97a97258178b87c0f3 : SUCCESS in 1h 02m 54s (non-voting)\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1805b50ac50543cba607eb7b4d125328 : SUCCESS in 18m 12s (non-voting)\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/6488c26abb634bfebddbc99e48f01f7c : SUCCESS in 56m 06s (non-voting)\n- kayobe-seed-images-rocky9-podman https://zuul.opendev.org/t/openstack/build/296592097c954f6793114b05fe7d7e7b : SUCCESS in 1h 13m 24s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f7fc42641e84175b10258c76bdf9f95 : SUCCESS in 41m 01s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b6c5b2dcddb94716a526f2d47bc781d0 : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":1},{"id":"79dc55abf8463538af816a262a01f1e480261baf","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-17 12:26:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ab4e25e2709259042c33fcaa30e2c36a2ea4d137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 12:26:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f60fe1c45f6738349278e4231b7e2f75eddaf3fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 15:18:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c5d5f542b9044adbfb11211cf9d526b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbd07fd6300d4ff3a1079a4b13d28fa7 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fad6347597144439941c8f9d75bf8316 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e67dcd5641424b94ab9d5564f88f9c4a : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1c60b087f134054bedc42898bc0f8b3 : SUCCESS in 2m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8fc304e842aa4d999db0ecd21510a760 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1a463380afe6447abd973b2f8bdaf8de : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e931eb039b4a46c3a9ed9bf4bd3a093d : SUCCESS in 12m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ac7af9f25dc549328ee5ff04d9045e82 : SUCCESS in 55m 10s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/1fd9ad78c036464eb1ca2b53ff59a385 : SUCCESS in 1h 11m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/221767d6103d4ebe9c917e9e1622571f : SUCCESS in 58m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2f3d51f07d504ab0a28efe4055fdc3bf : SUCCESS in 1h 08m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5440804a6013421690ff37d2e2768326 : SUCCESS in 41m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d14e82cd576844b1babe0b2e380d1380 : SUCCESS in 21m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e9801e7ec9d483caf021561b6e5d1f5 : SUCCESS in 16m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2c3801dc7fe949e0b378f384cb3cb4e3 : SUCCESS in 2h 36m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab0fa833482947749fead96ca9b8d68c : SUCCESS in 2h 37m 13s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c3facf80bd934c669a22bed6e78be602 : SUCCESS in 2h 45m 58s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6ae6d17e75149b79e32ff8104076601 : SUCCESS in 2h 26m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5877c34e7e444e0f906761c2031e34f9 : SUCCESS in 19m 17s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/52a257c13c474d1eb5cd344aab415575 : SUCCESS in 20m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7af37415d8434dbe937799d85228f7b5 : SUCCESS in 18m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1c4cff8b5f2c4205a8cb9c6d8f8a10b7 : SUCCESS in 17m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/87f182c8e7764032b74c3fa6e7e74f42 : SUCCESS in 30m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35cbd5e2d3f24c07bdd095631640ad9b : SUCCESS in 28m 03s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ef1d1212455142efa4fc9f5b1eb3533e : SUCCESS in 40m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bb65b26275f2442ba724aac1a696309b : SUCCESS in 13m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a08a025b985e48a48b0e60c74506a1dd : SUCCESS in 11m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a57fb71fbce646b7a9e83b3fb8fa90e1 : SUCCESS in 11m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/399b6e099ee945d383d1ad410fbc72bb : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6d2d3544b17a0063f490e94c9a63a868110c850b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 15:18:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"71e051ae09a4603c040126b3f6b56eefb5e32d1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 15:19:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbba8cfae1694b6b8145d9c7cfa648cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c202c81300c4d3995d2edd11d6680d0 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6755b15ecbd54796bd42f825fc468b49 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46ceb39ad84412ced69f0c66cdf48156d7a13a44","revisions":{"46ceb39ad84412ced69f0c66cdf48156d7a13a44":{"kind":"REWORK","_number":1,"created":"2025-06-12 11:04:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/952441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/952441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/952441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/952441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/952441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/952441/1"}}},"commit":{"parents":[{"commit":"92219a792ed94e0ee9ee7a9f8745f287f05f1e9e","subject":"Merge \"Make custom Gather facts run with default python for kayobe-target-venv\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92219a792ed94e0ee9ee7a9f8745f287f05f1e9e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-11 13:56:24.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-12 11:04:19.000000000","tz":0},"subject":"Use latest CentOS Stream 9 GenericCloud image","message":"Use latest CentOS Stream 9 GenericCloud image\n\nChange-Id: If07b7b5aeff61c4a1181d4e4a6cc256456125c0a\n(cherry picked from commit 46d28dc7996efafca5be3c9f8280fbd291806f48)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46ceb39ad84412ced69f0c66cdf48156d7a13a44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46ceb39ad84412ced69f0c66cdf48156d7a13a44"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
