)]}'
{"id":"openstack%2Fkayobe~952443","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~If07b7b5aeff61c4a1181d4e4a6cc256456125c0a","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-17 15:18:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If07b7b5aeff61c4a1181d4e4a6cc256456125c0a","subject":"Use latest CentOS Stream 9 GenericCloud image","status":"MERGED","created":"2025-06-12 11:10:13.000000000","updated":"2025-06-17 15:19:25.000000000","submitted":"2025-06-17 15:18:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952443","meta_rev_id":"d1090da9f9db7c10671a288411b734e0c01ef517","_number":952443,"virtual_id_number":952443,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-17 15:18:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-06-12 11:53:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-06-17 12:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-12 12:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-06-17 12:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-12 12:11:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-12 11:53:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-12 12:11:15.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-06-12 13:20:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-06-16 09:15:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"bb45a1f25ffe54cce2349626d228b101ac88647f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-12 11:10:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"feed13714c75415f3d5b43eabb1dc42baef336a5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-12 11:53:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fcd19fc9dd173afb1f1369bfd3e94c601f7a621d","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-12 12:11:15.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"307b650e7a7d794a5524e03c2b07a047c9cb3f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 13:20:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70fd14f2e5174ecfb815db4deb1fc320\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9aa0075ddd0242a3ae3b08d0d27dd452 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638bcfc1679243e5affb618b2fc59721 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f6e7f200cfd48b4b5853c6378ca35fd : SUCCESS in 2m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28da7b59dfd446709493699abdec4739 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/32f52403dc584d9d9d2086d615c6fbd8 : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e12d5fb71a144cb789a2a3a6325ed17b : SUCCESS in 9m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/52d91574c63c4b7ba04fc3dda1994633 : SUCCESS in 45m 56s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/568a80ca521c474384bba49015a6bde1 : SUCCESS in 1h 02m 02s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/321fb0be3bf94ecb8bd2e0e89ed9137c : SUCCESS in 54m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7de99c88ae6441a59d1a5b8f48046b16 : SUCCESS in 44m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d032e234c750420a930f08ef1e7d5c41 : SUCCESS in 1h 10m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/555248a5561d4bbfbea55353f236a19c : SUCCESS in 17m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/126c2227e64e42bf911b531629531b21 : SUCCESS in 17m 36s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b26c54019f13457f96e8255b0d87ac75 : SUCCESS in 12m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b51b120fbcae4e6aaf021e2acae52645 : SUCCESS in 16m 39s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5798a5692ba845eaac5af852a93ad7da : SUCCESS in 1h 49m 32s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f20077a1e2bf4973953dea19d43966e4 : SUCCESS in 2h 07m 52s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1b4c6497c4e04d05bd7259f4baa8c425 : SUCCESS in 21m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e06a5d50dfaa4b06a677cacbca2f13f6 : SUCCESS in 11m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dc458a1333d4c5ab1f4bdf37da18831 : SUCCESS in 21m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f135a082e90543568d3a532afb4c0cd8 : SUCCESS in 44m 11s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e45191c7d3b84151914b8c59dd0b2f1a : SUCCESS in 14m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aece66e84134c36965f552182fc92b0 : SUCCESS in 29m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8485b1e937734f4f8577d7fd6456b90e : SUCCESS in 16m 59s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/9ccc2e6fdf584718bd01149211efb41f : SUCCESS in 17m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf0d30a78042433eb26bb79da3feb474 : SUCCESS in 15m 36s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/1ecc7734cc034d89b76042a0304b40e4 : SUCCESS in 8m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed96e2cf47484845bad2fa80d6f47920 : SUCCESS in 13m 07s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/783d147bfe844606aed1f88864ae414d : SUCCESS in 13m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a8a68d791c574e0abd9fb0cc04d2e10c : SUCCESS in 14m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/268fdb0277344faba1471ee2f3771374 : SUCCESS in 14m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37ea4edcc1d7444d93de4e1859cad427 : SUCCESS in 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff19096bbaf45ea99a563c6cb4fd689 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a312c2a10390447a8a991d806e6fc176 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a8edcc3aa2e4c18acdc8982f6e0fd70 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f7c311c82024cbca3cedac2813038c9 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6d599586baf28e0f0626e0fba3dc017c9b03065e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-12 13:59:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77db9f9b444036b9839f65672cf7652b14f59588","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 14:49:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49f47fa7e190461fb2cbc458f8a3b098\n\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/e4ffb6debacc4c20ba6bf58f019c4bcc : SUCCESS in 48m 58s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8debff1b5de843148a3c1c1ae0595220 : SUCCESS in 49m 05s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bc50d7650c54aac876760f7c818c257 : SUCCESS in 46m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6ef36d04db2164699f03afab0c172b1762f9c510","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-06-17 12:26:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c2be700e0d31bd3f966bd2d40cdbb2880caeef7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 12:26:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f1d568c64bea38e31453ef8dd96abdd7646a7b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 15:18:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10693702158a48999d853563faff51b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ddd5c5f58f840dc9ee288b478472046 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9af2e9fede44b8a9aa13582bbf33d28 : SUCCESS in 2m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/30da27bc81fc475ab653ebf61f4f6c80 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f4acd35050b449fada079ae43be0038 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dceb4c2e1018484d88959e0987f3c3a3 : SUCCESS in 10m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6a090c9706ad4d20a7c50a6c61bb5ebd : SUCCESS in 1h 00m 42s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87b769b628344e0890afb9d1edce4b0c : SUCCESS in 55m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f62a8da5e884bb592a6851b7e8cb51e : SUCCESS in 48m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/aece4c05bb1b4abcab607e0962ad9801 : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2dcaa6ebea444dc0a7c5d3411db32056 : SUCCESS in 20m 09s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef373625806445a29392eaa2b4e867be : SUCCESS in 13m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9dc6edffb8bd43518b8abc2a9fdc0a6c : SUCCESS in 15m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5873c9334e5644509f47d00586e35f35 : SUCCESS in 17m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd8e832d23eb48c1bba55691b14fe39b : SUCCESS in 16m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37e98c5f7f0047d2ba268bd204777795 : SUCCESS in 19m 05s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0b8951ee820247d6802dc65c395f4be8 : SUCCESS in 27m 25s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c4509bd316c49b797347659e6621978 : SUCCESS in 29m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d441fe5add02429089f1b500defb2298 : SUCCESS in 16m 25s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5accdc5376b04e4c84bfe629e8e0f237 : SUCCESS in 13m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a29e0a4f5e4c4d7fbd835e0179fc81ce : SUCCESS in 15m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/de6c26c991744806853ee6293d769319 : SUCCESS in 17m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/490fea94962b49dfad76903d96ab2500 : SUCCESS in 15m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e2ad4f38689d4ceda457674a5b9d1b2a : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":1},{"id":"76f668d5ac513eccb9e2edbe80cfc46f6108c717","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 15:18:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d1090da9f9db7c10671a288411b734e0c01ef517","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 15:19:25.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/042fee6ca0ce4bc38553c43ff8c41d24\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21e16790cdcc49fc85efb16cb8b52197 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a71ce4c5b85c4e8eb52e79153086e660 : POST_FAILURE in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9a0af887bb0178036cb1a25e08197bb64372d3c","revisions":{"d9a0af887bb0178036cb1a25e08197bb64372d3c":{"kind":"REWORK","_number":1,"created":"2025-06-12 11:10:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/952443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/952443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/952443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/952443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/952443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/952443/1"}}},"commit":{"parents":[{"commit":"ddca6265f64fa8e9cab331cb571b8dff0a476804","subject":"Fix image registry conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddca6265f64fa8e9cab331cb571b8dff0a476804"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-11 13:56:24.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-12 11:09:28.000000000","tz":120},"subject":"Use latest CentOS Stream 9 GenericCloud image","message":"Use latest CentOS Stream 9 GenericCloud image\n\nChange-Id: If07b7b5aeff61c4a1181d4e4a6cc256456125c0a\n(cherry picked from commit 46d28dc7996efafca5be3c9f8280fbd291806f48)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9a0af887bb0178036cb1a25e08197bb64372d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9a0af887bb0178036cb1a25e08197bb64372d3c"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
