)]}'
{"id":"openstack%2Fkayobe~952636","triplet_id":"openstack%2Fkayobe~master~If0771fadb0efd4d48ba5af0e5cab8176dde35193","project":"openstack/kayobe","branch":"master","topic":"kolla-enable-bifrost-leftovers","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-06-24 11:49:07.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-06-24 11:49:07.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-06-24 11:49:07.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-24 10:20:14.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"If0771fadb0efd4d48ba5af0e5cab8176dde35193","subject":"Drop some leftovers","status":"MERGED","created":"2025-06-14 18:54:03.000000000","updated":"2025-06-24 11:49:57.000000000","submitted":"2025-06-24 11:49:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952636-kolla-enable-bifrost-leftovers","meta_rev_id":"e903c74fd4d9422584fc05b38962b776be98da3d","_number":952636,"virtual_id_number":952636,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-24 11:49:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"all":[{"value":1,"date":"2025-06-15 12:19:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-18 13:12:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":2,"date":"2025-06-24 10:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-06-18 13:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":1,"date":"2025-06-24 10:20:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-14 18:55:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-06-14 18:55:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-06-14 20:04:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-15 12:19:10.000000000","updated_by":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"reviewer":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"state":"REVIEWER"},{"updated":"2025-06-18 13:11:46.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-06-18 13:12:02.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2025-06-24 08:40:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"b6503a804e0f72a01a0feb09fd51e23d41caf261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-14 18:54:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9f9c048bb2584c2387aed1b54db9fdf9fdf4180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-14 20:04:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e67c49ede3184610a470c6c550de31d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a0256d089a4c09b95638b1fa5aff73 : SUCCESS in 7m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0d60328efe7f44f3968679a5fc2ae8bb : SUCCESS in 5m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/466240c57932474fab2474f7128a495f : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778d0ebd005c45a0b5d766d40977a090 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28e036bdeb554e6b953d79172f76605e : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35771c51d20345f1bf01a5cc6796f5c9 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c3364228d204866b4bcc0f04bd9d493 : SUCCESS in 6m 31s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3788db3790d421482ebf0ace8ccba89 : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d8e0b44b2114988ae4b5530ac3c19e1 : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/27137d239a8c421197b36b5b9fcfdddb : SUCCESS in 8m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/655d9196bef443f78bd967beb33d6570 : SUCCESS in 49m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/17024c97970a4a25af92d0fdab4e83bc : SUCCESS in 31m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/811fb449378b45c098eaf265ec215dd5 : SUCCESS in 12m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55f017a2e3a8479c9ef784bd7620d650 : SUCCESS in 6m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d509037bb03479bbd8daa7a10445cad : SUCCESS in 1h 09m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/75b86a843f3d4487ad45f718791e1e76 : SUCCESS in 12m 38s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/7063cfbdbe184295896afddb68d88386 : SUCCESS in 16m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d48f01a77882432dac371b6ffa067c36 : SUCCESS in 15m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/88f313b2ff0e4608bfb67e968fe105a7 : SUCCESS in 10m 27s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bad868a578964fcd8e3197d6c0457268 : SUCCESS in 18m 01s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7a50511b8135439fa21e8b235621951e : SUCCESS in 22m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/493235a2581943a9a8b38382c4ebe30d : SUCCESS in 27m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ad68d6f087b94092a5727ae6a818e0c8 : SUCCESS in 9m 22s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/412b47ccca64403bb9b3837e110a7450 : SUCCESS in 13m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bc413aadc0d47038f18c87935534ab0 : SUCCESS in 11m 41s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/453738c204c947269f03a46dbad372c2 : SUCCESS in 5m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9232bb68f7cb46eca6284d9a778f8491 : SUCCESS in 7m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02630266de544860b8ef4714eb276cb9 : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"bf63741317bd01249ad50f2b9e83c83a00f96e63","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-06-15 12:19:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ef99d21b6825ac47e5903a9fb96aae2e8d6f5f92","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-18 13:11:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"767ddbf5abfcf1b3b7a9fb673b2c952351750b83","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-06-18 13:12:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"918ff3a62f8d3eedefa7ec1cdd90bd8ba400fe9b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-24 10:20:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f76858f87f399349f75e21bd5f6d946a9e38d28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 10:20:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70104e46ba2cafd6f3714e461c6ec79ca9e5a448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 11:49:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58852491b3d044da83cff10304d44458\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a2b3fac23b549a9ab5747c6de99cd35 : SUCCESS in 4m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5edd3967652c428c9b7824186f2a6cee : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee4cf90674a248fcabf3c3f3767a3142 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bee0accbd784ad29bd0e9db12736227 : SUCCESS in 1m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79649a388fb94283923146b39ccc80dc : SUCCESS in 2m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b483a40a9e542c8b392ff244037245a : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db21ed82ba9144bd965d669486baaa29 : SUCCESS in 5m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16bac8d3d11442a4836c8831918648ee : SUCCESS in 5m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52d687d6cdbb4090baf655c00bb18d3d : SUCCESS in 43m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bee3df4ae2d44460b09d4e0d398409c0 : SUCCESS in 32m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3fa1cd18cf4045b1a1c60252be524a59 : SUCCESS in 14m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c468de2e8ad2487e8230b34eef3f1c6d : SUCCESS in 11m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42e63d73fe7540e3b581676b586c5ac5 : SUCCESS in 1h 28m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/15490779a8a243ac92fcc54d3bd30b0c : SUCCESS in 13m 48s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/b8c20087b8a5448385ed51a523e76cc0 : SUCCESS in 14m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a591a0ed995401b8b7b795b07e3e423 : SUCCESS in 11m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/679fad26ca74488187a598f674c24047 : SUCCESS in 13m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1edc409d2ed94c64ad03df81d8bbbe3a : SUCCESS in 31m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b43560cdcc84fd28327a6166c506f65 : SUCCESS in 27m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/51d86b58479b4b9592994471aa0545d3 : SUCCESS in 14m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a44127d0bf7a4aa9a8fd218449177e77 : SUCCESS in 13m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/016f74b3872244ada74b166ef6ba7b9f : SUCCESS in 12m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f19d9c52a5f44bb8a9a2dbe8704767c : SUCCESS in 9m 05s","accounts_in_message":[],"_revision_number":1},{"id":"027b850866e66235785b6e72cf5fe66d9c6e2677","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 11:49:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e903c74fd4d9422584fc05b38962b776be98da3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 11:49:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a273795da2d44966b1585fb43d81c3fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1673fc965c842c782b75dc86d166160 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d72f5c7e4b60f4b597ae1db7f0daed3d88a3d8a","revisions":{"5d72f5c7e4b60f4b597ae1db7f0daed3d88a3d8a":{"kind":"REWORK","_number":1,"created":"2025-06-14 18:54:03.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/36/952636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/952636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/952636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/952636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/952636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/952636/1"}}},"commit":{"parents":[{"commit":"1e145b039bfe468325e5c70b581eacc2cdf4477a","subject":"Merge \"Add support for making `Bifrost` optional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e145b039bfe468325e5c70b581eacc2cdf4477a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-06-14 18:52:48.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-06-14 18:53:28.000000000","tz":180},"subject":"Drop some leftovers","message":"Drop some leftovers\n\nFollow-up on I7c5ac03b3ceb3f558480044f385636936df90f8f\n\nChange-Id: If0771fadb0efd4d48ba5af0e5cab8176dde35193\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d72f5c7e4b60f4b597ae1db7f0daed3d88a3d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d72f5c7e4b60f4b597ae1db7f0daed3d88a3d8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
