)]}'
{"id":"openstack%2Fkayobe~952809","triplet_id":"openstack%2Fkayobe~master~I825ab7331fbe8f28c17b87ff79a436144426a78c","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2025-06-20 09:57:48.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-06-20 07:32:29.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-20 09:57:48.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-06-20 08:01:02.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I825ab7331fbe8f28c17b87ff79a436144426a78c","subject":"Fix infinite loop in seed container deployment","status":"MERGED","created":"2025-06-18 09:16:08.000000000","updated":"2025-06-20 09:58:46.000000000","submitted":"2025-06-20 09:57:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952809","meta_rev_id":"dec9228b40cdb89b1b51b6f92a828ba472168955","_number":952809,"virtual_id_number":952809,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-20 09:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-06-20 07:32:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-19 17:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-06-20 08:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-06-20 08:01:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-19 21:40:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-19 17:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-18 10:42:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-06-18 10:54:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-18 11:28:16.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-06-19 10:42:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"06d1af1718d3609493fd5dfb2bb956348e712e45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-18 09:16:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"483d42dc43985ea1065b6526e275bc1386cda127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-18 09:16:46.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a4937452cc695988f43e1899748bf1bf9356e8fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 10:54:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a23932cf02a4f569ca75460ee0c80a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/286fc7edc5944cdcb0f3112f3d6900d2 : SUCCESS in 9m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/92216eb2091a4a648279e5609d3daae4 : FAILURE in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16a2fd71fdbe4cc084148eb7c12fd216 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/820eed3f9c294b7e954f8164cffe4678 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68df4cc2cca247479d53347a9babea38 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b0595c2063643ed873b605d88b69c2e : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de632b38729241ceaebfd229c5a4da08 : SUCCESS in 9m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08d34236011e455f857a1008f7a4ca9f : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9bbef204d6994b079a8528cbed4b9398 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/93d33f1850a1420a8fd350e82309be06 : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/baa4631ec17c492194e26ba8f9e5fa41 : SUCCESS in 9m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f84b0367de17499d80e5cb1aba91d825 : SUCCESS in 58m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7973dc8af3534707b150cb15dfaef2d0 : SUCCESS in 53m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/198ad2ad8af143898bbd7b20afeb0d1e : SUCCESS in 18m 02s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7aad1f474454e0fa8d21dfa18ff2378 : SUCCESS in 13m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a6905d402f64b7da989717171ffd9f3 : SUCCESS in 1h 36m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bc136c271324498ba53b863e820765a6 : SUCCESS in 15m 15s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/fd325a39ca074847a405a5919b3ae150 : SUCCESS in 16m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a33438b669484e3cb6d58d4843ce2add : SUCCESS in 18m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1b0add91df2941daaeb5168d16a63dd6 : SUCCESS in 16m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/efb9886a9b5d4873a32c7e00df0904d5 : SUCCESS in 42m 21s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1d82e397773b477eb673dc793c724107 : SUCCESS in 35m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2464e4535824a31ab3350b0b86ac2fa : SUCCESS in 32m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eb0acd44c45645ad9e2b9f885c715509 : SUCCESS in 15m 34s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/1b0365aba57946c99f373204dc0d214c : SUCCESS in 15m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6f8a6c36a5345f59edbc670ec807617 : SUCCESS in 13m 32s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/c8722c17a969420890189da7f8ae11b3 : SUCCESS in 13m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f454ec4be39b41ca9a3fa0f21a96dabc : SUCCESS in 14m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f28c1511aed041f5aaf04a4db37dd19c : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c0c9bee941312c8b20ed8e1d98bbbe461d4d0e7c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-18 11:28:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"461c392f0293750672afac0608a83046a11e5445","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-19 10:21:11.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"aa65d7dae6a2f8f7a92623e77696736557eed418","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2025-06-19 10:22:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d64a60dbfe05925300a8b4d82c566b4abbf9fb81","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 10:42:21.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da2083c4051605f57e467fb9eacf6029a28685bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 12:24:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e673f6c7ac0e486fab1d9309de6d4833\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e304d82f8fa24b69b3353cdd7caed123 : SUCCESS in 9m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/62099c4168da462bb8638c1043a726ae : SUCCESS in 13m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2cb428cc9f24ba08295a9bb0d725e02 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5241ade9e62b4b789978d1c6e02720b7 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/666ee9ac3b414d69b54e58ceeaee37e1 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49621288213745b691de47e23d555858 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6ffc67439854514a37dae94b5a15399 : SUCCESS in 5m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0430bf1c0c854135b0942e069f32b7fd : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e033e73aabb64fa293841d3df627bf9b : SUCCESS in 10m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ef85aa2ca90431c98201eb9b909519e : SUCCESS in 20m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eb1aa28538d342879075c9eb7a83c67a : SUCCESS in 11m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e63ac127b554211a9876846c11c0db6 : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/44768d9cb5354f348e340b4c9a780210 : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fb8215f14c06460f94507bb300bac36e : SUCCESS in 30m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9f21b0c14904c7793b394802e364016 : SUCCESS in 25m 08s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25d3ab3e53384eec97174df4cdf621c4 : SUCCESS in 2h 01m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1996b27fb18740598a09413fb8ada9ce : SUCCESS in 30m 44s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/b4f2589dc1014fd7a74c2813df43ea65 : SUCCESS in 30m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/495f15c0e2584083b99e6815699255a2 : SUCCESS in 26m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ada000371a54ff0bb528a7fb1c3adb3 : SUCCESS in 25m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c787787fe7c34fc5b0e0b2ab7aa8c0b6 : SUCCESS in 55m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ad345fcc61fc48108a526ba9234e20b9 : SUCCESS in 48m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eff87200b1fd45c39dd135ffd9a2dfcb : SUCCESS in 50m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/afe58bfbba40421580da02d703930481 : SUCCESS in 29m 03s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/b4c9116b129b428183995a527219b935 : SUCCESS in 29m 26s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0703017b19b4fc08d4ae75e5c86599a : SUCCESS in 23m 02s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/f06d21687c254e2bb97623bcce07bf1f : SUCCESS in 23m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/57711d16655146e8b0dd9460f39b2510 : SUCCESS in 29m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/045789b5ddf8488a99b72e62224246f1 : SUCCESS in 23m 11s","accounts_in_message":[],"_revision_number":3},{"id":"34bf6eeabf887773cb691fed8ed0df238245401a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 17:29:54.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5b542a08adfcd4d44dbe9578c9e5a9774fc47f86","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 17:30:13.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"626a0998cf48bbaded331ded5398e068cb3ffdce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 19:37:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2db2c03dab3c4e7193bf72686c1a2acb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9dc290c7fe431f8b93819263d1bc8d : SUCCESS in 10m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6fd492701f314360a5b0fd3c5c590dd4 : SUCCESS in 7m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ae30d82a2274c75b019c4484757af00 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cfc0a37664b493a9e5357f1240cf014 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/986e5ac1dc9c49d9a8bd19459ed3d0b2 : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8a41b26b90e4e0281d5b1cd44a11ce2 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2619dfaa5c045cd9e149b8b9210bf43 : SUCCESS in 6m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4e93c60292145bda797ac957dfb9f20 : SUCCESS in 2m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d514901b1f144015b10ef0b99d8d92f9 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/27c45734e96e43b495e7b096ada1c226 : SUCCESS in 12m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/482c7d841dc1413986ecdefc7c0d0b9a : SUCCESS in 9m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/843e9870c82d4bc989d0ef0f20b29d51 : SUCCESS in 57m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cff8a1ca8b3b4efe8f26b36c823f1227 : SUCCESS in 1h 03m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f4ccc2b22e80499b98c529d9e52e1670 : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e736de8b0d7144be86253709766edddd : SUCCESS in 14m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9ffa5af47894025a36fff54628ddbbe : SUCCESS in 1h 54m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c6744f5143224b2a8749d21500797a41 : SUCCESS in 16m 04s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/060632ad3e4941f98b5325261ebf8531 : SUCCESS in 16m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b9f03555289459a9873f7d4241d1433 : SUCCESS in 14m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/238db6efafe544f094e8370e32bdb63e : SUCCESS in 14m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7815c63a72fd4d8ca2336cf6326ff4c2 : SUCCESS in 39m 34s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4bd8c4ef2c36400a94a294f9d2dbc393 : SUCCESS in 34m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b8f3a0dde71427d89c7bd54e46d7ab0 : SUCCESS in 31m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/168645731cc3488d8a644f9cc5ba5f5e : SUCCESS in 13m 28s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/6ad069414bb946e6856593232d174ed9 : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f6268196af44bd1a2ce3e9dd63debe3 : SUCCESS in 13m 07s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/27fb6d2b2ce4482ab44f88ed4dbe3be5 : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a0bdedad7caa47deb0eb12c054b8da72 : SUCCESS in 12m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8aa13300ee94c3596d99fa143adb05d : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":4},{"id":"24ea122d409cc40367d62f3009150ae4875ab0a5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 21:40:34.000000000","message":"Patch Set 4: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bc30c22c00fdeb110e53b31e4d9fcac5a81a7c5e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-06-20 07:32:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5792d6d468478e425acacab5706c99a5195fdd92","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-20 08:01:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"513c4ae32d15082250dbef1c4ccccde5c07fc90e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 08:01:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"eb0144b474c41d3760f9f94fce44df16f150b8be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 09:57:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e551a7db0e74beea4751789ea6b92e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebaf5dd55ab84204a5ed5fc47923a5a4 : SUCCESS in 9m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0896ce0561194b7b9dc5e76e81b19065 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66020abc63ef4964bb63fd21c854b8ed : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce12d1a0a0ac4ede9521a6ac306f0422 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1eaf56b4ddb43e0b99357842dbedc47 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/227ffad4bc594571b59294e9e21924c0 : SUCCESS in 2m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5cea3cf78bb441dd9a6cb2898a145a28 : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eb5c31c3a58e40dda40e1e943ba1c444 : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad167f7959124e399d592d4e10f12610 : SUCCESS in 10m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fdc3b7a73bd64af1a1d96146afb9d9be : SUCCESS in 55m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4609e00c319745dfbf37a210c944cc5d : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c7df0b3161894246a194eb31782d0f08 : SUCCESS in 17m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/916f5d3b4ffd4e01acfae733224e7cd0 : SUCCESS in 15m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77696faf05054fb5bf554b79699c7083 : SUCCESS in 1h 50m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fd73d22d964a4b7ca56695b154c4d93c : SUCCESS in 18m 11s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6ccf9af111574ade9ad90e1feffb4055 : SUCCESS in 17m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94a5e088d0654daea25d3619c52a93ac : SUCCESS in 14m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c6935be73dd549e9b8f8329fb4e50f3c : SUCCESS in 13m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1e9162cfb09b4b978bcd0878564c68f4 : SUCCESS in 29m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2ecd8fa3bf6420199df333d13700228 : SUCCESS in 27m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a68e3810c34f4ed880395ee95077ab81 : SUCCESS in 13m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2823f606c30f4fdaa95451e69eaca7c3 : SUCCESS in 12m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c57fe7e749e44d893d7ca9680a8db69 : SUCCESS in 11m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e89084e7335486aacd7ca8f30c370f0 : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":4},{"id":"095619fffb9c3c883785a7ad691b62ef9c327aea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 09:57:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"dec9228b40cdb89b1b51b6f92a828ba472168955","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 09:58:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb9ca05d8e7442939ba7d08904febab2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6c02d6415f47dd87184c730c19261a : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41459439f5564b2dbdcaa431bcbcf2a5 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7bd7929ea77f9843717a424d88422837d19d0d4a","revisions":{"dc61c341bb7684cb36e910b8c0fb39a3e0a9101f":{"kind":"REWORK","_number":1,"created":"2025-06-18 09:16:08.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/09/952809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/952809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/952809/1"}}},"commit":{"parents":[{"commit":"1e145b039bfe468325e5c70b581eacc2cdf4477a","subject":"Merge \"Add support for making `Bifrost` optional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e145b039bfe468325e5c70b581eacc2cdf4477a"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-18 09:09:56.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-18 09:17:33.000000000","tz":60},"subject":"Fix infinite loop in seed container deployment","message":"Fix infinite loop in seed container deployment\n\nThe deploy.yml playbook previously called deploy-container.yml, but this\nwas changed to call itself when podman support was added [1]. It would\nthen recursively loop until failure.This change reverts that, so\ndeploy-container.yml is called instead.\n\nCloses-Bug: #2114845\nChange-Id: I825ab7331fbe8f28c17b87ff79a436144426a78c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc61c341bb7684cb36e910b8c0fb39a3e0a9101f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc61c341bb7684cb36e910b8c0fb39a3e0a9101f"}]},"branch":"refs/heads/master"},"d0b522bb9bfc8164b9fbb0464232dcdf66f74870":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-18 09:16:46.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/09/952809/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/952809/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/952809/2"}}},"commit":{"parents":[{"commit":"1e145b039bfe468325e5c70b581eacc2cdf4477a","subject":"Merge \"Add support for making `Bifrost` optional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e145b039bfe468325e5c70b581eacc2cdf4477a"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-18 09:09:56.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-18 09:17:55.000000000","tz":60},"subject":"Fix infinite loop in seed container deployment","message":"Fix infinite loop in seed container deployment\n\nThe deploy.yml playbook previously called deploy-container.yml, but this\nwas changed to call itself when podman support was added [1]. It would\nthen recursively loop until failure. This change reverts that, so\ndeploy-container.yml is called instead.\n\nCloses-Bug: #2114845\nChange-Id: I825ab7331fbe8f28c17b87ff79a436144426a78c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0b522bb9bfc8164b9fbb0464232dcdf66f74870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0b522bb9bfc8164b9fbb0464232dcdf66f74870"}]},"branch":"refs/heads/master"},"d2ae6356f2199430c44e9e7303bd9d06df4a7c30":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-06-19 10:21:11.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/09/952809/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/952809/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/952809/3"}}},"commit":{"parents":[{"commit":"1e145b039bfe468325e5c70b581eacc2cdf4477a","subject":"Merge \"Add support for making `Bifrost` optional\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e145b039bfe468325e5c70b581eacc2cdf4477a"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-18 09:09:56.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2025-06-19 10:21:11.000000000","tz":0},"subject":"Fix infinite loop in seed container deployment","message":"Fix infinite loop in seed container deployment\n\nThe deploy.yml playbook previously called deploy-container.yml, but this\nwas changed to call itself when podman support was added [1]. It would\nthen recursively loop until failure. This change reverts that, so\ndeploy-container.yml is called instead.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/909686\n\nCloses-Bug: #2114845\nChange-Id: I825ab7331fbe8f28c17b87ff79a436144426a78c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2ae6356f2199430c44e9e7303bd9d06df4a7c30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2ae6356f2199430c44e9e7303bd9d06df4a7c30"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7bd7929ea77f9843717a424d88422837d19d0d4a":{"kind":"REWORK","_number":4,"created":"2025-06-19 17:29:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/09/952809/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/952809/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/952809/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/952809/4"}}},"commit":{"parents":[{"commit":"9e18a6e02a72ed08d9eef85d8b186d8f12cf472c","subject":"CI: Fix new failures with ansible-lint 25.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e18a6e02a72ed08d9eef85d8b186d8f12cf472c"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2025-06-18 09:09:56.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 17:29:38.000000000","tz":120},"subject":"Fix infinite loop in seed container deployment","message":"Fix infinite loop in seed container deployment\n\nThe deploy.yml playbook previously called deploy-container.yml, but this\nwas changed to call itself when podman support was added [1]. It would\nthen recursively loop until failure. This change reverts that, so\ndeploy-container.yml is called instead.\n\nAdd a test to seed jobs, to catch any regression in CI.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/909686\n\nCloses-Bug: #2114845\nChange-Id: I825ab7331fbe8f28c17b87ff79a436144426a78c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bd7929ea77f9843717a424d88422837d19d0d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bd7929ea77f9843717a424d88422837d19d0d4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
