)]}'
{"id":"openstack%2Fkayobe~952889","triplet_id":"openstack%2Fkayobe~master~I2b0bdb5b4d3db7298a43913cd550b5ee769c94b7","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-19 17:30:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2b0bdb5b4d3db7298a43913cd550b5ee769c94b7","subject":"CI: Test seed custom container deployment","status":"ABANDONED","created":"2025-06-19 10:41:12.000000000","updated":"2025-06-19 17:30:10.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c426d189575316a02b5d272e82e0d1dee024b52f","_number":952889,"virtual_id_number":952889,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-19 17:04:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-19 13:23:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c718f1eacce01a18827ec33a65e82b79e75e71c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 10:41:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86027e6de90c6d491b22d3837f6828cb5c732891","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 11:11:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5f840a14fd72a8466317082316f0a269182f0a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 11:12:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1cba6802a28e7f73fa0c0b121048d12283dd406e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 11:41:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"31b4897ea0d901a199cf615ab2e403ab6af9a3a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 11:42:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3b5dddec51c1a30de17d9f067c7e8a2e18367528","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 12:50:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6329a1c59d1fa332f719e60f01d4e543ffb1a671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 13:23:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05bb6ac7e9ea44debfb14032630230e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17a0f57be414963b4434948bcfcd415 : SUCCESS in 9m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3eed72ad04c94e5c941f5657ff88e584 : SUCCESS in 6m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1834de0d3d154688ae2cd8a295b797c5 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a1b0f20d7144c480c5a975aca7ab05 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83f03eccced3417c9897ae9aada9b482 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de99f0ba13484e14b789ae5ce40e68ee : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65d9589aa4d74134b5e7a547682447a4 : SUCCESS in 6m 47s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/927253dc60b146c6af0946e37d3a67dc : SUCCESS in 16m 15s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/7fb74ab590a14417809c74b6863dfd4f : SUCCESS in 16m 04s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6533dc3726645e68839e7baf27c89ab : SUCCESS in 14m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1618517ccb0148579675f52891cb5da2 : FAILURE in 8m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/804c3cddacc24b9fa48d9a6dce25b64c : SUCCESS in 32m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec0b21cee8e34ea9aa36259c9cbde071 : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":5},{"id":"189b6cfe284a703a95d23a59702060c1b1dc2241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 14:01:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"64e7d763d20d33902ea1b5c6b0f5e04716619227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 14:51:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/deb52e1e1a1e47f9800737741248242f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80717dd999404453a503e64d8df2f92c : SUCCESS in 10m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8a764f8ae778481bb5bba2609b9b35f0 : SUCCESS in 10m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb8d3a690fe24008812eb5c94b68bf1d : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/301b78980111484cb5447393def687ad : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/738a4bf3996c44cba8c869476b12e2db : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f0cfff6bde64b098b6862b7eb3a95f7 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/62ff6ca2a0af4baa9975e8e5b2a55129 : SUCCESS in 6m 07s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/810863c4e40d4799b8ee011318a2ad1f : FAILURE in 14m 00s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/c7891384d82543b18227946aa4145c91 : FAILURE in 14m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2370d3396bf248f38c6c8ec061269232 : SUCCESS in 25m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c2fd18585d044659aa1adf79ce1ae21b : SUCCESS in 25m 25s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8425fd0b35fa473bbf7f7e5f62b9629c : SUCCESS in 38m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d54a71fc53394fa89f7db5d61181ec02 : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":6},{"id":"27f8746b29396ecd7d1d74c3d5781c582df1a2ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 15:48:35.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"478152c6e2db3435b24fc3f9af93d57125aa8550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 16:04:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4d5ec7b7d5f2bfd786c12655697a1f85943717e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 16:29:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"987b68815f9f2e66be76182eb6fb0a66830279d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 17:04:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d97144e072e4ef987c7c6dfa74e9e7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c3d8ff5063b4762a6fbc74c3b3db640 : SUCCESS in 8m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/60f052b2fffd4a46884ccbd1a7f3d573 : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b80d77346e434600a53a6606c6d2cedd : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da660c7b9134c0db96597aa79084b55 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bea1452bf4be420ea1a935879814249d : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48fac800ee6a433a923877c0da23a1cb : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/341ca1c83e2d42e1a56f6c09a436f1b9 : SUCCESS in 5m 37s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bf0d66783fbf476184e4d4c2d80e2928 : SUCCESS in 18m 45s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/c2fe6542a0fc407d99e1a2bcdbf8f995 : SUCCESS in 16m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fa38d9bdc7b4024b5f6494b427423b5 : SUCCESS in 13m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/80edfff5ca814b4791466119437e738d : SUCCESS in 15m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e0521d1b69ca440e9327219d85c784e0 : SUCCESS in 29m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/066ebe79c6b6425a8f159a37b5da877c : SUCCESS in 29m 40s","accounts_in_message":[],"_revision_number":9},{"id":"c426d189575316a02b5d272e82e0d1dee024b52f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-19 17:30:10.000000000","message":"Abandoned\n\nMerged into https://review.opendev.org/c/openstack/kayobe/+/952809","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"7108b7c131cd1261b2efe9c38ea34420a602ea5c","revisions":{"3fd71feaaaa69dd2aca05e9b4fe92969d826c9f6":{"kind":"REWORK","_number":1,"created":"2025-06-19 10:41:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/952889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/952889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/952889/1"}}},"commit":{"parents":[{"commit":"9e18a6e02a72ed08d9eef85d8b186d8f12cf472c","subject":"CI: Fix new failures with ansible-lint 25.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e18a6e02a72ed08d9eef85d8b186d8f12cf472c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 10:39:21.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 10:39:21.000000000","tz":120},"subject":"CI: Test seed custom container deployment","message":"CI: Test seed custom container deployment\n\nChange-Id: I2b0bdb5b4d3db7298a43913cd550b5ee769c94b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3fd71feaaaa69dd2aca05e9b4fe92969d826c9f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3fd71feaaaa69dd2aca05e9b4fe92969d826c9f6"}]},"branch":"refs/heads/master"},"6ad932b80e4d70756935ee1754cc38821e29064b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-19 11:12:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/952889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/952889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/952889/2"}}},"commit":{"parents":[{"commit":"d2ae6356f2199430c44e9e7303bd9d06df4a7c30","subject":"Fix infinite loop in seed container deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2ae6356f2199430c44e9e7303bd9d06df4a7c30"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 10:39:21.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 11:12:04.000000000","tz":0},"subject":"CI: Test seed custom container deployment","message":"CI: Test seed custom container deployment\n\nChange-Id: I2b0bdb5b4d3db7298a43913cd550b5ee769c94b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ad932b80e4d70756935ee1754cc38821e29064b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ad932b80e4d70756935ee1754cc38821e29064b"}]},"branch":"refs/heads/master","description":"Rebase"},"6ac971029fab38bb854fdf37b960f674f6e9f1d7":{"kind":"REWORK","_number":3,"created":"2025-06-19 11:41:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/952889/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/952889/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/952889/3"}}},"commit":{"parents":[{"commit":"d2ae6356f2199430c44e9e7303bd9d06df4a7c30","subject":"Fix infinite loop in seed container deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2ae6356f2199430c44e9e7303bd9d06df4a7c30"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 11:41:38.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 11:41:38.000000000","tz":120},"subject":"CI: Test seed custom container deployment","message":"CI: Test seed custom container deployment\n\nChange-Id: I2b0bdb5b4d3db7298a43913cd550b5ee769c94b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ac971029fab38bb854fdf37b960f674f6e9f1d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ac971029fab38bb854fdf37b960f674f6e9f1d7"}]},"branch":"refs/heads/master"},"96856f3aae042d6c489da08949171b397b692edc":{"kind":"REWORK","_number":4,"created":"2025-06-19 11:42:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/952889/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/952889/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/952889/4"}}},"commit":{"parents":[{"commit":"d2ae6356f2199430c44e9e7303bd9d06df4a7c30","subject":"Fix infinite loop in seed container deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2ae6356f2199430c44e9e7303bd9d06df4a7c30"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 11:42:04.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 11:42:04.000000000","tz":120},"subject":"CI: Test seed custom container deployment","message":"CI: Test seed custom container deployment\n\nChange-Id: I2b0bdb5b4d3db7298a43913cd550b5ee769c94b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96856f3aae042d6c489da08949171b397b692edc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96856f3aae042d6c489da08949171b397b692edc"}]},"branch":"refs/heads/master"},"da2d846797a6acd0f1a5fcea988b1fc328156d74":{"kind":"REWORK","_number":5,"created":"2025-06-19 12:50:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/952889/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/952889/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/952889/5"}}},"commit":{"parents":[{"commit":"d2ae6356f2199430c44e9e7303bd9d06df4a7c30","subject":"Fix infinite loop in seed container deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2ae6356f2199430c44e9e7303bd9d06df4a7c30"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 12:50:25.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 12:50:25.000000000","tz":120},"subject":"CI: Test seed custom container deployment","message":"CI: Test seed custom container deployment\n\nChange-Id: I2b0bdb5b4d3db7298a43913cd550b5ee769c94b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da2d846797a6acd0f1a5fcea988b1fc328156d74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da2d846797a6acd0f1a5fcea988b1fc328156d74"}]},"branch":"refs/heads/master"},"a32ca598f586d121d71c341679f46a12604c1361":{"kind":"REWORK","_number":6,"created":"2025-06-19 14:01:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/952889/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/952889/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/952889/6"}}},"commit":{"parents":[{"commit":"d2ae6356f2199430c44e9e7303bd9d06df4a7c30","subject":"Fix infinite loop in seed container deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2ae6356f2199430c44e9e7303bd9d06df4a7c30"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 14:01:13.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 14:01:13.000000000","tz":120},"subject":"CI: Test seed custom container deployment","message":"CI: Test seed custom container deployment\n\nChange-Id: I2b0bdb5b4d3db7298a43913cd550b5ee769c94b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a32ca598f586d121d71c341679f46a12604c1361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a32ca598f586d121d71c341679f46a12604c1361"}]},"branch":"refs/heads/master"},"c9e6c5f4fbcdd7545208d755338ab888ab33e109":{"kind":"REWORK","_number":7,"created":"2025-06-19 15:48:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/952889/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/952889/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/952889/7"}}},"commit":{"parents":[{"commit":"d2ae6356f2199430c44e9e7303bd9d06df4a7c30","subject":"Fix infinite loop in seed container deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2ae6356f2199430c44e9e7303bd9d06df4a7c30"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 15:48:19.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 15:48:19.000000000","tz":120},"subject":"CI: Test seed custom container deployment","message":"CI: Test seed custom container deployment\n\nChange-Id: I2b0bdb5b4d3db7298a43913cd550b5ee769c94b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9e6c5f4fbcdd7545208d755338ab888ab33e109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9e6c5f4fbcdd7545208d755338ab888ab33e109"}]},"branch":"refs/heads/master"},"a4c345d870e1b1a8851c8aeae53b43d8b87321dd":{"kind":"REWORK","_number":8,"created":"2025-06-19 16:04:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/952889/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/952889/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/952889/8"}}},"commit":{"parents":[{"commit":"d2ae6356f2199430c44e9e7303bd9d06df4a7c30","subject":"Fix infinite loop in seed container deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2ae6356f2199430c44e9e7303bd9d06df4a7c30"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 16:04:30.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 16:04:30.000000000","tz":120},"subject":"CI: Test seed custom container deployment","message":"CI: Test seed custom container deployment\n\nChange-Id: I2b0bdb5b4d3db7298a43913cd550b5ee769c94b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4c345d870e1b1a8851c8aeae53b43d8b87321dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4c345d870e1b1a8851c8aeae53b43d8b87321dd"}]},"branch":"refs/heads/master"},"7108b7c131cd1261b2efe9c38ea34420a602ea5c":{"kind":"REWORK","_number":9,"created":"2025-06-19 16:29:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/952889/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/952889/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/952889/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/952889/9"}}},"commit":{"parents":[{"commit":"d2ae6356f2199430c44e9e7303bd9d06df4a7c30","subject":"Fix infinite loop in seed container deployment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2ae6356f2199430c44e9e7303bd9d06df4a7c30"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 16:28:57.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-19 16:28:57.000000000","tz":120},"subject":"CI: Test seed custom container deployment","message":"CI: Test seed custom container deployment\n\nChange-Id: I2b0bdb5b4d3db7298a43913cd550b5ee769c94b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7108b7c131cd1261b2efe9c38ea34420a602ea5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7108b7c131cd1261b2efe9c38ea34420a602ea5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
